)]}'
{"id":"openstack%2Frequirements~681499","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ib1e9c883c4f59303fcfc1e898fa594f28297d316","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ib1e9c883c4f59303fcfc1e898fa594f28297d316","subject":"update constraint for paunch to new release 4.5.1","status":"MERGED","created":"2019-09-11 14:23:58.000000000","updated":"2019-09-12 09:57:31.000000000","submitted":"2019-09-12 09:57:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681499-1568282251031-7eb2037b","meta_rev_id":"5002a5734a0d451a89ae4e49ee62f2c66790147d","_number":681499,"virtual_id_number":681499,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-12 09:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-12 00:16:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-12 00:16:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 00:16:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-12 09:57:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89d38c914b43f77cfcdd374d0d040d99752e089d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-11 14:23:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99aaa14eabcb9cea36ca17b1d0e257d2b2e770a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 23:57:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f8cdce059d304aad91d1e37915986ad6 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d69aeaeca24dd8803bb6d7eac3ab1e : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be17ffd0631a4b48ba7d3d235b4f6f6b : SUCCESS in 3m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eb12c1e6688443cd99c4681acb5736ca : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4d56ecb95d94562b0b8cd5922c406d6 : SUCCESS in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebbcee0af8034395899e3b691f2fe39a : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06dd0d866829495cbcd9eab725e2e141 : SUCCESS in 3m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0faf6c093c43496e95dae7a466563fd9 : SUCCESS in 7m 44s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/96ec1c69351a481c91c8e9869284ce87 : SUCCESS in 12m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6f44919a7e4d4bfa982a03c2ced67543 : SUCCESS in 7m 01s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7c3a7c326b544d99912b440d69ca292e : SUCCESS in 9m 06s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/57cc2d0b38a24e23ac4a32ae8064f24a : SUCCESS in 11m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b66bc31fd8ad4d6c9e324636c327b51e : SUCCESS in 4m 58s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/3cdcd223134b41e0856aab095fda8d9f : SUCCESS in 5m 53s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/41a49211a44141178d9fd520722fff82 : SUCCESS in 13m 21s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1ee466fa61c24959adefbadf32c41570 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e059b793e43343bf88b705213d7e496d : SUCCESS in 6m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0ecfe73a799c4340b17c29944adc674c : SUCCESS in 28m 14s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/ad133a82daec4310948335123a19e4fc : SUCCESS in 26m 54s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/78923fa161dc434781d38e62800b262d : SUCCESS in 29m 18s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4df926f1becc42869eb259dab1bbdb19 : SUCCESS in 36m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0e33140175544c91bcc7cc2381462de1 : SUCCESS in 11m 36s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/8c6c728c1e3f482e90379fc3b4081419 : SUCCESS in 12m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d68747b306d64fe29eb295f8687d9efe : SUCCESS in 19m 04s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/108a984d892e47f68e93803085f77388 : SUCCESS in 4m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ef37d659b1ba410790db6e5a8852bee9 : SUCCESS in 9m 28s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/4365fa3e02d644acb427ba6dab0738fd : SUCCESS in 8m 37s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/66fec383797741fd9274ef6e9e9224d8 : SUCCESS in 1h 01m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b46c9289f4fe4cd589fb08afb3807244 : SUCCESS in 1h 17m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7496546d5de941d593b45adf4e5c7b83 : SUCCESS in 1h 52m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d5ed994abed44318b277093507719a7b : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":1},{"id":"44058c5130504fa4c5762331ae2dd76f19392eae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-12 00:16:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a65f511d2ccedce218dad35a74f2646d13c88e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 00:20:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd3dfb3fc709da67dd93787caa088fee3f4ec196","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 05:14:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/900f83a1cf774b909bb2e8d7748a5f35 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c201a798c0114599974b4f43ec8612c6 : SUCCESS in 5m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3512d8861dc34565ad449963fd8d45b6 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d80ef1e542c74bcaabceb0e7cee595a5 : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5165ed0472f041edb62a8b124d542cfa : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9fc007cef5846e9bccfa38950aff868 : SUCCESS in 4m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0f57f6b5fc294a0bb3d0e63392b15e62 : SUCCESS in 11m 21s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d4a659f7392f43e08d0a7d9d8d61a9cf : SUCCESS in 11m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a65c00760cdd408bbe0670939468f63b : SUCCESS in 12m 59s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/01af328c1411486e9fc55f93c425ce59 : SUCCESS in 13m 09s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/54317998426a46d6a8b5ab922d23acf5 : SUCCESS in 7m 36s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/fd28755e66e04a2580ffcb6ace71c791 : FAILURE in 4m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e4fa2647d201401c86eca3346fd0d5b7 : SUCCESS in 13m 59s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/3a8f9b966d244a88bb67bb632fabc7db : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e416c70e397d47bbaa0706d4ab5d92e7 : SUCCESS in 10m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/36735c335fdc4ea4a2ad3f1424b1a416 : SUCCESS in 21m 01s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/1337e8120b6c430cade18ac678a0e7a3 : SUCCESS in 24m 31s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/241d688ff74c4a6cb833d4d5f7acdecc : SUCCESS in 28m 48s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/662f51698a2c40ff877c4ed0ccfe7160 : SUCCESS in 29m 07s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d05259c5f54645178af7af3ba98fcdf6 : SUCCESS in 13m 09s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/89be93ec14d249d69df7fc24f7ddd4ad : SUCCESS in 12m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bfd22894e1de41788ee3def1d2720e88 : SUCCESS in 21m 48s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/482bd84d4f8a4b129846ce844bc7268f : SUCCESS in 5m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f51ae2af4c914935a6170cb2d7b8a009 : SUCCESS in 10m 02s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/c40ceb471574498c9642e2c0a8d830fd : SUCCESS in 9m 19s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/efde9ebb1da24556905f239a8b66e2ef : SUCCESS in 1h 04m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/636cb34472924fa0966d9f7c32e8c9a1 : SUCCESS in 1h 05m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8a03dcd553e445e6a5f1db1449f6086a : SUCCESS in 1h 44m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/05dbb1e5ee9f4caeb6be1b9dfa25014b : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b20198efd284351a17c4d240695daf51aba4e1ca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-12 05:35:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e992ea1872c8c41c9b56eb1809fd12763e94667e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 07:21:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4da21f6b58f348f2bad8d1f847e8caaa : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeb3524f742c469d8c207e94f59c458f : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bcedb64c85d4d2c8e413f8642bf90ae : SUCCESS in 4m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c0e26f95a3704cbd967f8985d93d2e03 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2151345eee1f45b5bb05026f7cf33cd0 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77d2e0bbd8b34e8bb8d9755a81b9227f : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17906c878f704d8e83b2d66e2ca24e31 : SUCCESS in 3m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1415223fdf57494fb382ca9292db7cab : SUCCESS in 10m 25s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3bddc9f97afa4000b76cbba06345fe23 : SUCCESS in 7m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4fca8554d4ae4d0f90aecded13e6cebb : SUCCESS in 6m 35s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0c2852334b7447329a8c5f324432ecbc : SUCCESS in 10m 34s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/feb32b6c391e4e008f1acdd4472bea15 : SUCCESS in 11m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4afc0d61628649e591111e67aad79d91 : SUCCESS in 5m 56s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/9630db86de704a3ba506941f3fefcf09 : SUCCESS in 5m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/45d15bfa450d4f36bc91c79993055475 : SUCCESS in 10m 39s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/ecb65a5073ad4229a0852bab9ca18a0f : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c06e9605dc2b4e39996bbf65e7957d1b : SUCCESS in 8m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/60242647224b4604ba6293a63b177916 : SUCCESS in 23m 00s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/5a8b4b386d07467db4bc1d3285bc740a : SUCCESS in 22m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4e289ce99bbe48b1b228d7464d2c4d7f : SUCCESS in 25m 36s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/072a49b0dbcb431b913a1d2bc8927d50 : SUCCESS in 25m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ed657f7187074b83a5379ddea84f088a : SUCCESS in 15m 28s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/fad50633a8d6485889ef838b30a4d987 : SUCCESS in 13m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ecb0b7aac0e045f9ab4febcb0ce294af : SUCCESS in 19m 40s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/4da694c267724b6a9aa33a392c8593ce : SUCCESS in 6m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0f3e74aa64644b7bb052b6b26d555bbb : SUCCESS in 8m 35s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/a765950f98514eb183dbda389100b7ca : SUCCESS in 8m 22s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/04c817c323204c2eaa80b4084d9edbeb : SUCCESS in 1h 05m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a46c278426546038ce8de2d3aa2e777 : SUCCESS in 1h 10m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b558aab5355d4d93ab2d7f9606743e8f : SUCCESS in 1h 44m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cc2c8fa455e24303a1ffa85df9c632d3 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7ef5c28bb23bfe2a8c2f117d83b95608a817689e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 07:21:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"497a751c5b80e5dc38957bfc2f10e2a2ff56e0b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 09:57:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bfb460ca92849fa9a53d23a43e49e9e : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f7616d3817444a78e1718c49646ba4e : SUCCESS in 4m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/36093d5ac231489c9372f1a0f40580c4 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/525bc1362156490d921c28b54e0b213f : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbb67a4ef756463ebd92aaac35af07d5 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f74648fe5c854678b57ba080b5e28124 : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/09a0860f179a4e668a304704617fddc5 : SUCCESS in 11m 38s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/204d8463945a4433ab8778c1c3765cfb : SUCCESS in 8m 38s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b5c14c978cb94084995429ed192f6a70 : SUCCESS in 10m 59s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/47b44f1f3f5e4791bd742d06d2c92cbd : SUCCESS in 15m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/bed90b8b2a804acea7c9b2b3d30dd298 : SUCCESS in 4m 58s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/7ec5fdc05bdf4c67ba0611511e1b7fa7 : SUCCESS in 7m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/136c3b4eca1749ef92701df97948a0c9 : SUCCESS in 10m 40s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/64f2f7c630f84558980daab2c8376413 : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ec6431a56304b65b2caa3e8128d7bac : SUCCESS in 7m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d322eabac22443a6b5b8132883dd845b : SUCCESS in 25m 31s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/15bd5ab5077944f7ab840b7396c43d8b : SUCCESS in 36m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/26cbd4de9b5e4268896f5378ce17bb77 : SUCCESS in 28m 58s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/228ae0b0b64849489a3dafb45ad8e956 : SUCCESS in 28m 45s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f8962b308a2747469d1a3868a70bd4a1 : SUCCESS in 18m 21s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/6b38c688f1d34d0e91a2bc4f6b5e79ca : SUCCESS in 17m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3dc4fcb283c344abac9302df95f195e8 : SUCCESS in 26m 20s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/0b024a17e0b24cc1a22b1625244bfc19 : SUCCESS in 5m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1bd99e6c90784767a34f8ab67fbbf478 : SUCCESS in 8m 23s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/36b6bbf5ecab435c8cc11822726e1017 : SUCCESS in 8m 16s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9233a4ec9298413baa646cee12b2c8ae : SUCCESS in 1h 11m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e9446988f07471fa62eb2cffa767945 : SUCCESS in 1h 04m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9dcff5e0ecf34041beaa9e9ce166117a : SUCCESS in 1h 32m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e715e22e48724acfb23317adc8e3401a : SUCCESS in 1h 02m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5c9f6c7c1e64186418c7ae8d34e7a7b84e141d52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 09:57:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d12220482cd47c760c70b5ec367a5d12ac741de","revisions":{"4d12220482cd47c760c70b5ec367a5d12ac741de":{"kind":"REWORK","_number":1,"created":"2019-09-11 14:23:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/681499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/681499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/681499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/681499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/681499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/681499/1"}}},"commit":{"parents":[{"commit":"00df062c5811566268ee3e007254fdcf485ee06f","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00df062c5811566268ee3e007254fdcf485ee06f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-11 14:23:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-11 14:23:56.000000000","tz":0},"subject":"update constraint for paunch to new release 4.5.1","message":"update constraint for paunch to new release 4.5.1\n\nChange-Id: Ib1e9c883c4f59303fcfc1e898fa594f28297d316\nmeta:version: 4.5.1\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: I155861dcd6e2d3d57390de6a3b6f53b22ed36e1d\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d12220482cd47c760c70b5ec367a5d12ac741de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d12220482cd47c760c70b5ec367a5d12ac741de"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
