)]}'
{"id":"openstack%2Frequirements~681685","triplet_id":"openstack%2Frequirements~master~I906f67f9072bac991bb480bef485c374a7aeb61c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I906f67f9072bac991bb480bef485c374a7aeb61c","subject":"update constraint for python-masakariclient to new release 5.5.0","status":"MERGED","created":"2019-09-12 09:43:04.000000000","updated":"2019-09-14 13:51:30.000000000","submitted":"2019-09-14 13:49:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681685-1568468946928-c741f7bd","meta_rev_id":"4a88ec55f2603b2b28a277922eabf6f62b7086ef","_number":681685,"virtual_id_number":681685,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-09-14 13:49:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-13 13:49:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-09-14 12:07:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-09-13 15:09:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-09-14 08:59:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-13 13:49:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 13:49:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-13 15:09:38.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-09-14 08:59:30.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-09-14 12:07:34.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-09-14 13:49:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbaf969ba5e82287ac095cc9fcf9459687232a3b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-12 09:43:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33ab70529a9e3bc64ec3de85cd12b8f0d43584b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 20:33:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/784860ce5bff4fd29978d8945c42fd12 : SUCCESS in 3m 45s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b699aeba254d4e3880951b495dc053f0 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c726eaa6642d464ab757947d71378319 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aed2423186d74896bb099659861b5720 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a504493de58248d48967282f5f797091 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51fa2518e0ea4bda86f44cb8b1f79c84 : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36a090ef5b814cec85ef6acc0942bdf5 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1910882391664b3fa65206c6ccd8fa8d : SUCCESS in 2m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e1ecf6219fd64e37acd852c166c11660 : SUCCESS in 6m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d195bc0930ab412f87749861632fec40 : SUCCESS in 6m 39s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7621daf30b93453aa8c3004f31eb671d : SUCCESS in 8m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c273ce70001344e89c90e4efae4372a0 : SUCCESS in 8m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7a78c9de387a447ea2e1fd9c0bac721c : SUCCESS in 4m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f008ee71657c41e2832cb1ea4baf857b : FAILURE in 6m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/82a8780876a44d07b389b379d5ca63d4 : SUCCESS in 12m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ab103fdf20584f5da3a3f8addbd6ed25 : SUCCESS in 8m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fa482ce5d30c4da0a6586b743f955a68 : SUCCESS in 8m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b1efe98b4a254378b9789e80e83579e5 : SUCCESS in 40m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/de8f1fc69256431db2142b3a41b0f612 : SUCCESS in 34m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/62412e4c1d314e84b18772b0c9c48ac2 : SUCCESS in 34m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4629c8241e0947719c9f566e4ee4bd7c : SUCCESS in 23m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/177f19fd1ab2487b944c77ba49de086f : SUCCESS in 13m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d4eb55d0412f48239b45c2e300d51468 : SUCCESS in 18m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e0e517ef027e40a39a33913ea8de1d48 : SUCCESS in 12m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f5b671118aef442d98061c2b5390d0f5 : SUCCESS in 19m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e01e2f65617494980bfddf5ee498b79 : SUCCESS in 6m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/00185e6767894da9af28272d2a395dfe : SUCCESS in 8m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9d05cdea5e9d4aef9f1e106aa00aec65 : SUCCESS in 9m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4963338a994148d6af60f40ef165c5f3 : SUCCESS in 59m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ad9de6caa8c54a12862c16f76f58e3f2 : SUCCESS in 1h 19m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9774ed48543f46ccbfe1aa527fbafeb5 : POST_FAILURE in 1h 38m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3562f8cf45e641e79fa0e42770d4ab28 : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9e7fbea9a896c439a87a129768bd6ab70d9510fe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-12 23:26:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1c048650d759865c704f9d7c6b3deb3565afc57d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 07:23:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/337c170ced254f3087465aa0d8557e85 : SUCCESS in 3m 17s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5414ea2306d840d3bb79b0f998459671 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b3a0406c6a438e887922b8e84980b4 : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c39add43546428c84c7972a1372182f : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a19be4b17bf4c04a918250b1ebecd98 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a7155e735c44baf82f756042d421ff4 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a0a778a6b1142da843b8ef6f4e6996c : SUCCESS in 2m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53abe8a8af4a4122b3b87c35b9732b77 : SUCCESS in 2m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/edf569e0cdbd4a97a81074bfdc6b222f : SUCCESS in 7m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a1664d953b154b3780aa36413d90c526 : SUCCESS in 8m 11s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4667af3c8e1f44f7a381a251334b29c4 : SUCCESS in 10m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d723a46456a74358be8cfb5f65eb357b : SUCCESS in 7m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/649f25dfcf984255bf3a4b34cb9ca4b0 : SUCCESS in 4m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cd603668e36c479e8740abc47fc6f228 : SUCCESS in 4m 49s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4798d6e0a20d40689c3e445449b4734d : SUCCESS in 9m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/92576c59a880447e9d0103e1282f0ce7 : SUCCESS in 8m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99f42fa2bad04d26a4bb8a2f353e7d6f : SUCCESS in 8m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5cd29b103cf04ab79cf7690f2603cbc0 : SUCCESS in 43m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/05da6326809c4fc18812392fc9397286 : SUCCESS in 27m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2d569eed38b440b2ad03d472232240c3 : SUCCESS in 23m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/67c47915a00a43c49d3d82c188f77230 : SUCCESS in 22m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/cf7920f949be4551b208fc1f3fc8feff : SUCCESS in 12m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6fe45c101b4843638d71d9249b02c3a5 : SUCCESS in 11m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/650b0240f488495e9ca6cb230c95d46b : SUCCESS in 12m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c6346899ab8489fa7326968139cb945 : SUCCESS in 18m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05a908e5b75949b7a6a2f9d04997b551 : SUCCESS in 4m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ae2106f433624cdc8438c216ca2e74e5 : SUCCESS in 10m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5acf95a3e3ed4f779aa3ad4dbddab1f6 : SUCCESS in 9m 51s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0d3c4bda39ea439bbb16c269ea5b578d : SUCCESS in 56m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2021ba1c58134e5fae1c67ccd928349d : SUCCESS in 1h 14m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b94d3eff81fd46f0a4e3e83b8497efb6 : SUCCESS in 1h 37m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d52970d09471437699f8f59c0f6858b2 : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":1},{"id":"39aa38553ec06c78ced2386fe7f81b378aa6322b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-13 13:49:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e331714d65bd4be9acf2b03f330aa018d3c2db6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 13:50:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00403c028efd0f792a19e400d882c6b50cadaf7d","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-09-13 15:09:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ffe3bf5d98501df4bc374d7a145602d132676a9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 18:29:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd1ad7a48dbc458a89efbf2947821519 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83c03676fa89410382fc7d511075d157 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66a8f7a529cc4c418d23ff005fac6c6f : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a2b3ef1731246ecadf02506309433c4 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfcb01fe41ff4305a38df21da88b854b : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a67f7925f08d4eb98516f71a0b367542 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc45a9b4c70e43598c6370d1275c1192 : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/620eb302fd144cca91bb2eef547fde99 : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eca4b60ddaa94e8ca018afb07e48bba0 : SUCCESS in 7m 01s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3c178623f43e40fab4c74f5781af93b1 : FAILURE in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fc4df3d2952f401d92dfcbe77f4f9af4 : SUCCESS in 9m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e23c5925f6714a8ba1be8b1d30565f9f : SUCCESS in 6m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/787864b8d2bb43bcb0749826d9d0d746 : SUCCESS in 4m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fcd64c7746ee49ef82e3b7584cfc6251 : SUCCESS in 11m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2aec7b5c31d0461ca3d523fa4ce97ebb : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71c3769b874b444abedbbf0e9c9afac3 : SUCCESS in 7m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5f3951cdb22a4c50a4b8430e027fd4f7 : SUCCESS in 36m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a0e37a94d1c74a5e8feba62a00594825 : SUCCESS in 34m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/64d3636379d146968792d1394177b389 : SUCCESS in 23m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0d8937d5d5da4e3b93da543014b3b53c : SUCCESS in 23m 48s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/cd66e71beb644f45bbdc4d17a90f762d : FAILURE in 33m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/47f180790cba4a54b72b91b0b8a6f22a : SUCCESS in 13m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a2483a1689474ed0835594a38f6b8c68 : SUCCESS in 11m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/05bb1731457f4e039ce1e704fd19c7d3 : SUCCESS in 19m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5b43a3d8b85488eaada6e89241c05e3 : SUCCESS in 6m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/20c5f30125ee4e369fec5880ecb2f769 : SUCCESS in 9m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c4f4bac1d7d7448c9acc4e3a867ff47e : SUCCESS in 10m 25s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/404028d6f6394e10adfc2291405edccc : SUCCESS in 1h 04m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c87d46835224065bc97b5c04e308981 : SUCCESS in 1h 13m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/53bcf691308b46a9bbdb645346ec6b33 : SUCCESS in 1h 45m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/190c1a8d769e4b98aabf18a0bac72fc1 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b85af4adabd2f30ca1ee58877ea91d9517d331a7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-13 18:34:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22a91b19207ed6372ec86bdce289e3a403a0dd75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 03:27:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b44fb0bd108458ba9df8322caf9e188 : SUCCESS in 3m 49s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/58e2bff7e8564d6f98632e797d897fae : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a52bde2282a94aa59691972912876a6c : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2148af9268054f9a99371f9dada945e6 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/754ac565c69049208b1927fd722e289a : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/140060cf292242f290b752e9c28db04f : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d04443b46954c749927eb10e71a5a87 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b7f5997e1bc94e63b2aa04b8dbadb1d3 : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f3215a3142bd4eba90a51f39472d7d6b : SUCCESS in 9m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a31aa4a69ef04578b1f886c02cfd1fe6 : SUCCESS in 8m 50s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f94b240ef7d34b10b9025a9501ddd38e : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a0eb79f9e85f40f292b5068ac78c6a44 : SUCCESS in 8m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d0c5c47ba7594f9ca5e7f563e56b6596 : SUCCESS in 4m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/72a7052170414dc0af729eb007790ce9 : SUCCESS in 3m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7330163f30c4426188597455b6f6bb47 : SUCCESS in 10m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/898e35cca16b44609aa44efc2878256e : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bef80a46ab6f4d19baa7698fb93ff74d : SUCCESS in 6m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/09759d6c912044bbb2039d5c988cd332 : SUCCESS in 53m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dc08a74fcd784bbb83813c63cf9a50f2 : SUCCESS in 46m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/81c257b9186f4950819b51c1d5afe55f : SUCCESS in 24m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3d7a4803abd54691a951c7ea74f90fe1 : SUCCESS in 24m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e2567b375a5a4788b6d47261b9c726e3 : SUCCESS in 14m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5eb4da0cce3b4efd885c51ce28467102 : SUCCESS in 12m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ad8832815ad0429a81d642db371e08ed : SUCCESS in 12m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c37da585df24fffb8a1e577f7bc44ab : SUCCESS in 20m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c0f7593e95434393905ab1fe918bc0b1 : SUCCESS in 4m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/519d420792db42d68b3004d103869cc4 : SUCCESS in 7m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c24fcbde90f0460b96f4684f0b28d526 : SUCCESS in 9m 43s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/29f01fc92ba546beb1eb4d9e690325ff : SUCCESS in 57m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b33e3364c4474275894b9981205a3084 : SUCCESS in 1h 16m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4d1fdbb45e2f4c018ce15796dc6612ec : SUCCESS in 1h 53m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/23b5c0b6b1644778b4e848fdf006c113 : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ac40bca99020cd2694456a2a4a9b3b3833b81c03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 05:11:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d65773f0c7364b63b2b6eea327ff9e694c562562","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-09-14 08:59:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d5b67da5ab3702571d89958227be240a4e1b53a4","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-09-14 12:07:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e026b9fd441440f35444090e85fc39e7fb37fbc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 13:49:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bd0884c5f95a2a948ff47682fa7431851a55ddf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 13:49:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7940821b71d464ca57673f9176f5d83 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d8c7225d8054ad580fc2c04f901e27e : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0362e44826e646d6a6a8b2086c209356 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71bb947901ad4e4a841f78251efe8880 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2aade122e17f4ddca9a7c8816a38c639 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/28c6599150854ec3a2f3b00374aed2b8 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4fb42b1fb2e04fa89d11b241f0e55593 : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/69f4b6dfd8e844c4b8a5db1a3bce3db9 : SUCCESS in 8m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e3cff943a77346a2a3ca3dddbc897036 : SUCCESS in 7m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/11be14b02795476f9dd73887e13c5bf5 : SUCCESS in 9m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/64c2769832384ec993d1e2fb837f8784 : SUCCESS in 9m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5a44edc8279a4408afe1f6755e1c4e38 : SUCCESS in 5m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dd8444de89b34c7d9c521110112aeb00 : SUCCESS in 5m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e1e4219fa4cd45dea2c90e5c2de08ba9 : SUCCESS in 12m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/57df381521cf4bc0a92d6ccbe18087ab : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18b3427eb2d8408c8cdb29d068688f8e : SUCCESS in 9m 44s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/620547d12ac64c11bbcd304259d58d1d : SUCCESS in 56m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0a3ce10ac553442087a0e08b63bbbaee : SUCCESS in 37m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/750afed25d9c48a88833b2f1c5ad70d9 : SUCCESS in 31m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6241fc0233e048b4b0f0bb4fca086bac : SUCCESS in 23m 47s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c55da47a1f4348b8ae2ed36b67dc06fb : SUCCESS in 16m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/26611d010f6347d4982eb332207d12f8 : SUCCESS in 12m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a75415941e73438d88805b787ad5fe62 : SUCCESS in 12m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/02ddb47c06eb472c8aaa26396fc7c066 : SUCCESS in 16m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88463ca018c6427a888b3340d73f58de : SUCCESS in 7m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3b62f8f836db4b389311b570fca0b0f4 : SUCCESS in 9m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6f05303734b4449e89299538761fe0fb : SUCCESS in 9m 56s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5ea54929f9e74df992fbfa04d0957514 : SUCCESS in 50m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a6f7c9fdffe4daa89fc7a34fff8246a : SUCCESS in 1h 02m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0dbafd6e73814c43833dcac8d6ab1730 : SUCCESS in 1h 40m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9f85c85e7e6d44b8bee8a070d79f29db : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a45857aaf2693c98559ac4cdec653828457c6037","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 13:51:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b5797a16af9471bb14cb0016319bfd9 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f868096ec155918e52700e697956a12539921510","revisions":{"f868096ec155918e52700e697956a12539921510":{"kind":"REWORK","_number":1,"created":"2019-09-12 09:43:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/681685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/681685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/681685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/681685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/681685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/681685/1"}}},"commit":{"parents":[{"commit":"81deebb7be87df1a6c3cb4cf82e3bbf593e710de","subject":"Revert configparser update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81deebb7be87df1a6c3cb4cf82e3bbf593e710de"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 09:43:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 09:43:02.000000000","tz":0},"subject":"update constraint for python-masakariclient to new release 5.5.0","message":"update constraint for python-masakariclient to new release 5.5.0\n\nChange-Id: I906f67f9072bac991bb480bef485c374a7aeb61c\nmeta:version: 5.5.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Sampath Priyankara \u003csam47priya@gmail.com\u003e\nmeta:release:Commit: Sampath Priyankara \u003csam47priya@gmail.com\u003e\nmeta:release:Change-Id: Ic6b6283bbd2b068acfeb39615e44bc1926b29564\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f868096ec155918e52700e697956a12539921510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f868096ec155918e52700e697956a12539921510"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
