)]}'
{"id":"openstack%2Frequirements~681690","triplet_id":"openstack%2Frequirements~stable%2Frocky~Ib16fc0045faf2c64e70a434ed478291bd661c146","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"Ib16fc0045faf2c64e70a434ed478291bd661c146","subject":"update constraint for tripleo-common to new release 9.6.1","status":"MERGED","created":"2019-09-12 09:47:09.000000000","updated":"2019-09-13 10:55:10.000000000","submitted":"2019-09-13 10:55:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681690-1568372110285-e46e57b4","meta_rev_id":"320c60b30f0a75e826454752f104333192af3887","_number":681690,"virtual_id_number":681690,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-13 10:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-12 23:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-12 23:26:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 23:26:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-13 10:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41d1cf4ef0989ad970b35b549cbdced75df4101e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-12 09:47:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc8e343138623d7d25020453af16ad37e95632b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 20:28:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cf9456d40f844f49a23c2cb6f57bffbd : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ee5a6ce39649f58e36da7f7f897035 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e809c9f9779d49e0a5045adcd8ea7b52 : SUCCESS in 4m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6949e2de45d1487da9a364e89e4ec063 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3220405a3e8c464081031d6629664d8a : SUCCESS in 6m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8751f7f61e444bd18cfa7c34ca6375da : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a6b03cfb404e4303be1186ad37bd8462 : SUCCESS in 6m 36s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/884e0cb73889402297c1ce4178fa75da : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/702b048fae454203965e1e750f9c1e8f : SUCCESS in 9m 45s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e935fa82ad9147c1992ed2d77e249a8c : SUCCESS in 13m 33s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/2dd891a17ed24e40a2b3662220764a47 : SUCCESS in 9m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cdf1299949794da0a9b62e1217357bc1 : SUCCESS in 4m 10s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/cc18a4c65afe42c7b2afca431696ba87 : SUCCESS in 5m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/99a5a74143454d1fbb90bdf1bf40436d : SUCCESS in 11m 08s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/319a7db0b6f74567b26de4e4d93c99ab : SUCCESS in 9m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8cfa57fc37c14fd7823bbad2e4e43ecd : SUCCESS in 17m 05s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/88e86308a3054baaa4d4efb11949985d : SUCCESS in 15m 49s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e7cc6d9e84884e218f51c11b2152d272 : SUCCESS in 25m 38s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4b078de03b404e658ab4428e3b9220c7 : SUCCESS in 32m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a78ff6e62a5b4f1e839b48e5592922a2 : SUCCESS in 11m 30s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/dbd515af2fe441408ea242b3fc0fc6a1 : SUCCESS in 14m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b7d27734bc947e3abafbfb87669cc38 : SUCCESS in 21m 58s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/3956c0cf3d4c4def940a381991ad4887 : SUCCESS in 6m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8dc489b6db0d401ebba8e902e851fba5 : SUCCESS in 8m 20s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/d5b7e2b3b6aa460995cc5e3bb8055e31 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f22a163832af4294bcd65ce25e808245 : SUCCESS in 5m 25s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7ed600fb64d24422b29149e760777318 : SUCCESS in 58m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3880328902f46a1a7073beb21f0187a : SUCCESS in 1h 09m 05s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/a54702d8fa7a4837aa084f0e6ca66dc6 : SUCCESS in 1h 35m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/327014fced96439ab2d343e90ebd7bb3 : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":1},{"id":"33bb385f1c004678616ae9ace4e5f6b54075077e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-12 23:26:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90fa67a4795dbd45d385fcc5a0a5fd83e8676830","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 23:27:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c97cbc3aaa86ccc6cf91bd21590f925bcb6a455","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 01:24:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87f8eb605f04322967eaf31fcbef971 : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b5ecb69ccd543559d33e024d3fce2bc : SUCCESS in 3m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fce1746e52e540d4b8c20c69a0863098 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e224a517fe34a35b02c72bb83894126 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/356439b16bc84fecaf60df0a7ac03ff8 : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/82ea706899da43a285921a1641f5bf1d : SUCCESS in 26m 44s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3be268f18c0c40319cabe13741cb9982 : SUCCESS in 26m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/cf81712a3cc04d4e98e04e75d6fb29bf : SUCCESS in 10m 01s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/17d7b9c1bd7c46fb8ca598505fd900c2 : SUCCESS in 10m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/be4d9ab34cd847bb90c9630a1da6ff4f : SUCCESS in 4m 36s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/7302e1cc48f24d4bb7f9a025331f1dc8 : FAILURE in 7m 04s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/55e01620d36849c686c122aaacaee1d9 : SUCCESS in 10m 09s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a6a5e9180fcb45938fcec9447c816042 : SUCCESS in 8m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/86254c771bef457789d1d54998fdab99 : SUCCESS in 11m 30s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e2e62eb3603a4d268ee4dfcb97ad30f0 : SUCCESS in 12m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6664f7a584be4722ad04f63c187435e6 : SUCCESS in 25m 42s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/3b7e8f685a554aac865b2d877696d31e : SUCCESS in 29m 07s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e4c4aa3261124ccaad940a07bd146e71 : SUCCESS in 13m 02s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/32f2dd6a29b14ab299baf4073935a4e1 : SUCCESS in 13m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/753445c089c34d0cb56a1267dce321e1 : SUCCESS in 16m 48s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/559f4fe8740a4860847ca735feb0c710 : SUCCESS in 5m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7581d79866ef4b12a1b556ecfcca8a8a : SUCCESS in 10m 50s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/f265e01fd1af4064ad7ec112722e12f8 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/432e9ea94261413ea696eda985bc6e6d : SUCCESS in 3m 11s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/116822b6ca264eb9b6dd0659551055ca : SUCCESS in 58m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/73539fafcbce4b3ea83173ea570f1f86 : SUCCESS in 1h 13m 35s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/ee29d728039447c2bf44fcf284297d4c : SUCCESS in 1h 20m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c588c543d2004675b458af4ec391157b : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":1},{"id":"48b0b1f023f4ee4fdfb5140fa8f7562a38f03ed0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-13 01:28:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"49ab0d517caa096355b491e62cf841a8456d9ebe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 08:31:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/820fcb68986d46409dd938a8ae64fc6e : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c7f59fc2b564d06bbe9c50fe338eddf : SUCCESS in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e097aa905c34e0bb2c838cf7ccc1264 : SUCCESS in 3m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/37fd2a38cee941bb93c358ee100af07a : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f627dd7af124454ab22ce5e8bdc72d49 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6449220d7ff453d9ac2a58c0a17ce8e : SUCCESS in 2m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/50e39cb7cc564e66b987ccef31db14d3 : SUCCESS in 20m 18s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/2ebb570588104726a0ebcd1926a7cf9a : SUCCESS in 20m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0f45e17fbdad4f6ba4bfdf6692a0a6f8 : SUCCESS in 9m 20s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ea1cc8f149ca4ee3b658be1c1ae01ce8 : SUCCESS in 8m 27s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/29de4495a3b041008e52fc4dc5d271e1 : SUCCESS in 14m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6b22062f5a35475eb48831fa925dacf9 : SUCCESS in 5m 18s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/fc7249bc61e44d9299d1203ae10dcce4 : SUCCESS in 4m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fc17c4f250a446f3acf1e35e0a2ea35c : SUCCESS in 16m 40s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f8977e7164b7449391e2eef65223a54d : SUCCESS in 8m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7a0be08a30f04db09a58cf37344f11f9 : SUCCESS in 15m 16s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/2bc9b19b721c43158c07ce59b27835bd : SUCCESS in 12m 02s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6347494700574679bca4aab5192c7be4 : SUCCESS in 26m 07s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/050f7460cf804d59bb3c9613dc496cc2 : SUCCESS in 28m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dea340ecba0e47bda8753844b08319d9 : SUCCESS in 14m 01s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/f9a3f85ddc614f7db7160dd80bf1d878 : SUCCESS in 12m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c54b8aed07a48f3b8139c492544389b : SUCCESS in 15m 49s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/11b6c9396c4a45d8950383916fed1e26 : SUCCESS in 4m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ac31a608631d4059a1dc15e89b50eff8 : SUCCESS in 7m 31s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/56bc6a5076df497990ebbd310d57b0e6 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4765d06f75064f4e8535534af4f5da2c : SUCCESS in 3m 24s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0d40ba2e44c445eaba70656cfc2ac8f2 : SUCCESS in 56m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4482cd2f28154b28842aa3a292a0fc8c : SUCCESS in 1h 08m 32s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/f2beb45645d54259b374c5739b472c71 : SUCCESS in 1h 43m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b7eecc8150cc4f0191bbc73af11bb1ea : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e11927813ce85584918150996d8ba8e84bf60a77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 08:32:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7ed32d292aeef2fd974821296ccb06b2586b481","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 10:55:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1cfe2aea3b1423e992098cfb0fccb20 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58da52663b0f43aa963c7943af280bf4 : SUCCESS in 4m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/90154cb65c724feba59a8192055d8499 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac387188b33543eaa8b9570327b7d1dd : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4385e7345ffb4ae9bcce62f0da512144 : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b368a0252bcb40eeb95d348dc855d635 : SUCCESS in 6m 10s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/120bf0b7c8804167b2095b4dc2a287a4 : SUCCESS in 8m 15s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0ff5c74a622a443fae48bcc5a6d77a1f : SUCCESS in 8m 34s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/6f3ffb17354849aab652a7616e19d398 : SUCCESS in 10m 22s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0d9af267485243109e93f8016ea2e816 : SUCCESS in 4m 22s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2af003d3e24240a6842d6c5990d31f88 : SUCCESS in 4m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d7460969d12d4e4ab1fbd09a746a9bf8 : SUCCESS in 11m 08s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f18bd0ccb43f4d1289975bad189e1192 : SUCCESS in 8m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/145d2bf18f1444a29477d6e1d9aa93e4 : SUCCESS in 26m 23s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/fee47ad0e2ff4e37876f857c38d6658c : SUCCESS in 13m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ecc23d48de004f3f88c255875747733f : SUCCESS in 25m 54s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/eea11f21506d4b8ab7eaf7dce1230552 : SUCCESS in 27m 13s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7b5c32d962e040628ddb682be6d4cff9 : SUCCESS in 12m 38s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/05969f5cc07841048c114d82461c50aa : SUCCESS in 13m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c2242afb9b94c679fb40a89fad9abae : SUCCESS in 16m 55s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/b7dae498eda14d6d9b378a7bf0fce2de : SUCCESS in 5m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/52d9ed6b7aa847ebb668d2d003e08d63 : SUCCESS in 7m 17s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/f0df9129474e4c9f82dce3cb36308b4f : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c95859af42f40c3873769b45c466c19 : SUCCESS in 3m 10s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/634068adcbd94b6e982ade9d87048f7d : SUCCESS in 48m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d0222c0a4434d89bf1c0618401b59a2 : SUCCESS in 1h 04m 37s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/dcdbc1fbe8584740a87902d1bdb91c1d : SUCCESS in 1h 32m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c21a49d8313140138221033fff99ef8d : SUCCESS in 55m 17s","accounts_in_message":[],"_revision_number":1},{"id":"581d1373e6a65e3b01ddc052a9afe9195042ecd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 10:55:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de499276bc33312b411aa582f89898debfd6e498","revisions":{"de499276bc33312b411aa582f89898debfd6e498":{"kind":"REWORK","_number":1,"created":"2019-09-12 09:47:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/681690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/681690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/681690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/681690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/681690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/681690/1"}}},"commit":{"parents":[{"commit":"98ffe6f8b10bb92992df9201797e9516b7ebe0d0","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98ffe6f8b10bb92992df9201797e9516b7ebe0d0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 09:47:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 09:47:08.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 9.6.1","message":"update constraint for tripleo-common to new release 9.6.1\n\nChange-Id: Ib16fc0045faf2c64e70a434ed478291bd661c146\nmeta:version: 9.6.1\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: I20f966b80a5033429d42ac6f62085e3a086c19c3\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de499276bc33312b411aa582f89898debfd6e498"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de499276bc33312b411aa582f89898debfd6e498"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
