)]}'
{"id":"openstack%2Frequirements~681691","triplet_id":"openstack%2Frequirements~stable%2Frocky~I9be44d558ba752e334dc9d12169fd7cc695f79d3","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"I9be44d558ba752e334dc9d12169fd7cc695f79d3","subject":"update constraint for instack-undercloud to new release 9.5.1","status":"MERGED","created":"2019-09-12 09:47:41.000000000","updated":"2019-09-15 19:14:41.000000000","submitted":"2019-09-15 19:14:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681691-1568574881253-c741f7bd","meta_rev_id":"9ded21d3662f33af55d5068f218bb97f5f39f13c","_number":681691,"virtual_id_number":681691,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-09-15 19:14:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-12 23:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-09-14 11:46:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-09-13 15:11:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-09-14 08:58:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-12 23:26:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 23:26:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-13 15:11:24.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-09-14 08:58:03.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-09-14 11:46:40.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-09-15 19:14:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76a369378e014f85131d5f39937380ff1038083f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-12 09:47:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bad9312d7bcefdddfdb8460db5093442869e2354","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 20:52:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8aa9ca372ba140adaa4edc9fd150d170 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad93b5e06da7435db943003dd6a4eaa1 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ebd3d310e2247f6aa4011cc1f2de3b1 : SUCCESS in 3m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5323fdb0fe9b4cf2baaa67ad3ca4ec11 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90bf6aa7ce064fe0b111e24905196ff2 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8f01caa6957e4ac9b5b336c58d32b7f5 : SUCCESS in 4m 37s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8938d407a4544eb3a41ebd6a0c919dff : SUCCESS in 10m 31s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/192bbdb464104eafb1e7e243ca9ec7d1 : SUCCESS in 6m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/44846555ad0d4cd28a5f61460813d3dc : SUCCESS in 6m 19s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d4c6a3b7802f426883e726c15ce7860f : SUCCESS in 9m 02s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/bf5474845aba4f408cd02245421ecc81 : SUCCESS in 8m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7544338e3b5845b2a037affd68d99309 : SUCCESS in 4m 59s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d997729365c744cc953b37574c08cb40 : SUCCESS in 6m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7743442f92f04f77adb7f04dde8ca075 : SUCCESS in 13m 05s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/e93e056d46224d9681af89ba3da768d4 : SUCCESS in 7m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a357a3a091d64680a4cb57d719d033b4 : SUCCESS in 12m 32s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e6feedc5dad945b48fa6cde677f42939 : SUCCESS in 10m 47s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4132fd12bbde40f3b0699229420e2923 : SUCCESS in 37m 42s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/8db2202d186146ee8fa5f833e930e05a : SUCCESS in 24m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/67cf687a2d394abfbdb2c7aaa2dae480 : SUCCESS in 11m 36s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/008f506b773c472ca73c10422be5d3a9 : SUCCESS in 17m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b2a21ced92364d3ebc2bfbc045e162bb : SUCCESS in 18m 22s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/a44376c9648c48c0a3ebc89326437cf0 : SUCCESS in 5m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/09b31472645c4f079c44c4a4ea56bd84 : SUCCESS in 7m 53s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/7b5eee63c9334296b441e4101a0f4c6a : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a98d88b656dc46b6904f19fe979f8139 : SUCCESS in 3m 51s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/bdbec3a614ce4bddbefec29598d14edd : SUCCESS in 1h 01m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c570c80676342b4a9a573c1c770494b : SUCCESS in 1h 16m 29s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/0fca99cc3a9c4e29bdefc6c78e0d1636 : SUCCESS in 1h 31m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d68271dd83534500aa5e5d5fd8867322 : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":1},{"id":"53f05dd67dc7e6d6db47756966fc04dd359dd66f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-12 23:26:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eb39d247a2f8b88c1f18406942d29026ef9d46e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 23:27:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f24a9081dbde3e9c2f5e2650e9c1c944a71f3fef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 01:24:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e1066ed3704ca98ed69ad8a0fb33b4 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2afae5cc07a4c658129f4cacc1f2f37 : SUCCESS in 4m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/11c950a1cac8450e97324189cac52c4b : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/448ab795a16e46fe88c37f34d7921287 : SUCCESS in 5m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/41495b7ac2b04a99a1be4d39eec7edd2 : SUCCESS in 4m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8cc9733096754558a3a37e534233388b : SUCCESS in 7m 35s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/5c8c2ee9652945b2a24678139b5bd9c4 : SUCCESS in 7m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/462fa3816195455b893d5d73edaa9109 : SUCCESS in 12m 42s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9c9ad6defd174742b7bc599523c3ca5c : SUCCESS in 10m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c61045f3cdfb43a09fc336ebeca9380d : SUCCESS in 5m 06s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/85361259339542dda24e77c958ffa37e : SUCCESS in 5m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4d91c09382914c9497041960417e3f53 : SUCCESS in 10m 45s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/430cc3d938914f9b96a972a81336fcf5 : SUCCESS in 9m 28s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6de255abf85b4ec39446d25171ef1c28 : SUCCESS in 14m 39s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/d32391cc89f04b828b9e099cae7a64d7 : TIMED_OUT in 40m 57s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d525d3faf0f14a6a921977404285ba3a : SUCCESS in 22m 59s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/61d43124a7d147c29b2d5de94fb28262 : SUCCESS in 24m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7d7788f979324abcb7e72bb5fb26d242 : SUCCESS in 13m 28s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/38a564209f394e8a871f77ffac8c3175 : SUCCESS in 12m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d157d27b4cc4e9994ec2c2cced04700 : SUCCESS in 17m 52s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/133f35ded4454e6d8ebfa7ad46441a1d : SUCCESS in 5m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6a1b25e285cc40a1b75d298014a52fc0 : SUCCESS in 9m 22s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/1e9cf9a81398402fb5a72e3375cc709c : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f2d1adf44bf4beea29d57ece88e70ce : SUCCESS in 3m 32s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5f32817dc0f745dbbe08d7b5cb355e75 : SUCCESS in 1h 03m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8459386bb3744468b10532b575dba1ae : SUCCESS in 1h 13m 48s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/8eb2b764688a4333be3b0181b6c1167d : SUCCESS in 1h 31m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b45eb0e289d648ec89fa8f6598542ebe : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":1},{"id":"0321e8272e43a0bc76a7f8c68baa732a8eb163b9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-13 01:28:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22b009aab3b494d40640c54c95dc1d0ba915f3ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 09:51:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8043107720434442b22453ab44d194ec : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18463f534056424d81fc60b6c2cb0304 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d58155923ae41de8daac4b8f1c9c514 : SUCCESS in 3m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b89ca3efed1b4c34b7a2c925c182763d : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74398ec3a80a465887de6f0747bc528c : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9faea9e3fd6b4223991cb98403d2662f : SUCCESS in 2m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/98fd057fae3c478e88b477659ab8d9aa : SUCCESS in 6m 21s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f6c4d6cf502e438b94081d5d0afb4907 : SUCCESS in 6m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/952bd6390b634a3abb7d4822f2a741f1 : SUCCESS in 7m 06s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e3bdaa2c76f34bcb88e10df62a36a912 : SUCCESS in 10m 12s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9b940648790a432683182436f984beac : SUCCESS in 9m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/aa4c541995704bf6b511f243a536c298 : SUCCESS in 5m 58s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/fe61a7f1e6f64617b5a666ce79c022aa : SUCCESS in 7m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c4e4865258c845a3aa7619c31ffd3bbd : SUCCESS in 9m 21s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/229c056bd0544337ac5907a4f9c4a8e2 : SUCCESS in 10m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/41661eeef572440bbf65dfabe6eb85f0 : SUCCESS in 11m 11s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/32740184c42641eb90783026decf8311 : SUCCESS in 11m 13s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/931184fb7f63459cb512a115ffdbd1cd : SUCCESS in 26m 13s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/e48b4584d15944abad7679d607f263e3 : SUCCESS in 24m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/75d9faa007c946d4851a697d25169f74 : SUCCESS in 13m 38s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/4c2a6017f8c64196930d101f9a588d47 : SUCCESS in 12m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7cd03de1203e4238a6a9268b45c8d5c7 : SUCCESS in 18m 02s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/a5aa5920757d4a7fb4cffd8f2bba4f6e : SUCCESS in 5m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/77fb455d4d78401495ea3ff446d32ec2 : SUCCESS in 8m 26s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/53641bfc35f14a1f961a07280f289748 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d325fab9fd644d7aff58f0c64512b33 : SUCCESS in 3m 28s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5922368714484e1aa805064aa68628d2 : SUCCESS in 1h 06m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/69429f289d4643f99487423541ae94cb : SUCCESS in 1h 16m 07s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/339d91004d2b47a5ae78da433c12912f : SUCCESS in 1h 36m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7f9e78fd2e2a48f3aae4b19d0a73f47f : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":1},{"id":"08bb4407adfcb3cf74015d81447619554d9d57be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 09:52:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70e4e6013d6bb841d3efac901b48451890164e19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 11:27:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d10eaf15acd47cbb139825f36490d17 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39ccf08e3aba458f8ede82f1e3ac4dd7 : SUCCESS in 4m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/88bb0436bd8f4430af4268132e7c66c2 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62e83363f36f4ce88b30e68fa3c449cd : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a57f4b732287447ab87ceb586fa1adcb : SUCCESS in 3m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e9f50015f58d40af9156b4c0e6845fd7 : SUCCESS in 7m 32s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/0ea6c5bd085649a0ad2e83833cf7f76e : SUCCESS in 7m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/33733c6057eb45cf9408607a7bbce1e6 : SUCCESS in 10m 50s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/7c0b1f7cd9b04a70803eca3800094db0 : SUCCESS in 10m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/dc82f3468d5a478da71baa6a31366adc : SUCCESS in 6m 48s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/efac758049424486b3d8173bbecb1dfb : SUCCESS in 7m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/36c98e089d7045c2aaf10c2dcb7dd766 : SUCCESS in 10m 35s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/8aa992eb1dac47059f0e11aa52e1bdaa : SUCCESS in 9m 35s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2d710d45b6544324923466455d2af286 : SUCCESS in 13m 40s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/a9af92a4678c4d1382ec3a37c06bae6c : SUCCESS in 16m 08s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/94e91dea2fa04eaa896aff0674e267e8 : SUCCESS in 24m 53s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/0a4316abaf2a4752b74d54bd5174676d : SUCCESS in 23m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e13ba6f5c3b44ec597d7e605bf3d89e2 : FAILURE in 25m 04s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/3b912fdbe1c7445d9185ca967c1aec96 : SUCCESS in 14m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/849c1cf96e87490c9399535141a4d02d : SUCCESS in 16m 24s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/2ed4e006a66348cf8b0ce5996a8a80f3 : SUCCESS in 5m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a931b70ff1fc4bf184d48a70fe902663 : SUCCESS in 9m 12s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/91b7bfbc39814f43a3012b53a581921e : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2df43ff3d4b948a684fce8fecf6fda61 : SUCCESS in 3m 15s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/b023149d2ed4467ebc846070ffcf1032 : SUCCESS in 1h 00m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/60b1125f91724a61b678736294036ffc : SUCCESS in 1h 09m 02s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/73bf035532334a15aecd71b7ea06699f : SUCCESS in 1h 27m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/056b3a8ebf454c44b85218905e2c7665 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":1},{"id":"97aef1269c0786a929f01b22d279e1a3dba888b1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-13 13:50:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f795a99e0626ae92b7509dbfb50279a860e6f00d","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-09-13 15:11:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5429bcee8b1a96f910d59a5a835702515df79a01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 00:03:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/912dcbfa4e2346a4928ffb1c27bd6823 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e149d323ad8466fb21755fa175dc57e : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5373a4fcac3d46e49a436476da446c06 : SUCCESS in 3m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/027fef5d63b44f5ebfd7b8bb922fe651 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09ff0d1e4f724a9a8d42daf0ccf49460 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9152ec9057be4f95895234147c8b6de7 : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/33a6cfb83b764bfe97c503057acd652a : SUCCESS in 6m 15s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/4602f07c61104edabdfd4e7ca20ed948 : SUCCESS in 6m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/55c93e54570440aaa9d76e86d1fc1f12 : SUCCESS in 6m 32s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7a560e4c83684b6994b5f9f8ba013c17 : SUCCESS in 11m 52s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/ce2a0436179d47a0ba0fdf02642e534b : SUCCESS in 8m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/07aca81ec48d49bb8039ccebfb70aee6 : SUCCESS in 5m 27s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/65e9512bbba54433bf3f83605ec0d0ac : SUCCESS in 4m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/aa7df58e018e41479b208e9bcc05adda : SUCCESS in 10m 30s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/850c7d6515dd4ac4969d7dc8c2a24bcc : SUCCESS in 9m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f3fa78f401454f8091fc835eea081a33 : TIMED_OUT in 41m 22s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/bafa6e734c744ff9a48f4ea2ae95c587 : SUCCESS in 12m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e3da43781376478db9b991d87c8a60f0 : SUCCESS in 29m 17s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/7573858f579343c68602de92f746e1d2 : SUCCESS in 24m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5a37da42fd174e5daae28ebd6ebabe80 : SUCCESS in 12m 12s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/714bd7f856a846aeb6e9949ae39d406f : SUCCESS in 13m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d73220e7a76e46cd8272b300ae538aaa : SUCCESS in 22m 08s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/0dc8f864b77e4e7d836da910c51895df : SUCCESS in 5m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/40186555a7df44e38d508b30fe11b17b : SUCCESS in 7m 03s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/81943a04adbf471299c30d9997fde7de : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a11c103dd85c4b8ca2a3f824af7cf729 : SUCCESS in 2m 57s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c424352e65024e8e9341e474ffd33388 : SUCCESS in 53m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9e85f848e774b41a996fee1140d695f : FAILURE in 12m 15s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e4e921978dcd40edada30d548fbec115 : SUCCESS in 1h 36m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6cccafd747084b969a2d6f3fcc97b98c : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a8a641072b5f40d1b653e3bfdc63346d8b8629ff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-14 00:15:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1a2ceb89fb005ef75203a22ee7bd3f63e93510da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 05:15:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4d5b12aea54e4ba584e3ad8a1b10ab4e : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd30941e57a49f6a1c2670006dfbe07 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a332cd596d3748f9af820b905343dd84 : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/564fd53f0334454d80c6e56e680e10be : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5d68a4d43f647719bd2225da6c26ed1 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3185ae60df6e44c29a598c34a4bf331f : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5974623d8e8c4863a97a75c3946b8cc8 : SUCCESS in 28m 07s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/1e4a0ac21b8e42569d317050cd430e48 : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/59bfbec9ba05432a8885a03d05f557e2 : SUCCESS in 6m 15s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c3f8294f711443e097ee759404b37e6a : SUCCESS in 8m 51s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/6ea1180188774f66af0b4efed6c85d04 : SUCCESS in 9m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c680629c403149259dc3e129e58a883e : SUCCESS in 4m 19s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/6442ef948546412b9b44766428ae27bb : SUCCESS in 4m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/58b2b775feba43a6aff48d59db287de4 : SUCCESS in 10m 55s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/bc0bed0d03884b72839b5ebc7153a57b : SUCCESS in 10m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2fd9302dcccc40abbbd6a9128b39bc94 : SUCCESS in 19m 49s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/ae3aad1c272c467aab6bb7387344f6d6 : SUCCESS in 18m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ba695caddcf94d7d84a2fbe8f898c847 : SUCCESS in 39m 01s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/8ffcaddd3261453c80cc181161d9f38e : SUCCESS in 40m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/58ba356a1eec4e73a03fdd24a315bf8e : SUCCESS in 19m 45s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/afeea182bcec4d948fbeaee2c6971589 : SUCCESS in 20m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc69c30b83cd4038bd63db6bbebfc698 : SUCCESS in 20m 39s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/1b2eb3a8df8242959a77acc594e79350 : SUCCESS in 4m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c44acd4cdc76467b8342168ec7e13ae9 : SUCCESS in 6m 35s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/af03934258fa44a9968aa797fe2e3d82 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e45e6dfae6d47b9801923a537699db2 : SUCCESS in 3m 06s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3ccf4c4f2de044caa7500ead5b11ca5d : FAILURE in 37m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8502f0488c614972a691806c3b599609 : SUCCESS in 1h 08m 04s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/24e962cd72724bf48144ff08d4e7764d : SUCCESS in 1h 36m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f0d394e476eb4f8c8716bcb1e3b26adb : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5ff6c52993a5261e26618931a8e293a9140e0eb0","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-09-14 08:58:03.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7e2a7040208769759edd6994c27251e752265557","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 10:59:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6c1377ba64b14c19a1b2f0b21e7f67ab : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1909b97676244c6b4b71c84e352b006 : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1fb3b20c07144018b59d97c14718135d : SUCCESS in 4m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e0176a310b09401a8d83e252ac106831 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f22e410881be41359fdb505088ee70b6 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/839c3d3807fc4db1843475023ee23208 : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e3ae4bdb00cd4210ab37fc807cbe6ff3 : SUCCESS in 6m 31s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/490c9f866f214d50b4d580b5b9878354 : SUCCESS in 11m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b892e7247ccf4f29b43c62e7a66360f8 : SUCCESS in 6m 51s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c7030abfb2134be5b9677c2671e2119f : SUCCESS in 12m 18s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/d95f2c2a4d2b4b4c8a65117745ca6c7a : SUCCESS in 11m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9853413a82ba4e5ab8314e30555a0211 : SUCCESS in 7m 45s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/0d2ce1caa3a04dbdba7d8f5120a9904f : SUCCESS in 4m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/40703f8538d54de8b45b1f2ac60d5871 : SUCCESS in 10m 47s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/ea4335d49ea44fd6b3242fdd2bcec1f1 : SUCCESS in 9m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/31eaee8064d54175b3e36902385dfc7c : SUCCESS in 13m 59s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/0ae26bde3ca94ec4962c88ad6628641a : SUCCESS in 11m 54s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1a590d129b9d4b88ab1a0b0167400071 : SUCCESS in 23m 09s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/7093197f94aa4295ae7a77513ce5c0f9 : SUCCESS in 23m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/13e374a84dd846a3aa79d7acd9a606d0 : SUCCESS in 13m 36s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/62a7daab5bde4195be7e673fe6b395d0 : SUCCESS in 12m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d4bb589828a41268dc677d7e1a278cb : SUCCESS in 16m 49s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/d5b81765f71b4aa3acbc2fad782b40c1 : SUCCESS in 4m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/13b783c752764e6a96a21fcf3806495d : SUCCESS in 8m 15s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/34927a88b7be46e089e1e31728815733 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/addcd9d5c5c84fd69252175972d95a6e : SUCCESS in 4m 32s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5fa424fa23f24ae898083553a3582c05 : FAILURE in 36m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/965f36d38ce5440186f28ea2a9dc85ac : SUCCESS in 1h 06m 05s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d2eb1497d43f41ddaacb44dae6672b15 : SUCCESS in 1h 30m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/39f7b88950e74c15a2d2b1d394ba91fa : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"dfe5c0c44cd97b8013982733abc699aa9e37dacc","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-09-14 11:46:40.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"93a77e69026256314f104ef9bac0e635da049c56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 13:21:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1bc3edd88fbc4d17b00c4835958b6dce : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815ac3bd460d48d087f0cce14479d1b9 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41c87bf0111b4d53b58c098c3fdee95c : SUCCESS in 4m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cbec4a70ef354f33bd106c7418b5ffa6 : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8d42dfc5fe84ee3bf50f04149f02a7b : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00f92e1f8a2d447599db5b08fc20a877 : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ae940070ffe4477997515d220a2c55d0 : SUCCESS in 5m 20s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/577a5c53ac414c4a8cd4d4e571e08e1e : SUCCESS in 6m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e8baf1287a394a3db965745e6fb6edfb : SUCCESS in 5m 38s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ecc48697e4fc4ef2b78c074aa13e337f : SUCCESS in 7m 41s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/0e26a79aa119430e82d2884356682735 : SUCCESS in 8m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/34e5e2502a594494be06c4a7b8ccd514 : SUCCESS in 6m 04s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/e4a9881e88d44c5283bcebcbd89a9f6d : SUCCESS in 4m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a27f8f97e3564360a3cf41d6988f28c5 : SUCCESS in 11m 50s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/06cc44bc6bc94d7188d3ca002831b939 : SUCCESS in 9m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d0ccd59d873342b49aa0c48b15f078b7 : SUCCESS in 11m 56s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/d16e709176754ce6bc906fbc4caf7a59 : SUCCESS in 13m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8fd67d41651041f594c16fe20e118539 : TIMED_OUT in 41m 01s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4375dcc5b966483bbf9f88056eef8390 : SUCCESS in 21m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/899a588548a3431cb505a40a52c71f43 : SUCCESS in 20m 12s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/16e183c3ed0b406ba3f40680dde677b0 : SUCCESS in 14m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/efb55a496e7a4978996cbca6147cefc0 : SUCCESS in 18m 48s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/7a2c5140867c474fba814ea7791ddba6 : SUCCESS in 4m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8e596dfeb9d846cda4b8bb23d00d30ec : SUCCESS in 8m 29s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/08e0194b7c0e4b6687f4a6013dda5e8f : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0b456252c6841c3ab21a19515007f2f : SUCCESS in 3m 10s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/08468078547d4c089002fe8022a2fc6a : SUCCESS in 1h 03m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d4f288661654eb88a1ca8314e48aaec : SUCCESS in 1h 13m 52s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/3f9b121634d04023ac5955c3a9081a67 : SUCCESS in 1h 28m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4df299dfe23b48a4b1a0756ba41d6071 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"bc514f5667b25d6e7eaf46466a43859d3ddadda1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 01:33:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"08062b20cd0e283ee9029118493a94609c453f57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 03:06:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c4c8b679b2ef4f688dfb529b84c30feb : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be74e66b21044ba394dd2c0f9e203123 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0ce33d2be4940e7ad59b0d9919fb7a8 : SUCCESS in 4m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/208101072c84439caf76c57e40c4379d : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11665590d11444448b808b479ad6041d : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e57d3bd578c48af84d463e3df891f24 : SUCCESS in 4m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/51343256b8da4547927633e990a0e3fa : SUCCESS in 8m 10s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/c5ce60fbec93405389826b5171539f0a : SUCCESS in 7m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4eeedcd3d7614cd2aaeb08938edad838 : SUCCESS in 7m 22s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a327b4dba069404ca21937a88ce5302e : SUCCESS in 9m 53s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/42aad5fce5574e1794bc553f62455324 : SUCCESS in 10m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/029b08bc5314466e986fd9b40b25fa7f : SUCCESS in 5m 13s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/7fe01adb87124f3483770d270986ddfb : SUCCESS in 6m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3ccbe645025841e0b824272b97f7c940 : SUCCESS in 11m 33s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/d5ebf883fa0b44c0918f258a266e06c8 : SUCCESS in 9m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7877aa6f24234f1aa3ee2c1d4d6a09ad : SUCCESS in 13m 06s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/f6f6bedb692b400a954bb48c60e0fb3f : SUCCESS in 14m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5c9e4ee702b443dd82b4b250f11724be : SUCCESS in 23m 58s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/6ad77626918c44dcad812281d1672e6b : SUCCESS in 25m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b9e84153e8ab4945aff5dbed7f252982 : SUCCESS in 13m 54s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/c15b2e9898f34dd18e8fb150b4ff0b43 : SUCCESS in 15m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/93bad45c67344ea4a1c10d3dcc28b628 : SUCCESS in 21m 23s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/f6f369c863ea4185a19495fc43b38e55 : SUCCESS in 6m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f806e51c5b2846939ada199e09940575 : SUCCESS in 8m 59s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/55b09b9089b04630937d4643a0361be4 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5f568e21bd64ca0955aff2cdef5f385 : SUCCESS in 5m 47s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/555671618e44458bb6f1b378cd444813 : SUCCESS in 1h 00m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/77e774ebc0794565a5063a818ba8416c : SUCCESS in 1h 06m 07s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e367dc50e1094467b5b2658ed4453b0e : SUCCESS in 1h 21m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e9bb9f7066ba493fab7d89aa3c7f4401 : SUCCESS in 57m 10s","accounts_in_message":[],"_revision_number":1},{"id":"af38b11575a0f39c5604493902ffbb05500d8428","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 06:35:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c8af24e705a3743b719e03609d8046b6598cbb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 09:54:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f069a74f0eab4fd2b1230c0c58bd9137 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a89e0157b04d4336b020c178495ffa43 : SUCCESS in 4m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8521c0901ffb4d8b95a90f2236edf359 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91b902c23b664a59a2e5c3a4ea4ceee8 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc9b4733dcfc4615b722cc72cbb9a076 : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/36ef00478f2941098773e24390a6f788 : SUCCESS in 7m 04s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d139cb76ed684b6f8413026a9e4c331e : SUCCESS in 6m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6d18d1455ca74a518d5b415aff0d3b46 : FAILURE in 11m 53s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/e72df871753949a09d0f879a2edf2001 : TIMED_OUT in 41m 05s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8823d94aacd64d20b1adde91797fe90a : SUCCESS in 6m 43s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2fa81cbee33b4d0783f26a72ece5dd0a : SUCCESS in 7m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bf05081af31e49119f50f08a0ad0b03b : SUCCESS in 9m 54s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a145f0e88ab04cfdb38600a48b796332 : SUCCESS in 8m 33s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ced4b931f2ba431facca3617a1b4ac8b : TIMED_OUT in 40m 49s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/67c0bff3e6114515a124c95389ba5ffd : TIMED_OUT in 41m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/961964c9aa32430287ee2e39866ea37e : SUCCESS in 23m 34s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/02e6a60f39634a7f82ae2255e71d6d02 : SUCCESS in 27m 45s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6f79512a511845629e5da112a365bdc5 : SUCCESS in 14m 32s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/9eaf567b13e14fb180d28183e4ad49b3 : SUCCESS in 14m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0bcdfcbc55c84fc3af3efbbd17efc5d9 : SUCCESS in 18m 53s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/67ba74c418394e20b95947f2c45b396d : SUCCESS in 7m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/10ef48fdac02467caf14bff8189e2da9 : SUCCESS in 9m 37s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b873a2c98b6249628ae0d5e56668335b : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b07b7a41f9734a9695caead925e7f2df : SUCCESS in 4m 55s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e60213479b7e466c9f5b42c0706b02d3 : SUCCESS in 58m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9bcff802bd740379e1170ecc3ddcebf : SUCCESS in 1h 09m 44s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/8842d2913f36402da1d53eeeef7f1016 : SUCCESS in 1h 39m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ed5ebf3ee3924fe8afb0618bf13873a9 : SUCCESS in 1h 07m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e4391a50a0c88b2c884a7faa988b8fe9a57ffa6c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 15:26:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e056221be3d0471a4f0d759d684e8ac1cca2ba8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 17:19:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b0691537915a41c5b737d1f680f77c85 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f21923eb00e4a01a308660bf1e0f5be : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5273178a840047809b05b94b89194bee : SUCCESS in 6m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0d7a5f568c954fd1aa12fca2a0e469cc : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97df217fa0ec49979633dc558e9a0873 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/203da6d463974a37bb6bbb65b71c9046 : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/17c22d3c369c4590941d69a3f9123aad : SUCCESS in 29m 46s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/bd6d0b6b50c64a9382544a8f84167208 : SUCCESS in 29m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fbfea450e14741bd9bdf58d50f43e293 : SUCCESS in 10m 16s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9f1ca0e6be2849dd9ae14121758ce963 : SUCCESS in 11m 24s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/40ca8d99f2ce4cd9932ed3cf7fcc875d : SUCCESS in 9m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/93b8c30d86c046bdb300872c63b5597f : SUCCESS in 6m 51s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b90220455c364ec49bc273a37f88bf6e : SUCCESS in 3m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/809bb1cc569c41caa94b316f69eb0d8b : SUCCESS in 13m 42s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/9474b05e85e54d79a2106c30f0474cae : SUCCESS in 9m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/87d6a91a6278408ab12cd98ed96f64cf : SUCCESS in 13m 46s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b11dcc2209d84317b94cffc96be4b538 : SUCCESS in 12m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/841ba3865d7f47a0817185a6736b031f : SUCCESS in 26m 05s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/8176744f36884805b4f74156e509f77b : SUCCESS in 22m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/54cdb19a0d094248865a346a7e78c4b3 : SUCCESS in 12m 24s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/524e258d910c4e98b3a5ff305cdee8e6 : SUCCESS in 16m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a6481b4c66fe4c0cab113801adcc133a : SUCCESS in 20m 58s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/12392a9db35543008d1aa65b71515ad4 : SUCCESS in 4m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5d0807bdd926484e9f8443584d5c1f7c : SUCCESS in 9m 39s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/c30fe13a0f4b4648a86de515abd0c5ab : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47d96186f22e4ef28303b0bc686c6716 : SUCCESS in 3m 26s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/493b0a58f3e840d7afd1c2c2a78e449a : SUCCESS in 52m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3356dc2f4ac44b44935ac04f8a933bca : SUCCESS in 1h 02m 34s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/162b4a9e702b4413b0ccfaef460bb69b : SUCCESS in 1h 48m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6523692d6a754c4992d0bc739cff3a86 : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":1},{"id":"662f79e14c5e8fd09225c871cbf725c1adcd919c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 17:19:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dde02c545e8f79be333ad9a22282aab4529bf1e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 19:14:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05ebd81e4bee40ea9cd8e21315ef06b5 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fcc69a391aa84c60ab9337030026f1b3 : SUCCESS in 3m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/147934c0dba347d5b9a1c4698dc07d24 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/843c2b95b17e47a8a3f25f3bb3219f15 : SUCCESS in 2m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/147be5f35e0d480b873c38957e23e6c9 : SUCCESS in 2m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b86e60b127704b749879ce9f303b4cbf : SUCCESS in 6m 05s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/85056df331a44f1e8b27c105b04468b4 : SUCCESS in 11m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f9bcbdde93014307b3daa3b8f9e3de46 : SUCCESS in 8m 41s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/845b3c3d74d042fb984b8d7aea496d75 : SUCCESS in 11m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/74360e690107435389dcbb0d13bf873f : SUCCESS in 3m 52s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/bf8d20fe209547248dd300de84fe6128 : SUCCESS in 6m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/465df07837ae412f9c29448e6ce2f8fe : SUCCESS in 9m 39s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/cef5d6f8a9784efb84f8b569af05e008 : SUCCESS in 11m 40s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/79282265ff5743e39d88794109e6c14d : SUCCESS in 12m 13s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/f0eb31897938465e9cac059e907662fd : SUCCESS in 12m 42s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8f43b6ecabca4bb2a9bd433cc4419050 : SUCCESS in 28m 36s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/7026c305a5624ecabf73ec9103d13b81 : SUCCESS in 23m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e70bda13b1f34b2b839d217d09525566 : SUCCESS in 12m 14s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/04484de6030242b193f5e23fbbddfea9 : SUCCESS in 12m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ed59d2347c304c1691f9a04d02763b39 : SUCCESS in 19m 47s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/1d9541bcc0d341a49c53ccfda6d84bb0 : SUCCESS in 4m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dc966eef2a1945abbcf6b0c895b41bc9 : SUCCESS in 7m 31s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/dccd1998200c40d48ec805bb1f51087d : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae8b16b9a1a7434c8b0c1d6c8e425aff : SUCCESS in 2m 56s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/90aa8c9844cf4b528d9cd882fcaa7e1d : SUCCESS in 58m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd96c5a031cb4d07bce8a59fe5d81611 : SUCCESS in 1h 13m 34s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/4708320b479845f88739677fd7f14902 : SUCCESS in 1h 25m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0cf0bac644b94caf885423f18e8f6a01 : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4d9803662c3cea94b57d949dbbbe6cfdbe87cd47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 19:14:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f52789326a46563ee53db70b5a4b0b3177a84448","revisions":{"f52789326a46563ee53db70b5a4b0b3177a84448":{"kind":"REWORK","_number":1,"created":"2019-09-12 09:47:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/681691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/681691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/681691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/681691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/681691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/681691/1"}}},"commit":{"parents":[{"commit":"98ffe6f8b10bb92992df9201797e9516b7ebe0d0","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98ffe6f8b10bb92992df9201797e9516b7ebe0d0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 09:47:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 09:47:39.000000000","tz":0},"subject":"update constraint for instack-undercloud to new release 9.5.1","message":"update constraint for instack-undercloud to new release 9.5.1\n\nChange-Id: I9be44d558ba752e334dc9d12169fd7cc695f79d3\nmeta:version: 9.5.1\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: I20f966b80a5033429d42ac6f62085e3a086c19c3\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f52789326a46563ee53db70b5a4b0b3177a84448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f52789326a46563ee53db70b5a4b0b3177a84448"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
