)]}'
{"id":"openstack%2Frequirements~681692","triplet_id":"openstack%2Frequirements~master~Ide1889c08ecbe0232aba35702b3aa7e28fcc2d77","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ide1889c08ecbe0232aba35702b3aa7e28fcc2d77","subject":"update constraint for python-openstackclient to new release 4.0.0","status":"MERGED","created":"2019-09-12 09:48:27.000000000","updated":"2019-09-14 16:40:13.000000000","submitted":"2019-09-14 16:37:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681692-1568479072711-b56adfd4","meta_rev_id":"b4200301728bca5ac69f0dc21f9e05682e04014a","_number":681692,"virtual_id_number":681692,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-09-14 16:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-13 13:50:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-09-14 11:49:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-09-13 15:09:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-09-14 08:56:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-13 13:50:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 13:50:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-13 15:09:48.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-09-14 08:56:35.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-09-14 11:49:13.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-09-14 16:37:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3eaf71bf9310b6ae3a6180196cc17496c356aa46","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-12 09:48:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bbe0a245fdbbb969f654ba1f6be94680073e408","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 21:10:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33a60216d227467e958467b7ac5db626 : SUCCESS in 3m 50s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1646e7cfdaef4ef5b55864a80b80662f : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c7194f948264285a0c1de762fe45d88 : SUCCESS in 2m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3c1d22ed9da4b1b9eda2612c21add90 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/125b7f44f9034d498667d6a213271848 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a43ca65dcfee411e98e573d3ebd036f4 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b1b474fa5bb4c54ad0cdeadcbd146fe : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ed2480df6374800af2e2f18e1b6b12e : SUCCESS in 3m 05s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/957f6dc474cd436e8a312a8fd3dbd052 : SUCCESS in 6m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/482066f50b4d442baebdb46aa7f90ca8 : SUCCESS in 7m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e67e42b97c3d4e0b837ff569169b028d : SUCCESS in 8m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/88595d9d33d84cb2b24f2681d7a36aab : SUCCESS in 10m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c6feefcff82d4ccc805684eb6f227d07 : SUCCESS in 5m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3eba82d7256747a684ca0ea523e4e3e5 : SUCCESS in 5m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4768b525be82437baa0df510ab8d031f : SUCCESS in 11m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/59af1f8800f64bcb9592669e1b328815 : SUCCESS in 9m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a10c631457614b5fad7af026fa6d55d7 : SUCCESS in 7m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a0d81bbb26cc42f78c3ef13d3bd0d532 : SUCCESS in 32m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b179b9054a32444896c6562e58d608a5 : SUCCESS in 34m 49s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1c04fb0c0ffc4caead87364db33d3039 : SUCCESS in 27m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ede19b8c2aab47cf9153484412078247 : SUCCESS in 22m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ea7f5cac88924e82bc8d8d000f9b7a3c : SUCCESS in 13m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9ecf112a7c0e40bc98fc1cd62d2fe689 : SUCCESS in 13m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5d9d06de4b484bc59ab61831861545d2 : SUCCESS in 12m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ff5c9fd05d54958929cf3a45bc05200 : SUCCESS in 22m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/acefe7eec4654d5fb6aff47afb2d7b88 : SUCCESS in 6m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/504b05691e9b4d32869282dcba21583a : SUCCESS in 8m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b669ac35a5154aa1a5954800c0b8a08d : SUCCESS in 8m 50s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0d2f9a1b871a4566809d19dd6084e2c0 : SUCCESS in 1h 01m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c087319749a4389924baaa0983933e0 : SUCCESS in 1h 20m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/267c87bc5fc5428e8e623b88fb4bbda5 : TIMED_OUT in 2h 05m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/41a0af57506c487db54669099806f257 : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":1},{"id":"850ed952632834bd6a27a5790390f4ef76157c05","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-12 23:26:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8d9e09e2bf34a1cfe9f52f8993c64c79feaf4d13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 07:30:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a4219823d7e4955a6f9280d76f01f34 : SUCCESS in 3m 45s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/dfb5509a6f5a41b391af0259a2710f47 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ee351d5c424bfea5b896333a3be68b : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27ec39b53ec2456993150fca92ed9d75 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce2fad21508a4f2b864ad1177d334ec3 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f18d4566ac8d49b3bbe9e79625554275 : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c54518c14404431819781fe492f1494 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ad157d5bb6ea4ec39f17a8c18a7fcf7f : SUCCESS in 3m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/91d7020ca41d4ac584e1e1bd2fa79244 : SUCCESS in 6m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1587e28e0e664c9aaad42ca8e5c11706 : SUCCESS in 7m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/43e4019a5e084c928e43c1e11f485f32 : SUCCESS in 11m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8d2a26c0035b48b1a401effe275264ab : SUCCESS in 8m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/41b18e9ca85b41548a285364eeabca5e : SUCCESS in 4m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/154f8e95917045d6b86a9285fca58bbc : SUCCESS in 5m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9477dee0180b4d5d9cbad8f4ca9edcd6 : SUCCESS in 11m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3d5059de9f9a412cadcc758bbd15157c : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a2d129dc05e4459ca425b336d0a214e0 : SUCCESS in 6m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/dc92312f07e943f68c694e39bf525ce7 : SUCCESS in 34m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e7c13beec20a4239aa567e38f9cd3298 : SUCCESS in 33m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/358785fb3f40447dbb4da1f5f7ac48f4 : SUCCESS in 23m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f380c762276c4c44a7a76d40a5439569 : SUCCESS in 22m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9ff6b032fcb9418fa2e0ad4784d38bab : SUCCESS in 16m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/be8be6e67df84380b5f55bd96ee98aa6 : SUCCESS in 15m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/00b7518574e045f6b7d240bbe32f1de8 : SUCCESS in 13m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e60692677c804e7b99975a1f15acf297 : SUCCESS in 23m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6585b4613b74fae9ffa4f8c02fdc94f : SUCCESS in 5m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8e0c241a496b419da983b2ac6ad52220 : SUCCESS in 14m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/23e38edceb4445bf9dc220b368bb657a : SUCCESS in 8m 43s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5489b513f62d40f29e4c06a9adff767b : SUCCESS in 54m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/416af8c0b36148b9a98915d44cfd7b59 : SUCCESS in 1h 09m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1820bade5f9b450e831ca5aeef8f2cac : SUCCESS in 1h 43m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/703b1c1b2581406791d791dd04989146 : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b05f8bc45aa656a6de36ad76a7110cb12b2b41f5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-13 13:50:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ad8e1fc2c50a06ae2b8819e9b5e19c67b2d5397","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 13:50:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee37bf546595dd8df5664c930451625ba14c6b79","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-09-13 15:09:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9670124149475775f213bceb0585c36a3f00caa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 18:37:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/934111f9ece1442e84ebd2078279e393 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46ac9cda3f764733a6f98edec96ccf0e : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c2419f3be974341b56b87434213c30b : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6783e6c039142989f9a16ee8a7763d5 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5138b55af2624fc8891325b2d850b0ee : SUCCESS in 6m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a6dc835bafc44c879728f1c7db4ac9ba : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dce069f0b8d94eee844f957ffe83a7ab : SUCCESS in 4m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1cf81b7dd9474847a34f9e2f6d9d62b9 : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39629554ad4647a2a195706b68d8daf8 : SUCCESS in 6m 43s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/55935d35b02c436fb6db847c885e6069 : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b160fcfcfd394d33af8b719566626442 : SUCCESS in 8m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2c30c24e09cf48c58e3c0d8ed67d3b95 : SUCCESS in 5m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b5c11e7ba41b400c9c833bab002cc50c : SUCCESS in 5m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/64d2e28bfed6485da7d895097cdbb6cb : SUCCESS in 10m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ad9aaed145554c1b9e6205632c03746b : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62243a06509c4f5db198a2100ee6730e : SUCCESS in 8m 40s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0d7ecdf8030744a78e04efb9182acb1d : SUCCESS in 42m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/364597a5d2934a7db180219c3e292bdd : SUCCESS in 37m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4a85793ec2ce4db7aab16890f51956f7 : SUCCESS in 32m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/daf081af624d453b9cff3a6ec9a904a2 : SUCCESS in 23m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9ffdf6dc36a54cdbbb7a27b6ee623bed : SUCCESS in 21m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cb483e9a49584994a9bf2c4d5b8ad336 : SUCCESS in 20m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/56046ae3aea949e2a820ff94c161d843 : SUCCESS in 21m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e36655e28214f88bb679ae1cd60426e : FAILURE in 23m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbfad3b49faa4393b940bc14fd72dc06 : SUCCESS in 5m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c762a44438e348d983cb2f8f6798aaa2 : SUCCESS in 12m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2e7f8e97ab0c4584b7f236c811165a06 : SUCCESS in 12m 26s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9b541ce5932c43ee85710e560ad3ee60 : SUCCESS in 1h 05m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/752a13865ec448609763e188f0c36ae7 : SUCCESS in 1h 08m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/55bf823da5da4c119220216006bb4f36 : SUCCESS in 1h 34m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/33e71419163c47b48d7f4ff25fb1e36b : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":1},{"id":"04e43ea840725c100c7e0d96ca7a39f884ea6e84","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-13 18:44:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e07ab2f738b555435c2308178474246ee497c43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 03:41:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19abe23cd59b456089e1d65e90315044 : SUCCESS in 3m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/41383cc5038c4e788492b249db1305c2 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85ceb15dbc444c35b3ad42106f92a3c5 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b42d76e17075400fa8d961f8f17e1045 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/107e5a96717249859b2b372c2e8b8ddb : POST_FAILURE in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09d927c4f4604a31af6e15f305dbce48 : SUCCESS in 5m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/63943c663e6a4fe88de7a92b1b000f00 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72e4db53ecbf47acb49ccab50d82b5cf : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/eb0304c2ea604434b89203f9aa7db2f4 : SUCCESS in 7m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e730eeb9e89b4abb9acee31e42edb7a3 : SUCCESS in 5m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/62117d443dec488fa39ed8cfdc317a45 : SUCCESS in 9m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4f74bea976ca42f2a71cf689f97c39e0 : SUCCESS in 9m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f6f00a65bd07417c89388bd8b7625002 : SUCCESS in 5m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9e7c1ff3c8ba44c596b4f9fdcb50e503 : SUCCESS in 5m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/477d071d004a446b9cdff54a6bd366d9 : SUCCESS in 11m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/790de5660a2f4782b194db1b1c08259f : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/263cafa32f9842faa706ada820b91d15 : SUCCESS in 7m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4ee6d31f56a14d57859b45e410e296c5 : SUCCESS in 40m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/db958f492a214f7b9d561ba47f1b5a27 : SUCCESS in 43m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/268b5dfefe204a2d89a06d84d7be4376 : SUCCESS in 30m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3b85d1e25c8843769fc26113eb1c5d00 : SUCCESS in 28m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a4618436ff6f46189b9febafbe185af6 : SUCCESS in 18m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8611cc1c553c46dcab5916630467479a : SUCCESS in 12m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6a079800c65a4fb8985d9eb30b1a78e6 : SUCCESS in 12m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b50e16b52484293be6e15b43f9f19a1 : SUCCESS in 22m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5518e21fcd76461e85df979ba6c001ab : SUCCESS in 6m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cb31a88fd86d4236a7a4096a46b7fe16 : SUCCESS in 9m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/37a6e0e57dc041ba9e7470e9b8c239c9 : SUCCESS in 10m 20s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0d7619da74d64695b8899ae123ac60f4 : SUCCESS in 1h 00m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7b03d1092814bc48ecffaac42945d51 : SUCCESS in 1h 12m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c660404968f74eb4a976fd14ee899a0a : SUCCESS in 1h 41m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/24acc549fb024b039da2e49648fc93d9 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d927a66bb595d5958724e391fdf3ea4d974ea2ab","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-14 03:48:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a1b972bd2fba934ff96cc51ca20d54c3bb66d97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 06:01:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2d1b1976343422b926af88731d47b3e : SUCCESS in 3m 29s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/070d175d7f2740a8ab7c1e03c2db422d : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1681178d72444222a0e35384d11f4b4b : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47947a6559624fe981bfaaf97c15fa7c : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6af645dc1f9243b695675d23643f38c5 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99c52e4ded234dec974a981a3e9d16ed : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb239bc97279414096dbe30d7c089312 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4c2b3afc980d494fa6228e27fb15c23f : SUCCESS in 4m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4df18df2f76843b494a888a6911d886f : SUCCESS in 5m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/04c1c50f2bf64e1a9b1df5e2e927f7af : SUCCESS in 6m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/eeb8c5e52843446c9315e39f0bf93c69 : SUCCESS in 8m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f40a82e675ea46e6838ee93e7c7578b6 : SUCCESS in 9m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6b6832a60b394502a470ceca0f39151b : SUCCESS in 4m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b7754dc8bf854193a8ab56d2e1e7476d : SUCCESS in 4m 10s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3faf02be10ec4f9fb627da0c3e637dda : SUCCESS in 10m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/89aa8c4f8f25442ea2e60d1aec6b3696 : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0460be426eec4304be4ca3e2cad683de : SUCCESS in 9m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9d1c668396ed439fbf202bad38d7afc3 : SUCCESS in 38m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6c16e6dbee5c45e697bafb7daf4b500b : SUCCESS in 38m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c42bedeb4bf840458e71f668071241f1 : SUCCESS in 22m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/536d49c6f09e4fac967985eba99ae2c2 : SUCCESS in 21m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a6694833507740cebac95e5118e455e0 : SUCCESS in 13m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7d513f1aa24e4d4ba0993a41e64e44e1 : SUCCESS in 11m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e6a5d9f524164f2b817367bbb640a7da : SUCCESS in 22m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/15a63a1aeb894bb887d31d6ab67557c6 : SUCCESS in 20m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/362cc56af73a4aa383fa8f6f492cbdca : SUCCESS in 5m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c51d632736ee4a59a5b1a8c1b07a74d0 : SUCCESS in 9m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/89af2b07638d4d5ba713389b054b269c : SUCCESS in 9m 38s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/83f4bb79c91e492eb6954ed300bb2a7e : SUCCESS in 56m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/efea51881ae74ab9ae767f9e6e99e052 : SUCCESS in 1h 11m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5c2bc38975a04529af032f00065e782f : SUCCESS in 1h 45m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d64c5533200546d29f1000bf39fba514 : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":1},{"id":"63f6a18c551717671310e645fe56801b28d907e9","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-09-14 08:56:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1ba2ce044bb804691dbfe2bef8b36c8f0f7bea4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 10:20:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"63ee7c79b1f98fab773271b240d86bde65568d93","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-09-14 11:49:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"613411ae3336f30fcf970dbca1c0365c718401bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:37:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c90e6a676cac0a563a785373d3df61f6a7a6d64b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:37:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5a34f987b2e4669970c2b786e42c4b4 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56db88f44436403588e962773f32bb50 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32ec863885bb4f05a503562f343a4f85 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c68bd8db9d342b4ba976e0557acc513 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a679d9846995476e958d964789bbc532 : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a02ddd7c3fcd4ea3b1b1a37a9df6a7bb : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88a0b1ee256e4cc281f8ffb1532bd533 : SUCCESS in 4m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/34d937b74eb54626b2065f7ffa6a8036 : SUCCESS in 7m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3109ff5b77f54525a3b1fc13ced783c1 : SUCCESS in 18m 13s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/411d6fddcd4249d6a0c6264ee6391e16 : SUCCESS in 10m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/004adf41d3804f7e8debb471ffa65aa4 : SUCCESS in 14m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f2332027bce543c68c41c964499456e2 : SUCCESS in 4m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3b2c3f57bce343e29bb3fa208801d1e7 : SUCCESS in 6m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ea3c9a0bfe0b482ab641a664d59fef5b : SUCCESS in 10m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d985702b207a4fd8bf242b874a76510a : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e73fe3fba3b0453c9df1ae5ba7f56c07 : SUCCESS in 7m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a09a0d2a2df1421d8350aab30a4d7429 : SUCCESS in 46m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/53405d44fb114646ba4615bd0e7e76a9 : SUCCESS in 35m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/696e614e62044154aacf046d34ac848c : SUCCESS in 20m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/10678baff9fc4b83ae8156c1aaafa996 : SUCCESS in 26m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0460f10d5ef4461e96b3277310328dcb : SUCCESS in 13m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/40cbc137dafb4f43ab41d83e518c0eb3 : SUCCESS in 15m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/443689c7c7f6463a85c58fa9deaea16d : SUCCESS in 10m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2dbac9ab215443ca912190a60f17290 : SUCCESS in 22m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ffb4f0ceea24a43ba95781f57618ae0 : SUCCESS in 7m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6d92c36548324dbf99a9c376c5e294ab : SUCCESS in 8m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cb2ef9bd97c24cb9a669d50362485dbb : SUCCESS in 8m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/32dbfdf052d84aa09bea4aa43c4cf9ea : SUCCESS in 57m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/00e67a933e1141b790bd1172b2d6d776 : SUCCESS in 1h 18m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/edc6f08de1dd4ec99989764eb65aee6e : SUCCESS in 1h 47m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/712581328c46408aa1810ca816f56a42 : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":1},{"id":"37df6586ef503a987590e1cf149e6d84f256ef98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:40:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3a743ad29674b9bb1d333532722abba : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6eb291d4fbaf81c1444ce9aff302318b66d73dd7","revisions":{"6eb291d4fbaf81c1444ce9aff302318b66d73dd7":{"kind":"REWORK","_number":1,"created":"2019-09-12 09:48:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/681692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/681692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/681692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/681692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/681692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/681692/1"}}},"commit":{"parents":[{"commit":"81deebb7be87df1a6c3cb4cf82e3bbf593e710de","subject":"Revert configparser update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81deebb7be87df1a6c3cb4cf82e3bbf593e710de"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 09:48:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 09:48:25.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 4.0.0","message":"update constraint for python-openstackclient to new release 4.0.0\n\nChange-Id: Ide1889c08ecbe0232aba35702b3aa7e28fcc2d77\nmeta:version: 4.0.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Dean Troyer \u003cdtroyer@gmail.com\u003e\nmeta:release:Commit: Dean Troyer \u003cdtroyer@gmail.com\u003e\nmeta:release:Change-Id: I7773124f6c22bf2c58d3b02f0427a4d79c071090\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6eb291d4fbaf81c1444ce9aff302318b66d73dd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6eb291d4fbaf81c1444ce9aff302318b66d73dd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
