)]}'
{"id":"openstack%2Frequirements~681819","triplet_id":"openstack%2Frequirements~master~I1716473256dc6b7ef374bbfd1f5d5ab7b309356b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1716473256dc6b7ef374bbfd1f5d5ab7b309356b","subject":"update constraint for tosca-parser to new release 1.6.0","status":"MERGED","created":"2019-09-12 16:26:33.000000000","updated":"2019-09-13 16:40:18.000000000","submitted":"2019-09-13 16:38:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681819-1568392724135-4596321f","meta_rev_id":"747896943325cf4907ebb27a99511ce1a6c0dc8b","_number":681819,"virtual_id_number":681819,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-09-13 16:38:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-13 13:43:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-09-13 09:14:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-13 13:43:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-13 10:06:32.000000000","_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 09:14:58.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-09-13 10:06:32.000000000","updated_by":{"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"},"reviewer":{"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"},"state":"REVIEWER"},{"updated":"2019-09-13 13:43:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-13 16:38:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5afe12f316becab2c4b0dbd931757857221c817a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-12 16:26:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43f5725302b02ea91166437fc9e9c64cbc06bd14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 04:09:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4daeb3eb29b496e86603dd97d520697 : SUCCESS in 4m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/fdde1d194d284884aa32dfb953aea312 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7580b0db95da492fb96aca6bc3a9bb8c : SUCCESS in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88457db3619e45b18b589a9bc27128a1 : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/674a753bbb6f4fb4814eca0885552853 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7fc6f1ac245425487707b3934020a7b : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a36129414d8140f698589d5405f9dcce : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc4bd34a663d4d53bccdbb7139f92d9c : SUCCESS in 2m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8e3845e158be45edb17faebadafbd279 : SUCCESS in 14m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7664513a9e4841969c92115ce16acb50 : SUCCESS in 7m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c3c8c166ee294bf9ab1280f06deb0a6e : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/96ab2ccfb20f49278ed1ffb9b246c23a : SUCCESS in 7m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/835308de8a534d67a72f7407da480e9d : SUCCESS in 3m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/98d22bd736584d328334252f54bf033e : SUCCESS in 4m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/567582ddb66545daa23f50f5f89168c8 : SUCCESS in 9m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bd3b88a57e534fbabfd92366c12b2553 : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/67f79a5011c141f68651363e1a9ba5bd : SUCCESS in 9m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e00b131b8b7444b19561dd63f0db1765 : SUCCESS in 32m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b810bf306a4a4e5fbd42ff6e68a95798 : SUCCESS in 32m 55s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b914a257fa6c429ea2edd0e68787ba67 : SUCCESS in 27m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/81fa945e2caf4bddad41940698af9ff7 : SUCCESS in 23m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6fe7b3b4f1914c6f82d3147760764db2 : SUCCESS in 16m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/79d3900e59b4402098d17d22ba58f4a3 : SUCCESS in 13m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/42e2f47eb1a5421e85b20116527dcaf4 : SUCCESS in 12m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f6bdf9b750c44e269d4507f49b1c4436 : SUCCESS in 22m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/caaf516ce77b4999bd12af7b888b77be : SUCCESS in 6m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4e9d17d3d6c64589afaa41d1c556da92 : SUCCESS in 9m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/071057cc404641768f475651230f6632 : SUCCESS in 8m 56s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ef2e600bf536453dade41d27fd2b7bb8 : SUCCESS in 57m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c12e44bb9c44063b4338d0761ab4c68 : SUCCESS in 1h 22m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/813d4d8778fa4a79a14137d0455335a8 : SUCCESS in 1h 54m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c57fafeef92d484d892e696c89a7afaf : SUCCESS in 1h 10m 54s","accounts_in_message":[],"_revision_number":1},{"id":"46088ae6a56d8ae2dfb2505e4b3076784802bdb9","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-13 09:14:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c6854ae735c9817071cd0635198eb7529fc4e9c3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-13 13:43:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73deac475ae7efa4fbbc54def220c732d512c540","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 13:44:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5f19a155b962cfb609099b0f67a6fd185229e83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:38:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9debf237a1b441fd963f5376282e9080 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/927f84c86ede4eb68e0e899ef34646b2 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d9b004cfae364a05824edd787654b3e0 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/685b677ab8174d4295c31c163d1fd629 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dccda36e52f44da7bef551e5c214ce85 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/357c67d7c6c4468abc29a32f524e8243 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b061620ca934254a75a38f49cb53c9b : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/00ee34b4d0a84a5b8ab6b390b6332ca0 : SUCCESS in 7m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e580dae64189416baffaa5de93c5ed5f : SUCCESS in 6m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/40cfaa3cecd54fc4b90720e91ce0661e : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7521c3a45d434b7aadfbe5d77a826702 : SUCCESS in 17m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fbd59f7440f449f1a223cede347f4a2e : SUCCESS in 7m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0832438d894a4eb79b9ec035359766df : SUCCESS in 7m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9a4303e5356c44e88f503d862e924ed1 : SUCCESS in 10m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c9ad51b0b079433f99191f7329abea31 : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/37f83baed0974e05a5381317fa0a09ad : SUCCESS in 7m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ae46c780e60a41e9bb2df3216fbeeb6a : SUCCESS in 39m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b432474ffb6142e398a1c9ba7a4ac97a : SUCCESS in 39m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fd08efc2145c4bf4b9f2bb6b303b7b39 : SUCCESS in 26m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9aff2c792bb140239047d883350f961b : SUCCESS in 28m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f8a17f3c4c1c45ed85b888d375916d2b : SUCCESS in 13m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d7261c998d2f4eb9883ea6c24d5e0283 : SUCCESS in 12m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/424c94962979410bb984de2c8c0d56f9 : SUCCESS in 14m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a525dc6aba484404ae1a6500beaff137 : SUCCESS in 19m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/340bc092202b4684b64af74121020496 : SUCCESS in 6m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/62ef8f682b0e46c0b8aa5214743ff841 : SUCCESS in 11m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7536e170d6bb4b23b740860b8d444ba0 : SUCCESS in 10m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/458e819ae99c4190a49d413b47daf307 : SUCCESS in 1h 11m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df7fd0f246664a0193b75f94fa0a4395 : SUCCESS in 1h 14m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3cee04f621e442e7a8549f5d9513c32c : SUCCESS in 1h 43m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/36e7cfc681bd4e50a65186b57c588db1 : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1af0b7f67887f970f4c1c74787afb695b3d2b1f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:38:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"243848befec0de76ef93fcb4ce517511640283ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:40:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee2f68f6c54d42a1ad9b414a99c415da : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"633bc4595c5048c321a71485601519e0662517cb","revisions":{"633bc4595c5048c321a71485601519e0662517cb":{"kind":"REWORK","_number":1,"created":"2019-09-12 16:26:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/681819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/681819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/681819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/681819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/681819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/681819/1"}}},"commit":{"parents":[{"commit":"81deebb7be87df1a6c3cb4cf82e3bbf593e710de","subject":"Revert configparser update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81deebb7be87df1a6c3cb4cf82e3bbf593e710de"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 16:26:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 16:26:32.000000000","tz":0},"subject":"update constraint for tosca-parser to new release 1.6.0","message":"update constraint for tosca-parser to new release 1.6.0\n\nChange-Id: I1716473256dc6b7ef374bbfd1f5d5ab7b309356b\nmeta:version: 1.6.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: dharmendra \u003cdharmendra.kushwaha@india.nec.com\u003e\nmeta:release:Commit: dharmendra \u003cdharmendra.kushwaha@india.nec.com\u003e\nmeta:release:Change-Id: Ia36fb4a432de0fee0f1844ab27d689aaf77abb85\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: dharmendra kushwaha \u003cdharmendra.kushwaha@india.nec.com\u003e\nmeta:release:Code-Review+1: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Bob Haddleton \u003cbob.haddleton@nokia.com\u003e\nmeta:release:Code-Review+1: Sahdev Zala \u003cspzala@us.ibm.com\u003e\nmeta:release:Code-Review+1: Rico Lin \u003crico.lin.guanyu@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/633bc4595c5048c321a71485601519e0662517cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/633bc4595c5048c321a71485601519e0662517cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
