)]}'
{"id":"openstack%2Frequirements~681826","triplet_id":"openstack%2Frequirements~master~I11d32dc33da476a5779e9754ea426386a8ce368c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I11d32dc33da476a5779e9754ea426386a8ce368c","subject":"update constraint for oslo.service to new release 1.40.2","status":"MERGED","created":"2019-09-12 16:42:38.000000000","updated":"2019-09-13 16:46:11.000000000","submitted":"2019-09-13 16:44:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681826-1568393085269-600a17fd","meta_rev_id":"b2dcda6d04dcbef24c50974c5aae5806d613ff0e","_number":681826,"virtual_id_number":681826,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-09-13 16:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-13 13:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-09-13 15:11:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-13 13:43:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 13:43:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-13 15:11:35.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-09-13 16:44:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e967af566cf77803c07bd676ef050a384054de57","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-12 16:42:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9edc986d975b56297830d29355a969c115f2ec11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 03:53:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e77d107bfdc4889b81ee2caa945a87b : SUCCESS in 3m 33s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/289eb4ec653041768325568f10b766e8 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41f01268c2fa4ba19db1b95483645c3f : SUCCESS in 2m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/07ce3a6c543d4198b54d90e838644264 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4e4caea8a284f29be10aa3c202c84ab : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67637a1ee7e8432e9b131a79d138c67c : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c76be275261848ec9f0fd7eec60552eb : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c9c3c0de7c245728b6ea6be272369e0 : SUCCESS in 3m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fb308030466b4045ba9eeab0449dc525 : SUCCESS in 5m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6e4734c80205459993ffedec5b99f9c9 : SUCCESS in 6m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/12edc1b734b34b529742ed635b255a8b : SUCCESS in 8m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/402b7ad8c885486091d7a7aa3332fa3f : SUCCESS in 9m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/dcdd663c4a074ed289f0e24abe1e8794 : SUCCESS in 4m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/23eb8e11b42947299f9c8c05615bfff8 : SUCCESS in 4m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/59b2b35d072f4723988326fc665e0945 : SUCCESS in 9m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0b8878af46324a488370444b62b00670 : SUCCESS in 8m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed2307d6658147d39e91b3c03489ec89 : SUCCESS in 6m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e08dd7baedcd4d08b2256b5f25a5671e : SUCCESS in 38m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6d799da6e1d14674bda740298d3f18e2 : SUCCESS in 42m 21s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e4b62354c64d4aa9af07e06ceda75abd : SUCCESS in 24m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c3d73b6a9fee44289a8d43852a41597c : SUCCESS in 21m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d377abb7c50740b59ab8ceb474fc3255 : SUCCESS in 13m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/99f3c3b9c42d4e16963d44bcb76afebf : SUCCESS in 15m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b9473c2fe7bf458c98ea04ab7d510e3d : SUCCESS in 13m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3cfda56dfce0451eb3cea4c535943d1b : SUCCESS in 20m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/27451a56fd6740569e2dfc96e874140f : SUCCESS in 5m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3ae8d82356a1469cad1d9e9cee06b1ff : SUCCESS in 8m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/11b31fab9d1a413c9c0f2bcb29bf569b : SUCCESS in 8m 41s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8780ccb427864864b2e998a252b42893 : SUCCESS in 1h 02m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b2affb0a0fc458cac43643067bc03bd : SUCCESS in 1h 14m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8a2f058bb507449290d2de561e645df6 : SUCCESS in 1h 35m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/91fad7ad74db49f28cf234a674b1e45a : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e6dedba0737301f37bd05923d3f691bbc3e4ae9c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-13 13:43:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"794a865a9c37763a56eb7ec0ae4d786f48e026e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 13:44:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7245bc1e7b27580b58ce48e8180db5cdcdce274","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-09-13 15:11:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"51f8c52389fae037a4cc03fe4d77f63d959e9bae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:44:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f4c3a662d94cb1a1cc4aa2aaf1d235 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9648d7249e004b5cb306a1a6a656e00d : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e809ff7daca948cf94e04296d96cee3b : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ec16441230142d1872f9df0a08c8f48 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d2630ea5eda4cda82d908680b3c8c32 : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a652046b0b845ba9f35a9798b7e8eb2 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/26fc1621856949b0960a53b59cc8fd4f : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e0748d3bee1e4cde8c9b09c930e01302 : SUCCESS in 8m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/933f54d6694145abb7bc9dc942b579cf : SUCCESS in 8m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fc32b995c3ea4b8c85455041d5aa0702 : SUCCESS in 11m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a0cae75cc18044c6ac168cac38de4f9d : SUCCESS in 7m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/54d43b7a38cd4f7db29855992d39b776 : SUCCESS in 9m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eff506e16a2e4df5b736bc6aee0a539a : SUCCESS in 6m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/66898d211ebd40d49a4a153a7c1e121c : SUCCESS in 10m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c32972013f3845cda0247d963155e10a : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea97cb79eff142779b1fb57e355d0871 : SUCCESS in 10m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5cb5ff58d4bb48969b44c3abdf91a4e8 : SUCCESS in 44m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a005d422fc59456faa184bf118c3ae0a : SUCCESS in 45m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/34c2962ad9bd4ea59eac3280c2b922f2 : SUCCESS in 30m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bc9a000f940c4f55ab46521d79176897 : SUCCESS in 30m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a0dfd691417f4417beb304d2861597a4 : SUCCESS in 14m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7b10b86d53134c4f9e170a1398cd439e : SUCCESS in 11m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ef4e149a52e24604a87bbb2d3f5beb0f : SUCCESS in 13m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b07b021a8d7c4934a9b7bf56f84c7e8a : SUCCESS in 23m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1910e96ec0e5419d875b6f18a08177e7 : SUCCESS in 5m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1573dd8a7f4c4937a817191bd9436386 : SUCCESS in 10m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/10f8584bc53e49159c6da710a0acba63 : SUCCESS in 10m 10s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/055869c1eceb49469818bfa589e4f8ee : SUCCESS in 1h 00m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7e26a256067446cbb90afafb0eae8fc : SUCCESS in 1h 12m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f7ceac617973425bbcbf12887cada891 : SUCCESS in 1h 52m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a20b2dd1d78e4322993675e89803f42d : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9e9fb7aaf81c6a138e854e7a22ca59d3fc31ecd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:44:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"37a49528c2e796fc527258688624107bcde587be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:46:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23485af25be34037b9bf4e6901f7041f : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e27f78a8c602a61aefd9192a7cc06491ed1a8ba","revisions":{"9e27f78a8c602a61aefd9192a7cc06491ed1a8ba":{"kind":"REWORK","_number":1,"created":"2019-09-12 16:42:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/26/681826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/681826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/681826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/681826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/681826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/681826/1"}}},"commit":{"parents":[{"commit":"81deebb7be87df1a6c3cb4cf82e3bbf593e710de","subject":"Revert configparser update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81deebb7be87df1a6c3cb4cf82e3bbf593e710de"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 16:42:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 16:42:37.000000000","tz":0},"subject":"update constraint for oslo.service to new release 1.40.2","message":"update constraint for oslo.service to new release 1.40.2\n\nChange-Id: I11d32dc33da476a5779e9754ea426386a8ce368c\nmeta:version: 1.40.2\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ie41cc4192f8c763a8b7155c127001ac2effbd387\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e27f78a8c602a61aefd9192a7cc06491ed1a8ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e27f78a8c602a61aefd9192a7cc06491ed1a8ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
