)]}'
{"id":"openstack%2Frequirements~682083","triplet_id":"openstack%2Frequirements~master~I63faa5de49d515bcd464377e1fec77e8637dde2d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I63faa5de49d515bcd464377e1fec77e8637dde2d","subject":"update constraint for python-saharaclient to new release 2.3.0","status":"MERGED","created":"2019-09-13 15:12:50.000000000","updated":"2019-09-15 10:38:57.000000000","submitted":"2019-09-15 10:37:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682083-1568543823478-453d4686","meta_rev_id":"cfce8ed8aa5b0db9abeadadf4b89e26bb8363e03","_number":682083,"virtual_id_number":682083,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-15 10:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-15 01:33:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-15 01:33:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-15 01:33:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-15 10:37:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f045305155ace9dc52f47ee623b286445dd4c6a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-13 15:12:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8b2ebb0000e5a01052786e73ecc4082e48a6451","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:02:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62713c516dde4af0abda0dcfeccab6a1 : SUCCESS in 3m 52s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ee1aac3dcb6243a69e12451577929560 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59a1c86e91bd41f49b4a460e8b3324c6 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/009dd46003904c56aae09d5482aa958b : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc87867fc60549438427a3cbf5763ad7 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f10b00850034f1383d309a78cf152f5 : SUCCESS in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e18dd538052f44b383f36e9d0626b076 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/994ec3e93b8a4b5f980e5d9d09d0bdcb : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/82a68ce576e74119924f04756a6ab4d7 : SUCCESS in 5m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fe2bf062250b4a039aa7179e08f11026 : SUCCESS in 5m 47s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ffa1fb21510942ccbc530d794ffdf029 : SUCCESS in 8m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2ebcd626d573466e918f54ce9b6adb42 : SUCCESS in 8m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fbf5ab7a46964ca6ad1551f4991fad49 : SUCCESS in 5m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cb838c9525cb4fbbb4331c0049fadd30 : SUCCESS in 4m 30s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/800ebf1d492d48989e339858d77a4e81 : SUCCESS in 9m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e9450ffbaf664b8fa41b9ff04427428b : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1491ebd2f694f6da5d2a2b91ca7ca9f : SUCCESS in 8m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/97f4807252974455926c0fc48b29a505 : SUCCESS in 50m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c91027bb76264fa2955f3b47a9cf32b2 : SUCCESS in 41m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/235dec67c5e743c3a905417da1f84156 : SUCCESS in 30m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bc05d38b1c374bbc8cd810f32625b8b0 : SUCCESS in 34m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/677cc207e6a94088874b73f1a278c331 : SUCCESS in 13m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/befa7aa30e9d4c458a72dc8a4055a245 : SUCCESS in 12m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a32971092a5048c39caeb4ed76a8c854 : SUCCESS in 13m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e3132603da384f1786749ca228dec27d : FAILURE in 21m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a2c76714c5544d8a2420f5dade2dcf6 : SUCCESS in 5m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9a255676acc74c4bae3656b4381909eb : SUCCESS in 7m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7932c10ebd72426987c461aef652f7ff : SUCCESS in 8m 38s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/2befba927f204029a62b1548599189c5 : SUCCESS in 1h 02m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/09b04e8824ec4689a1ff77e58debefb4 : SUCCESS in 1h 23m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bcc2464f00314d3f8c4cdf4c784385a4 : SUCCESS in 1h 46m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b43e71d9bdfc4e3181628ad2ab85cbd5 : SUCCESS in 1h 10m 30s","accounts_in_message":[],"_revision_number":1},{"id":"77e577fe869a68f9ab5e954d0f8b9d9644240341","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-14 01:42:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"92a112d91e21e80ec83fc747bc02af6e9ecb71c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 06:02:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6554c3148f0a4fb4b2a69789e21c77b3 : SUCCESS in 4m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e210c6ab62f148608a11a5f2452793fc : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e3f9b39737b4a62b035c07e57ef14df : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e1247c18f80f4e1ead71e68d7e24e52c : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d79d838ad56402889469aaa3a9c7dbe : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae118d709d884532b69d109f8a0eab14 : SUCCESS in 2m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f087efc0f774dd383125636a31654b5 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ef0de340e8441fc906ba041df8f7bd9 : SUCCESS in 4m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/172c84a8f9764810b7bc4e04489d5b15 : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dc5f52ff1b4049b9b1b4a4101b3391fb : SUCCESS in 6m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/511692b0ef2e4e529e6641b4eb7af401 : SUCCESS in 8m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8c69a15fbbe9450c9a1a2b3be01f94d0 : SUCCESS in 13m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/df9fbf6e247e462198df14da7f2f942c : SUCCESS in 5m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9bb957ac16ec40f1a819da25166d70a7 : SUCCESS in 4m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/db0e7084c6614358b0060ebfd0a41337 : SUCCESS in 10m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9db29dad1ea8472892db80e88dfe6de9 : SUCCESS in 14m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a48299d5d97c42faa8eeff5e49bf049c : SUCCESS in 7m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/702360e1bd1d45ab9c26e64458203952 : SUCCESS in 47m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fd791c1ed7524d058a89734693891e02 : SUCCESS in 38m 57s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1b96e77e984641279e042e14458541dd : SUCCESS in 29m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/453edb3a61d64bbeb232a94f328dbf6e : SUCCESS in 35m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7accc6712eb5425783505ee1b2b7b214 : FAILURE in 33m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/29d26efdc21545eca7d8ca9859cbd123 : SUCCESS in 17m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/717b4f17f260449cad1a11836ca1ea82 : SUCCESS in 16m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e0c50072d1d4910a8800d74e929ce3e : SUCCESS in 25m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b6827f62e544ce6b7673e07fba7e377 : SUCCESS in 6m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3df80e7d877d4c789ba4f295d6106c79 : SUCCESS in 9m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/18ff0bd43d9a4e4d9d5059b81ad9a04c : SUCCESS in 8m 49s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/11e5f2efc83b4269b69a0a8c2f234e47 : SUCCESS in 53m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d9fa1b347504dbd8b3bcf7dd59f925a : SUCCESS in 1h 11m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/793c48bbf23a4de0a00dc2b45b3c1e85 : SUCCESS in 2h 01m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7646058d0cef41beb9ce0ec5ac8cfdee : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5141821705071a7b86dd908384b6496538a611cf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 01:33:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6cdced054970c6eb4e01b5a2cf267f560d24b4be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 03:27:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8db242d98094af5a8c612a3cae2eda7 : SUCCESS in 4m 25s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e3f789ff4bb943e0a8d34eb501226839 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/284d9847e9064f6faa465fe681bb398a : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab0a11006db642159c0ba48bf6f3217a : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9f08b57715f42128954486aa3f0d004 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/813b8247c54b48b1b91c628e7595bccc : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/693a7c861f46444887257b54f762a1c9 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e000f5624ba54ddab998a3df5b1ca19a : SUCCESS in 4m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2441aae27e3142f5a3507057e0417278 : SUCCESS in 7m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/581a831c39f04a57ba32ac61955edfe6 : SUCCESS in 7m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2d1ae0995b424877a31f68a4c19b44f6 : SUCCESS in 9m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aa2aab448307450a915055bf7bf6e7ce : SUCCESS in 11m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d020a758a0e944bf913dd14a6a9b8f2b : SUCCESS in 3m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e06d184bcd6b46b4a64a985a47fa9dff : SUCCESS in 5m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/08cf4c43cf964bbb8d068d1a127530d8 : SUCCESS in 11m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/464cd40a5ea24373b653c46f9e519564 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1a8812c7dc148039e65e2d8cd139172 : SUCCESS in 9m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/edaad2b36a8847abad127751d44716b2 : SUCCESS in 49m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/41a0567f72c84c48b4044d9199c176e4 : SUCCESS in 46m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/02791f134853487586463894a008c4f2 : SUCCESS in 28m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/defbe72ec07d4f988938c4341f9d19a3 : SUCCESS in 24m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2f3a819e3e304a82a75fc04bbb6e42c3 : SUCCESS in 13m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9e7744b578054c329a62017e0fa27ab4 : SUCCESS in 15m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2480c2eeaccb441493b0ea2253893315 : SUCCESS in 14m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d736dfc80d5844499df6ec2a1f9ba104 : SUCCESS in 20m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b8ecdf5299c4745add2c4a6e2adf4bb : SUCCESS in 7m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/33f30e1a2a5040ec86f7dac5340c8d3f : SUCCESS in 10m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/232e645c3a3140f7be5d245a06da66a4 : SUCCESS in 10m 06s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a31fa83d455d46d3a0045f1f06035bf7 : SUCCESS in 59m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b627ecb01a4d495eb8ad0de857e60a5c : SUCCESS in 1h 15m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/67f3346f8dd640498d86611017795ff0 : SUCCESS in 1h 42m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3610141ee7ca40ecbce2775fe470c918 : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f124af84af261026bb657bf350dffe22a23ee882","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 06:35:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b93f15e042226f82f395efe022259cac6a3c52f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 10:37:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8d998a5677f3f825a4b7eeea4b1855162b639853","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 10:37:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02aa712fd6a846718ec6ce8c73212493 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d252d6e38748bb8b74fa1493a54e2e : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa2a83eab6db411984b5396835f8d25d : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d93067bd92a45d686c72257cfd7a07b : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27b29674d8e2472a8513cd0821aec452 : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55d731157be748759810a42308f10782 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c0bd10898a8840c8b8832233a13ca39e : SUCCESS in 4m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7dbe642c86d24294805a20523b3061f5 : SUCCESS in 8m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/10cad9c234a7463bab28eabe8916cabc : SUCCESS in 7m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4330318024a24dbca735a871c0bf0486 : SUCCESS in 8m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/947c0aa690af47d2a56f05cc4f6e2369 : SUCCESS in 9m 47s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/18cab66a81f6410fb4c32b437394a77e : SUCCESS in 6m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1300488407444603b41f7906005a7053 : SUCCESS in 5m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/51884f268e18473db52d7a43e4a6d172 : SUCCESS in 9m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c37d16c181ab492aa5af983aefec7261 : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e271ee0cd6374da58f20299ad7104a1e : SUCCESS in 9m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7b28fbdd1f7740b4b7c45deaf9fbd273 : SUCCESS in 59m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/878c2af3294740848c92664818c30a8e : SUCCESS in 53m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/81b9a8d05e034924b250cd62d6cd03bf : SUCCESS in 27m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/395aa3c8b442448993adc70efbc0da5b : SUCCESS in 23m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a7fb074376de4d16bdff689509039d71 : SUCCESS in 14m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b6454b27087e42308cab01bcb571cad7 : SUCCESS in 16m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/18c7d52ddfa74fb6a555dce0bc99317f : SUCCESS in 15m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2dacd81cf2f4609b4d154b4cd965c7c : SUCCESS in 20m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c7eb13d72b9448659132a20bb367932f : SUCCESS in 5m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3e3f42ef04a74c0cb7fbd4c32133a433 : SUCCESS in 9m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/60fa7f3f3a95490da6b83cce2290b470 : SUCCESS in 10m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a2309447040747f0936b0f3964e61d18 : SUCCESS in 55m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3dec446ffee4d518a294041b00da457 : SUCCESS in 1h 15m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1d20690bc4674d77ab30ca2c5c037100 : SUCCESS in 1h 23m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cb76325c641841b5ac9bd560f22cf026 : SUCCESS in 52m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f1e7680f7307be1c5099ed419744a61967de47ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 10:38:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3478ce3043054650b3d92691c73ceb8a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"394af621bdb022ac5c39c7dae0e57f8732ebe74c","revisions":{"394af621bdb022ac5c39c7dae0e57f8732ebe74c":{"kind":"REWORK","_number":1,"created":"2019-09-13 15:12:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/682083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/682083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/682083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/682083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/682083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/682083/1"}}},"commit":{"parents":[{"commit":"e8ff05b17776fececb7d8d78c4cbb3ac397aa501","subject":"update constraint for python-watcherclient to new release 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8ff05b17776fececb7d8d78c4cbb3ac397aa501"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 15:12:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 15:12:48.000000000","tz":0},"subject":"update constraint for python-saharaclient to new release 2.3.0","message":"update constraint for python-saharaclient to new release 2.3.0\n\nChange-Id: I63faa5de49d515bcd464377e1fec77e8637dde2d\nmeta:version: 2.3.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Commit: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Change-Id: I4239c4f94cc3c0396a81d7e7de3a2ef9f3465926\nmeta:release:Code-Review+1: Jeremy Freudberg \u003cjeremyfreudberg@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/394af621bdb022ac5c39c7dae0e57f8732ebe74c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/394af621bdb022ac5c39c7dae0e57f8732ebe74c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
