)]}'
{"id":"openstack%2Frequirements~682086","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I694e25d5bb8ae2d85b9512202c72266b17d89da1","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I694e25d5bb8ae2d85b9512202c72266b17d89da1","subject":"update constraint for tripleo-common to new release 8.7.1","status":"MERGED","created":"2019-09-13 15:18:51.000000000","updated":"2019-09-15 10:37:04.000000000","submitted":"2019-09-15 10:37:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682086-1568543824878-b9dbb748","meta_rev_id":"81c00cc637f13e8b6eccb81a03d79fb448744514","_number":682086,"virtual_id_number":682086,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-15 10:37:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-15 01:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-15 01:33:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-15 01:33:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-15 10:37:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25307de942e154db215cdb6fe57b2f44cce43096","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-13 15:18:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2da99f960ce4caf76a8fc9caf8c40072896e279b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:06:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/78b2d01c2ffd404f86684997d952656b : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75dc5305f5b74d82b36cb5650a457a38 : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c1a8499ae5e45eb98e1704ec88880bd : SUCCESS in 4m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/02320c4dfc6e48d086f4f342597d42b2 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/138b4422c55642e0951165d407736c3d : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8ba65293b4b4e1983d83e7438ce1d8e : SUCCESS in 3m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/36d5a8fac51246ff91b94d78215bf368 : SUCCESS in 26m 11s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/8599d7e017b94242ac49d3895edf822a : SUCCESS in 26m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d33584c204ea4ae0b42dc403d287e21b : SUCCESS in 9m 27s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/e71b0718979743ed84fcb615428f5193 : SUCCESS in 9m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/047e7df881ea4853bfad7415db477fe7 : SUCCESS in 4m 42s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/529d1636e7ef492cb783badcd67dfcb8 : SUCCESS in 5m 04s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5ece197ff64d4e5d96c02bf28cbf05ed : SUCCESS in 9m 03s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/9298d48b9bb34c57933bcacf49eea638 : SUCCESS in 11m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/49539b7015a94027bb93b2635421bc25 : SUCCESS in 12m 59s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/f1941789bfea4248a7f2cfaeed9b32e3 : SUCCESS in 12m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4acfab2624bd4d0098797b40add26b1e : SUCCESS in 23m 51s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/233187aa7637448ca80fc8a78c71e5bb : SUCCESS in 32m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f4c11ebdddad448db5bf285e859be1a3 : FAILURE in 26m 15s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/3e10ef9e09e34af7828a3ed7e10fb087 : SUCCESS in 16m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63623baa0d3c46f39bab528228e7fcd3 : SUCCESS in 21m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/593e1b7b6f404dc4a25c71406310bc0b : SUCCESS in 5m 45s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e397ef85cb9d4c328ac50e533cf62006 : SUCCESS in 4m 14s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9cce8cdb23ee4471bb83608f0ec56f10 : SUCCESS in 54m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4042da16bc34f3db797b4ed00f69ce2 : SUCCESS in 1h 06m 32s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/24c209189ebd46fcb58094da2b1dbb6f : SUCCESS in 1h 27m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/54a07f1a693c49339f7222119002058e : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d4f5ee7ee422be854a308dfe849a4022faf7bb45","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-14 01:42:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0cc1b31f0224c3a5ba6af398acb6dda5e22af973","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 05:34:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a0ba05ce80514543b5f1932c34e5a980 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14c6e35e1b5949b6bfd95ec1a00546af : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69dd41ea32914684bda30b79ed70a736 : SUCCESS in 5m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ef72e977158f4e6592ce31bb67157540 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36903dc7190a4a3796015d40f41dc5da : SUCCESS in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4296d8ed6c47435d917e347d8d58d2a7 : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ff8217e1696e443294031ec9cec8d5e8 : SUCCESS in 7m 08s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/09041d0dca2643e6a63f5cac45f763f2 : SUCCESS in 7m 21s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/02be35a6fd6c4ce7bd8f3c79f3f615a7 : SUCCESS in 10m 20s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/34639470e0684f99be1b956710c5b52c : FAILURE in 11m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/45970f6450894d32a59baa7cb3a49bc2 : SUCCESS in 4m 59s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/003c18f30cc84544aad0b4aef425900c : SUCCESS in 4m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f18ce055315a4ae1a984be9f37d55af0 : SUCCESS in 11m 42s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/6f3b247e120948458b43293cf0db5a81 : SUCCESS in 10m 22s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d944f637b8ed438a804302b5fe82e148 : SUCCESS in 12m 04s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/5bdeb1e70e0b48f48a9b75779e9d5876 : SUCCESS in 16m 55s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d8b924b9d50740558d5e6cd11891aab8 : SUCCESS in 23m 56s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/49ac9d1eccd04f2a8128f6592158ecd2 : SUCCESS in 31m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/962fa34246f741d1ace80d4c08440737 : SUCCESS in 12m 57s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/8b173178c47b4e13a2d616b104f51f46 : SUCCESS in 17m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ad42c2142544985a860c1e06c43decb : SUCCESS in 33m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/64c196dc6b12417b97a7dffa80d1b141 : SUCCESS in 8m 18s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/a161c2eb7fe04d5295b4fb61a7b1ea83 : SUCCESS in 3m 57s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a16c1fce640c468db650cbac504d2cb0 : SUCCESS in 53m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4cc67bb458d44f38beb5c7aa9597d2a : SUCCESS in 1h 01m 46s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/1410ae3d015c47a8b7ba472e2f97db50 : SUCCESS in 1h 27m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/39f85364bde64fb8a68f8274d1ed50a1 : SUCCESS in 49m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9223187de91a3b600375a09da3b63a4634f431ae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 01:33:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ccf22a5b94115d94ea1f071fa0366b0649c70127","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 03:19:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/facf8257f4d749a78d85d13f3b4514cf : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8571207525b44b0d9832fa4bb12a13f4 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00b88860c64e4c498e85fd762cc2df96 : SUCCESS in 7m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/052312a5416047c79d9feab035648816 : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/31f6be6bb8b845408fc39f3b3fed8642 : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dcb7ff9b8ff7496b9b91f59a68659a2c : SUCCESS in 5m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9cdff38cbe7e4ef5bbb4bf31d3dd6fee : SUCCESS in 6m 21s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3fef64fa64914c51889b3aefc1636b78 : SUCCESS in 7m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/71665290c5b84f45a3ebdd592d48d0c0 : FAILURE in 14m 44s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/00507da4f475445dbea3f5be79324ecb : SUCCESS in 9m 16s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2b18c16af4bf4a8c9669c031356d9d4b : SUCCESS in 6m 20s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/329d4b98c56946538d265f240a241e37 : SUCCESS in 7m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/57d3d615a0134902bfe13db0134f622a : SUCCESS in 9m 19s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/6a9eb412489d49ffae1b36b4794b629e : SUCCESS in 13m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/dca54915b5bd400dbc5d652e02a5b417 : SUCCESS in 17m 45s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/78ad4a77072c490e9b51702e402598e6 : SUCCESS in 19m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/327d17aa02d14986b1a04e291d1c104a : SUCCESS in 31m 20s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/8f21644351bc43eba2bbdb4d6bd9fd5f : SUCCESS in 28m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a45b8e32a34c42d1ab8ce72f2b043032 : SUCCESS in 12m 18s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/2dd2499cba20438e93d176785892109a : SUCCESS in 14m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd23f71e5e8546e98cdb174b531fd99e : SUCCESS in 20m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/86428420617f435585045e2cd83e5a81 : SUCCESS in 7m 11s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/f13b006a58ee41e9a9f15cd6c9f62981 : SUCCESS in 5m 06s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3a2dd6968540452abf19dd97ee53552a : SUCCESS in 42m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/837b777141004fcca081aed5f3b61f9d : SUCCESS in 1h 10m 04s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/86c1e5cbe7cd41f79a0b9e9eff9a71df : SUCCESS in 1h 28m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0ea30bec3b6441f1a9c8d79374e0744d : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":1},{"id":"95940e4a0a46b58a673cfb18d4d9f1af4eb5590c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 03:34:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd2fe208674c52604cdd7432c01344bb17b514a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 04:56:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5d32a6cca91a4e54809cef5c96fd3a2a : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed2e6fafba724ad599222e85dcf77d2c : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8a2b19e46964fe9abe65d6d9f87d873 : SUCCESS in 3m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/95f9317e8d4f452ea96ce2eb4e3b0daa : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80d7b3d69cfb4050b57fa0f3537383e0 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8266a5bd4c0a40278a23deba7aebbb47 : SUCCESS in 2m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cdbe94438f444fee9c3297d44eb2f2ac : SUCCESS in 5m 33s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f6a6d3b1b6b2412aa43eac5bba0d40fc : SUCCESS in 6m 13s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1c84874a36354ff492120ebf66fda9c8 : SUCCESS in 10m 06s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/144592bc1cdc4a53b9714952eeaa2d03 : SUCCESS in 8m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/95a233af8ded4a98a7b7b188ad591bdf : SUCCESS in 4m 27s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/4f0d0ef3107f40388e4bfe23d9396f05 : SUCCESS in 5m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f03f82470a9f41d197c1e1a38562ea6b : SUCCESS in 10m 54s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/5ea68ccbd507443c87a61feea6f7302c : SUCCESS in 10m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3ce0bff9eac54a96b7fb9f6c8257a26b : SUCCESS in 12m 24s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/669956421fd64a58a2208f2690273230 : SUCCESS in 15m 08s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9d8aa6f554ba4430bfbd9c25707708fc : SUCCESS in 23m 43s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/b44330e6447d480abce0ed620f996cca : SUCCESS in 28m 43s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bef6d01063fe44ff9a8a1e856b98e50b : SUCCESS in 11m 48s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/cf8d3a451b164b5b9f9fa27129dc7543 : SUCCESS in 13m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8bcc0c2da6a241d1a7373e5b3f648fe8 : SUCCESS in 18m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0fbcf59ee33b401882ae6cfc57cb9199 : SUCCESS in 5m 45s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/a953f83db02b4992bedc6075075ae1c0 : SUCCESS in 6m 44s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ef36dac321654563a9377038d000527f : SUCCESS in 56m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e44725c33c94c21ad8d8aab4b0f5f46 : SUCCESS in 1h 00m 08s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/0d0f7c11480c42e3ae3aedd0a3bc2b5f : SUCCESS in 1h 20m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f94e73d3145641cb8974e6911900bd4a : SUCCESS in 56m 32s","accounts_in_message":[],"_revision_number":1},{"id":"bbf25747fca71bb6f67ab362321826bd32046c7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 06:36:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"957600aaa21599156170eb3335d5f76cf46827ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 10:37:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8eeec0619607a84c98830c23ce7175ac2c493538","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 10:37:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85931a9bf27f45f4b17ea5dfe7e0ed45 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb46e410f0f9457aa035fa44dbdf5ab5 : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/adebab3658984c56bf852952d6f24e1e : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/323fbc5d64a44863ab5dbde2ab0fb846 : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/37746f558c3a415a84a5751637b67053 : SUCCESS in 5m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/44e7ba335e0c46feb0abfe201ef89e0a : SUCCESS in 6m 56s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3d35d120ff4240ae84bbe2678fa4bf1c : SUCCESS in 6m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a2ec00777e90455db03bb0421dce900f : SUCCESS in 10m 04s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/ce22bbe900de4c0697e09074926df2b4 : SUCCESS in 8m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4ddf5b4d60534e7bab7e1f18d07141a6 : SUCCESS in 6m 15s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/68493c3cb4dd43e886191fd7f27f1dc4 : SUCCESS in 6m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/694d65ecf6c742859cc1dfc7128db0fd : SUCCESS in 10m 58s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/5f6e4cd05ab9443ca17cbedc272ca7e9 : SUCCESS in 9m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/da6f4e2e03ef4daab75e323f0803e411 : SUCCESS in 15m 09s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/01311c71144748b7b7c14ed523031d97 : SUCCESS in 15m 14s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9c84977967d74a37a1e9b8cce059d0fa : SUCCESS in 25m 42s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5f132378888543abada2c93164affe8e : SUCCESS in 25m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/50afb1ea82484d27b0f8750d5e3e8c4b : SUCCESS in 13m 31s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/27adccc8a7224752a1ea6aee227d051b : SUCCESS in 18m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f3601fd4f5a7437582548bbbc260c2c0 : SUCCESS in 22m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b1880a797b134b13a7e45bbd138598b8 : SUCCESS in 6m 32s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/17035f13b8194d878d61b03db33f7770 : SUCCESS in 4m 07s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3c3b0852d5f34cfd853447bf0a9a266c : SUCCESS in 1h 04m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa25a83fa05e421d802ba90021832e8a : SUCCESS in 1h 17m 11s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/798515c7935b4bc09f25e3cec07b2c6e : SUCCESS in 1h 20m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3aebc7b26a664c9ea355ffc922fc5eba : SUCCESS in 59m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab31b52c75eb4ec2d47998fe8661294de4d54e31","revisions":{"ab31b52c75eb4ec2d47998fe8661294de4d54e31":{"kind":"REWORK","_number":1,"created":"2019-09-13 15:18:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/682086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/682086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/682086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/682086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/682086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/682086/1"}}},"commit":{"parents":[{"commit":"f8d0b078ffe32a9995f9b2c34721c4a569e7eaf5","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8d0b078ffe32a9995f9b2c34721c4a569e7eaf5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 15:18:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 15:18:49.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 8.7.1","message":"update constraint for tripleo-common to new release 8.7.1\n\nChange-Id: I694e25d5bb8ae2d85b9512202c72266b17d89da1\nmeta:version: 8.7.1\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Change-Id: Id3a17cc2f110fac15cdd10537a5c97de78d2ca48\nmeta:release:Code-Review+1: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Andriy Shevchenko \u003chuang.zhiping@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab31b52c75eb4ec2d47998fe8661294de4d54e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab31b52c75eb4ec2d47998fe8661294de4d54e31"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
