)]}'
{"id":"openstack%2Frequirements~682087","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I1f43e9558c290a9c3cd7956edebde66509ab9604","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I1f43e9558c290a9c3cd7956edebde66509ab9604","subject":"update constraint for instack-undercloud to new release 8.4.9","status":"MERGED","created":"2019-09-13 15:31:41.000000000","updated":"2019-09-16 00:04:17.000000000","submitted":"2019-09-16 00:04:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682087-1568592257605-600a17fd","meta_rev_id":"f8449a908949db6d40ae8161443409ee93fc3fcb","_number":682087,"virtual_id_number":682087,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-16 00:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-14 01:41:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-14 01:41:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-14 01:41:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-16 00:04:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"edb42236684067438ac8713cb27c338c1c945a55","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-13 15:31:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db44fecd0d4bb412e3ba1c3f2f4220c9d7bc40d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:16:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0ff03c30870c4cc7a9d56824fd3d2e4b : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae2c36c8c704432a066fdeab67f286b : SUCCESS in 2m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6bd33788710847bd984aee2aaf972b90 : SUCCESS in 3m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bb1ee5312a8e4da080b35f8538312159 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d577728b025465983f1896c2feaa471 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/664a9d5b936847afb0698312bd6781e5 : SUCCESS in 2m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/590db37778f44dd5b4e4c5abe9d8275d : SUCCESS in 6m 16s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/51cd0e6e7213474fbc3413b51558ade6 : SUCCESS in 8m 48s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/46a79db809c54f7486f96552b9c28b19 : SUCCESS in 8m 34s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/7d22acb37ddd45bda1256bd5ef59f764 : SUCCESS in 8m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1f58c20835124a569a7292cb27f36c72 : SUCCESS in 7m 53s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/98182aaa57584acdb8e80e4035084191 : SUCCESS in 4m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8a32d4cf3c7c4c75aa6a8880223618a0 : SUCCESS in 10m 35s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/20a2548a775f4874b504470be69752f1 : SUCCESS in 10m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d8e020ac6316442da3a618da7346c0b5 : SUCCESS in 17m 46s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/c52a67ac57b146a4b16cff11c2bf8727 : SUCCESS in 12m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/bfc2ffac20664e5aa3cee1bc1fc15540 : SUCCESS in 25m 18s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/67f1403a45fb4b68af57719bb2933939 : SUCCESS in 25m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a9859fbb43034d6589488c5c7bc092af : SUCCESS in 14m 44s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/12f98291437e45df8256c172e5f492e0 : SUCCESS in 14m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f6d0d4867f849b69c48782c2e0ba3a6 : SUCCESS in 23m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ff5fdcdb730f4fcfb4b506b56cac4061 : SUCCESS in 9m 08s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/40b2dd299ca6445c9447188460c1e04b : SUCCESS in 4m 34s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f293a95893434b21b17451093bf3253b : SUCCESS in 54m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c32f5ce9af9347628aa713fbf9cbc9f9 : SUCCESS in 1h 09m 58s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/f80cba5cb92f4db3913304a6c4770233 : SUCCESS in 1h 30m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6ae86117cc9446b39e36e22a0eaa0985 : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9fbd7dd5262400d5692305b60e1c466cbfd96e2b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-14 01:41:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e48995428c5d70f7610a5e6703bd615c957a2e94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 03:33:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"830459a3fc63aea16599d36a0d6da585f6dddbb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 08:33:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c02917fc6ee4d538876d90648848af4 : SUCCESS in 9m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86b0641a2033410f92bb71f9ef9e60c0 : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e8abc0651d0d49e7bb6428c96dda0fbc : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ffc663978d349d7acfdc4f317133268 : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ede9f6e62898453987b83be796401849 : SUCCESS in 4m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/32e82e7f459645deb3c7f79370e1e42c : SUCCESS in 8m 16s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/6a699b2c335444d6974495b96719f1f7 : SUCCESS in 26m 13s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8f0b92abd86a47d8989a6b4ad89dc02b : FAILURE in 11m 15s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/eae0e36514f2424494c56b3d8e96788d : SUCCESS in 9m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ed3cbc376caf44d7b04052a6fee7bfe5 : SUCCESS in 9m 32s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/956cff20553b4f98ba64ff99452c48f3 : SUCCESS in 10m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4b261c1c4f2e4f7d9ebac5f92f908ee1 : SUCCESS in 12m 35s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/2d70d930e10b4e5794b8497bdc57e965 : SUCCESS in 11m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/07d5b1feb094401e86a11bbe2c2771e1 : SUCCESS in 14m 37s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/d354d222803542f4a1ad088bea4fe524 : SUCCESS in 22m 13s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0ed8c0125e0745c0a4fc8ce9c16abf9d : SUCCESS in 25m 37s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/58da829b5985436a8ce5298efe97dd86 : SUCCESS in 25m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0420c07845904264aa7a4b7ade7de716 : SUCCESS in 22m 33s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/f394ca0f25334deda149ddb20ed68542 : SUCCESS in 17m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3efffe7b17904f7c9887dca785233de0 : SUCCESS in 20m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dd708fbced64426f9502a119c55b2dda : SUCCESS in 10m 47s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/ad5d94d106954891a2a2e7567ff90707 : SUCCESS in 4m 03s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e0442702ed9d4e0e8e5d9c3098ca48f2 : FAILURE in 34m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d78b06bcfd344a0e8cb1973f46d48e2a : SUCCESS in 1h 16m 56s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/2738b5aa5b144177b53d2d140dc07f8d : SUCCESS in 1h 20m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3adcb2d9121f40f1b5e34b20245451a1 : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"64dc7874cdd920a70cdf6f5447713fa746d3c3e4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 01:33:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"39a7ce100a298532782fb91b77ef9642546427b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 03:16:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/85701cf4015d4442bb23bc9b206eb2dd : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9d8233ff5ed42b8834e7be0d964f6b1 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a2ea7e791874ea6afa55d1a4c87345b : SUCCESS in 3m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bf33fbea5a49496aaa631b4923a99aef : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f6156b89afc4976bb5b5385ed15a57c : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/87af41adf816488b95ac026eb7df3b9a : SUCCESS in 4m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c3e0dfa2286c42d49cea8ada07091753 : SUCCESS in 6m 35s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e5c7fe77dbf14990b880691a2a8ac616 : SUCCESS in 6m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9fc06662045d49709d2591347a874840 : SUCCESS in 10m 08s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/8083c55083134463b49dc7da72093697 : SUCCESS in 10m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/25b51b697634482d9daa5bc1cbad695a : SUCCESS in 6m 32s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/4b3c343c12934c60891217cf4134a33e : SUCCESS in 6m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/df9d4e660dde4bf28e91a53dd9269e0b : SUCCESS in 11m 42s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/5bbac8444a0345939c39e45a2f0dd9de : SUCCESS in 10m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2b5dc2fa98df44648612f0a03a8e1edb : SUCCESS in 13m 33s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/93d76e01fc694aeb802f48db1fc840fa : SUCCESS in 14m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/53c30e214a1c4685b53bc3edac66b947 : SUCCESS in 29m 09s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/9e40e0e4db0c4258adf74e65e736f760 : SUCCESS in 31m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/76482c9de8ff448ea89f3e152000b3f1 : SUCCESS in 13m 13s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/b527e5b6894147059e7fa000bef9d39e : SUCCESS in 13m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/39bf8e975d374a2ba1a67d1fbd51d834 : SUCCESS in 19m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9133902ebb59401c9b27188a88958fd7 : SUCCESS in 5m 46s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0a5d5f3d1f93414fb3086b3122f891f5 : SUCCESS in 4m 44s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/b01441369faf474d92e5891f26907029 : SUCCESS in 54m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd0d28d8b056405c8d55722200726f78 : SUCCESS in 1h 06m 33s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/6d231ce5fc444dc3a012adc0b2388e9b : SUCCESS in 1h 23m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/abb007bd34b042a6a65c8269a8bd606f : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":1},{"id":"643f6f7a2a86e639da22c6740485f09ce6119a14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 06:35:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"128865a1d03bc03f67ce055aa8b70ba45e8efd82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 10:06:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87f5f14212224235864fe9d6f1b976b3 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f3d757791804aa4892a4ea907e3e39f : SUCCESS in 3m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cf17a5ad1072439ea5d622b214e40e56 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11312c250e6f4fada57cc60ce9231a62 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4f8e205831c42e2861d6f2ae6f3f196 : SUCCESS in 3m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/41b46050dcb642a4be97374d67250346 : SUCCESS in 6m 17s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3f506b569cbc4abea92632866ab527e9 : SUCCESS in 6m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/17daa96178d74a679531de1de2337236 : SUCCESS in 8m 23s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/aaf1a49076ee43dba1a3248ab9d24f8b : FAILURE in 11m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a0d5535b81ec4c7ab323b5154388aa7c : SUCCESS in 5m 45s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/4edd6e4a11f64659a545e1f9bbe58fe4 : SUCCESS in 5m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/314817f06fb64c4d9433eb549a62ad5a : SUCCESS in 9m 44s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/e6f3c87d1de14d82adcab18bfe49acf6 : SUCCESS in 11m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/eb0b2448014b4cb2b33f56858e9782ef : SUCCESS in 18m 39s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/9247ef1c2a11482180c3da544f5ea1c4 : SUCCESS in 15m 31s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4276067faa2e460881b5f084b3b1a5f7 : SUCCESS in 29m 02s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/a3e4867952ad48f794a476b55172113f : SUCCESS in 24m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/124ba14a8cb84dcfa269448984570686 : SUCCESS in 13m 37s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/b31e41c466bb459f992d646d26b1bda3 : SUCCESS in 15m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6775abf6735d4c9ab3608461a4077321 : SUCCESS in 22m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5d03299f6f784e6788353b8b3d9fafd0 : SUCCESS in 7m 06s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/d175fa05c4be48e596171d3ba4ac1715 : SUCCESS in 4m 43s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/16608e83a4684e5bbf660c94b2d8fdc3 : SUCCESS in 53m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/31ba86bdd7ae4b9cb9a87d748fd4cc2a : SUCCESS in 1h 11m 20s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/4a906bd7dcab4d718f15e2e0533e9e34 : SUCCESS in 1h 30m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/30996745af8b4b81804d9c39f6110f13 : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":1},{"id":"682e26f4236dd361b08f4561feb3f871bda2ce2f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 15:26:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1088a9d44344b7752d10a772389b3656e0ed68c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 17:13:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/845c1b5574a0453db16f2d57d6fac647 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c572074d81546ada821c546f2866cec : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f66f4ae304b4405c95ed239dc9101c52 : SUCCESS in 4m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/31506a7c2af5446b9741a1eb6583f7d6 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9d0cc22db0e40bdb01fe60f0e5eee30 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d46945c902ae4da4bb58a4305de1329a : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/afe7804274ca496389ab8d9db4fcdf76 : SUCCESS in 7m 04s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/1889e77e70ea4f97b152c30616b22936 : SUCCESS in 9m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/77ba530527914208aecff3e3f0f7ae07 : SUCCESS in 9m 35s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/3cfddf03f4194bd9bd024e732946c067 : SUCCESS in 11m 10s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a39e5db194884976b746c17714d26452 : SUCCESS in 5m 12s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/002dbca8dd084215a7cf23cf7c6def26 : SUCCESS in 5m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0a172d68471247ebbdc236e5dcf41bc1 : SUCCESS in 11m 17s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1897e74366944c1dbe112253db69cc02 : SUCCESS in 9m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4c9a0913c9624994a0d13045720efeea : SUCCESS in 19m 41s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b7829b0ab8624aa4aa81979b4db23659 : SUCCESS in 12m 21s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/45da2090c0e24b7e94db3c29bb30d664 : SUCCESS in 24m 57s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/029a30effbeb4219a988f4e88a0c51b8 : SUCCESS in 25m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f03a4e4fda654a5a8758e5f0a7d45e0a : SUCCESS in 17m 31s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/c479eafabcc24abc833c81474e3832c4 : SUCCESS in 15m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a376dbbe73d54fcd872d094108a9a96b : SUCCESS in 20m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5c2d4ec19d6f429badd6f3a44ef65d07 : SUCCESS in 6m 42s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/928da481491f4578b73d19805245c4ba : SUCCESS in 4m 45s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7b1eec539d2e41639704674b87ee2a3a : SUCCESS in 48m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c91515a3e2ae43b8804b36e5fc6e0ec7 : SUCCESS in 1h 16m 58s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/1ec0c840e4c14524992d859209e6159a : FAILURE in 1h 16m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3fef5d8e690847ae826d6d32b17da5a6 : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":1},{"id":"acadad26132c68f018247084f76e238324abcec0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 17:29:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a97a4485714354d042daabbee22bf7abd691abea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 19:41:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4d029996df454d25a27dc720cc15ba2e : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d49607c97494ddb82cdccc2e932ca04 : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d57d76af70dd4953ad7eb22fdf49e52b : SUCCESS in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/11a378ddf6704551af38ad8280fe6e64 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e31436e1200a4f7ca6ca1359da0f152b : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22b982f5438444909b842a77cecad473 : SUCCESS in 2m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/13a4c8b71abc4a20ba0f7a991f2ca759 : SUCCESS in 5m 53s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/b8b0e69544c8439fbaeebd5fc639f872 : SUCCESS in 5m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/26c4ca1a54c74053838d187e1fa7788b : FAILURE in 12m 58s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/a52aa4f8b2104cbda5e7433ce1cacfe7 : SUCCESS in 8m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a8e811519fb349c293ee03cf75d7446f : SUCCESS in 5m 52s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/62b2f576fc9d4286a42ed325b4f572aa : SUCCESS in 5m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f3f912054ab54675b650300a4e2a3425 : SUCCESS in 9m 43s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1249c8d191b240d389b720857c1e5d5c : SUCCESS in 7m 53s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fbd2e00f47c6476b981764d54c2bb67b : SUCCESS in 27m 34s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/7a607d50dfb34dcc824777d74fa267ed : SUCCESS in 23m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e7d5ed049ae04675ba33abba2f59015e : SUCCESS in 29m 11s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/c3866649bc71471c8eb9709f26f913ad : SUCCESS in 29m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1a4a8ec00ad14725b3d4f7f1a4faa567 : SUCCESS in 18m 56s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/48a78ef4c4394390a6b3ceea72e07f5f : SUCCESS in 12m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa992c2b7afb45a18a3abd31cabe9508 : SUCCESS in 20m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a8aa4cd401d04bd0be2d3596a23563ee : SUCCESS in 7m 36s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/4cbeea9c00634febaeaf94a9e7e21707 : SUCCESS in 4m 28s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4a164fd419504fd2a8267cdf40035d3d : SUCCESS in 50m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/110891c64db0474bae8a72c8648ac20a : SUCCESS in 1h 01m 22s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/004e7cdb5d3f4f7393ffc9e303f9a0aa : SUCCESS in 1h 45m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5800220f1208434e98e4eb5aedcf331f : SUCCESS in 45m 29s","accounts_in_message":[],"_revision_number":1},{"id":"727c2ad23c9143783bea628c7367658db7c31336","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 19:57:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4396bd5194a9a76c9db591f3110013671eeba3e3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 20:28:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"26ef132cc0b066236b820b48c6f18954bea89c7e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 20:28:33.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"4ecaa7a6dfce6970f92f64e8c3788ef860a02b29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 21:55:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/14b0c5e1677046339a63e19b33e71289 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/482b6b0afd604e6ca4e6b763cf575137 : SUCCESS in 2m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ba17da8b59b4986b3c099cd229c69c4 : SUCCESS in 3m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9e481b629b2949e0b1abe5caee4ba04b : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cca2cb1dcba24348ab705951ccc8dd51 : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3da45af0912c46c5a8e8d88e9846cae5 : SUCCESS in 2m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5dd45d55df254515935ad37778d999c4 : SUCCESS in 8m 48s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/47d711ac0de149e49ddd941cc93966ea : SUCCESS in 9m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/da135d43b5f64ef8b2179a6b27e7eebd : SUCCESS in 10m 30s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/3675ebbce12940c8af891a34fdc1926a : SUCCESS in 9m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5ea3f97b41dc4560a73411266c47660a : SUCCESS in 5m 53s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/8024ef71ff0247a3a6a2f9b5a5e196a9 : SUCCESS in 5m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/54660f4668744ac9b15bcb2414214e1c : SUCCESS in 11m 12s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/3e9948bee65a4ea0a1e87c572074f39b : SUCCESS in 10m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fe64c42e933b42d4ae6a57db371bcbbf : SUCCESS in 13m 21s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/30f3324dce274da1b5c341879df841b4 : SUCCESS in 10m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/83d8c7648e444ebea4585a34c848d659 : SUCCESS in 21m 16s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/623b9d5c3d2c4074852119b54faa29ec : SUCCESS in 28m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e030657a3c72457aa8c68f2efaaeabb9 : SUCCESS in 16m 13s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/cf690dc1bb0e4713a4d76aec1e38ae5b : SUCCESS in 24m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/531d2759b92a467d985d0fda43b0a73d : SUCCESS in 23m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/64117647f2184621a468b88a72aa0b9d : SUCCESS in 5m 21s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/4201a9008fce40c1a5746d340a6250bc : SUCCESS in 2m 52s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8c33142d4c5c4397870c61e04c3da24f : SUCCESS in 1h 03m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/69de50e6b273468d89a883c150c3dd7e : SUCCESS in 1h 15m 44s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e28031d1d7d24c3dae2d742b3f043cdb : SUCCESS in 1h 25m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/15f66ee0883a458398c8a5bd3947775b : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1478a6497caa05b83886d688a55ee4081455a654","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 21:56:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cc973d0f6182ed6bac16fecfd19354c9d5cebce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 00:04:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82d0b3ef06274c069ff42b983241ea8c : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0dcc44e879748e5a2512d7e7b543fc2 : SUCCESS in 5m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c0fea22262404b3c87d50e4bbaa53bd7 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ec3771e5b144ba18d0044de7846be9b : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1549dc4f2cf49a48f5cd115aa9592a1 : SUCCESS in 2m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c37727a572c94a43b6a3bde60cbc5af0 : SUCCESS in 6m 12s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d90ccf03281c43b1837a44aef81b4897 : SUCCESS in 8m 45s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/82864b7587b143238bfcb5768e014c07 : SUCCESS in 8m 52s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/57b77a4af5b44013b646916c70092bb8 : SUCCESS in 8m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/aa5797c8b41e4e899c85c5741e021a96 : SUCCESS in 3m 52s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/27ea682ae6784a8f898f09a4338018b7 : SUCCESS in 7m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1e5d9d8311ab4562bfe90642113c7748 : SUCCESS in 9m 15s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/ed4296a0d7af45e990fedbbded518eb9 : SUCCESS in 9m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a0492fb98661409d8af53a1d93cb66e7 : SUCCESS in 16m 19s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e7669560f2314e0daf06296d796d8cba : SUCCESS in 12m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9db5b58ef4fc44088ee637089ff5946b : SUCCESS in 23m 41s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/94c13a19bd4748ee80dab2ed4d6fc77d : SUCCESS in 24m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/06a22bd4ec7c4d908fd384f22cf222b6 : SUCCESS in 13m 42s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/93ea38c44a6e4d20876f99d877a7782b : SUCCESS in 15m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d7dba194c8a4c66883f8f6adb9f6812 : SUCCESS in 19m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/66df636dce1446b189bf333d771acc71 : SUCCESS in 6m 05s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/fcaa3e4197e6436d94708302641475a8 : SUCCESS in 3m 33s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6ee717197f8842da9622ee5017ce077f : SUCCESS in 59m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c1fbf22ccef94a65bb46661290b4d758 : SUCCESS in 1h 03m 11s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/8a736ff725214ca58687191fc53090f2 : SUCCESS in 1h 29m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/da2519a05f814c4b8e5c1200b2ba7e58 : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1dca73637a656314e51ccb6dc4a97167db90a91a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 00:04:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e28b932778b75c4215ab280a99d2844f64dd471","revisions":{"1e28b932778b75c4215ab280a99d2844f64dd471":{"kind":"REWORK","_number":1,"created":"2019-09-13 15:31:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/682087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/682087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/682087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/682087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/682087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/682087/1"}}},"commit":{"parents":[{"commit":"f8d0b078ffe32a9995f9b2c34721c4a569e7eaf5","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8d0b078ffe32a9995f9b2c34721c4a569e7eaf5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 15:31:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 15:31:39.000000000","tz":0},"subject":"update constraint for instack-undercloud to new release 8.4.9","message":"update constraint for instack-undercloud to new release 8.4.9\n\nChange-Id: I1f43e9558c290a9c3cd7956edebde66509ab9604\nmeta:version: 8.4.9\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Change-Id: Id3a17cc2f110fac15cdd10537a5c97de78d2ca48\nmeta:release:Code-Review+1: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Andriy Shevchenko \u003chuang.zhiping@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e28b932778b75c4215ab280a99d2844f64dd471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e28b932778b75c4215ab280a99d2844f64dd471"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
