)]}'
{"id":"openstack%2Frequirements~682090","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I8ba50b9dc705498724a62de60fcc0803758535ca","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I8ba50b9dc705498724a62de60fcc0803758535ca","subject":"update constraint for os-net-config to new release 8.5.1","status":"MERGED","created":"2019-09-13 15:37:18.000000000","updated":"2019-09-15 04:23:03.000000000","submitted":"2019-09-15 04:23:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682090-1568521383173-d4b93e20","meta_rev_id":"d1aa5081e5f768757a101125fa0eb81ea76b090c","_number":682090,"virtual_id_number":682090,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-09-15 04:23:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-14 01:41:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-09-14 12:07:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-14 01:41:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-14 01:41:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-14 12:07:53.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-09-15 04:23:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f23c9a9c666eec9d786b3214d3b1e048eb9510fc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-13 15:37:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec6a1ae58000049a3812bf8ccb380392cf08bc01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:11:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/96f47be58fda4c3abd6effd8b744cbb9 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/616aee3d200a44f2aab2a702785cb478 : SUCCESS in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/554309877b704f42ad96ad2ba2b51f39 : SUCCESS in 6m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3a7bae6f93fb46f880957613103b6893 : SUCCESS in 3m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0bd7f3148364c3ba5f737eec6a1af19 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56d0374fa87748f98ab9200d6246430e : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/57a0f214aad440b9b0fbb26ab2864cd9 : SUCCESS in 10m 50s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/68d1c7a64ac74242b8333f79aa07076f : SUCCESS in 6m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a04b256aff3c41898371fe68bc770cd5 : SUCCESS in 9m 15s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/aab737aee0104d9f9f06be63699bf2b9 : SUCCESS in 8m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0b37286c61ab4be5b7a7617a77a3058b : SUCCESS in 7m 22s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/cb604bf9ce7b49e19bfa77e00417b7e2 : SUCCESS in 6m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/23c70f32dbb7414f9c7b6ae47bbbd177 : SUCCESS in 11m 54s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/2a072927b33648bdbb339459f062bf5e : SUCCESS in 9m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ca6ddf3035af4893b2a517f3af227647 : SUCCESS in 14m 38s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/362d4d53ea7c4744a656ce4d43b323aa : SUCCESS in 12m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3b581f2ef4dc4f8cacf20aa1b30711e7 : SUCCESS in 23m 47s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/e963506b079e4da0b9458631167ddae2 : SUCCESS in 28m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3ad1384aef5e4989875964c82a6ed39c : SUCCESS in 12m 34s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/1f068a7740384915bc1ea23afcd4d9a0 : SUCCESS in 13m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/72187d05696f4bf59117933afec4de87 : SUCCESS in 24m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9612405b0c684543b3a3ebdb9108fdf5 : SUCCESS in 6m 22s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/712a425535a9455da8b5d5c91d7118f1 : SUCCESS in 4m 40s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/dbfb074cb5b645bcacbe1dddf4099f5e : SUCCESS in 58m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/71a7d9cf044f4864a5aa8545e11054db : SUCCESS in 1h 08m 49s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/c89434fd4ffe40a798ddacc46ff10de2 : SUCCESS in 1h 28m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d7ecc807d0374747a79054ddecf81917 : SUCCESS in 1h 27m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b254d3a1b123cf5a0220252e06676b7004095e94","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-14 01:41:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c78c9408341b5ca86129e3688fbf9e3da9ed0f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 03:33:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b0e14bf676ae32af86c7955061b9e8503bf7e1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 09:03:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83f2d7ec1f3f4fd0adfd475a7f381b3b : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8905cbcfc9bc49308ff4afa16da1b5ec : SUCCESS in 5m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/14a319e07f694205b7e1a08a904a5c58 : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6cfd74e36fe14e57a9c3470fa6fe0cf6 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/347743cf459748fdae1484a145d7ad91 : SUCCESS in 2m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/087e7a3864274dce87a56ed23a69f16a : SUCCESS in 6m 36s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/bea922f4d8684d3fb9879f6140d37dee : SUCCESS in 12m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5d6955d3f24b4ebf94f9844840f9c168 : SUCCESS in 11m 00s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/e60cb584ce7e47c8b320cd9f4160d731 : FAILURE in 10m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/692cd10809f44f6e8d6b8fdb56d7abec : SUCCESS in 8m 34s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c389f4cc7e184f4f984023abe67470cb : SUCCESS in 5m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d8718e1a1f284f0bbca9e7176bd4cf1d : SUCCESS in 10m 56s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/88478f25ce6b43ada1a26c9952f31f94 : SUCCESS in 9m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bf2ae8e50d264de99d430ddab4ef2e31 : SUCCESS in 19m 18s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/3cb359bc35da4a55bd94311a9e0ccfe4 : SUCCESS in 37m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e15792c3b9374d819c2ca0fa9d68b6c0 : SUCCESS in 24m 42s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/84608e83d68540f28c4d249edb0feda3 : SUCCESS in 27m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bddf65dd91ba435a940d645ab781ea62 : SUCCESS in 18m 38s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/07ad30899c88476b981ffad628fe7aff : SUCCESS in 15m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e74fc71660e424b9a9dc5b0c9319449 : SUCCESS in 26m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/311bb99257994204ad89d655019cf75b : SUCCESS in 6m 53s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/ee4f7a98d4004519aeba9ba1dbf0a504 : SUCCESS in 6m 46s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/70704045a4c24f82baa2a7d713eb053d : SUCCESS in 1h 06m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/68cadaa3cce24fd6885bd8933578d3f9 : SUCCESS in 1h 14m 35s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/7e9b2e64a9eb49e8856dae1e6b978753 : SUCCESS in 1h 36m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ae590396975e444a9b446d6bd45983e1 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f57cbef5a8388fd02d8775e938773bdd28d41d6e","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-09-14 12:07:53.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3f84719e7e88de6ba1c14a5c764a464ece66d1c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 13:41:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/25fec4d652654c69bdfce0618190dab5 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618808ccab3e4a7d89e2340609230b94 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3e22ea3794e458f8bbf2c522ecc70ce : SUCCESS in 2m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/df8ddc01059b4be8aade5f97cbad9026 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/784b235a40ac46c4b833d79e2dafa639 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/883a1e37280642e5b6a4c32fe98a7016 : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cd644aa4c5bf4bc5a8aaa6025958a24b : SUCCESS in 14m 42s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/8ed971d17fa44428a7d0df235e66770e : SUCCESS in 11m 44s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7f915ef747f54b63881ebfaf80d9f746 : SUCCESS in 12m 51s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b392149b1ab54160b8278d0dd03ec187 : SUCCESS in 8m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7cb0c0620f3d47d597f305228a3f0ee3 : SUCCESS in 4m 33s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/8032e298aace40c3baa5fc1c4241c9b2 : SUCCESS in 4m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2e6f4bbf942641fd8ab32ad6386d0bfa : SUCCESS in 10m 50s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/ea417a4492a244808a7cc1d8c8405008 : SUCCESS in 9m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9ad4d2cdaf5740dfa4eccd4962ef94d6 : SUCCESS in 18m 31s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/be0d13a4146b4011a12466ba6c39abb5 : SUCCESS in 12m 48s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/20c72949e8ea438589c9936000501de4 : SUCCESS in 28m 35s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/a52add8496ce4a13aa3d3b5625c7bf29 : SUCCESS in 24m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/15473ffb06b441a68aaf3f9e99acd862 : SUCCESS in 13m 20s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/945301472fb043ae996953bed6a97d58 : SUCCESS in 13m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/69e439cc356748fea45fbc7fabb14e6f : SUCCESS in 27m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/22e4d90c540b42d79284b849cb6f91cb : SUCCESS in 6m 49s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b3e02da8e56949319f84e3a457a5d055 : SUCCESS in 4m 41s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/2b3b92c27b5341a889c865f514a72dc7 : SUCCESS in 52m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/306db32b16aa45d3ac0fce2f0f3cacbd : SUCCESS in 1h 04m 12s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/6bce0ba723cc46649fbc8ef26bcd119d : SUCCESS in 1h 15m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/19fa35024bba46189921f80d07f3f337 : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":1},{"id":"202a75c7dc8841daa0ee62a8fa7ec982949dd457","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:30:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f601a420893d7e5de4392af56141f0f8a0f5163","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 01:34:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"576164b8d346048dd8bebfbf8c137dec955c33e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 03:22:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/84f39e2da01947e29a04f28a246eebc1 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bdfa04fe91e4a57a22875a4f0ad682d : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2798c4f580994bbbb06ca7486b62e0bd : SUCCESS in 3m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/465a46b0edf545a2ab6e459d183f49f8 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85ead6a83e2e4ba394b2ec8c2ea79b39 : SUCCESS in 2m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7cb9dd50d95b498eb1997089a5556a52 : SUCCESS in 4m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/343e3fef43bd4df8aeefffde0f92be8d : SUCCESS in 5m 36s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/a967d05ac5d6479482a991884f3366e4 : SUCCESS in 6m 48s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3c64740f2aa04375af5d25a5dad45213 : SUCCESS in 9m 19s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/391350e9054045ab9e4e155f752c287d : SUCCESS in 9m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/201a2b3ea42f4bde8fd30126ef4a07a9 : SUCCESS in 4m 44s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b0de9809c0bc49a6a8c771506e252e30 : SUCCESS in 6m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2a2063d78ab24368b234e3f6737618c6 : SUCCESS in 11m 11s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/8290a1c3a2af4812a46deac7b4d4c67e : SUCCESS in 10m 40s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2996f25e8c63404d9dca67b399d521da : SUCCESS in 12m 20s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/8ced517894dc45ccb4a1be85fc60f8aa : SUCCESS in 13m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a5e3f511cd70416fa63a186361d8717a : SUCCESS in 24m 28s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/cef2c5979a3240c1b2b2eedf998dbd3d : SUCCESS in 24m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1ec07cc260544888a0dca73f40010494 : SUCCESS in 16m 10s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/037d8903429f4d458fadd5f3856fcb92 : SUCCESS in 17m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/59dc20c24f254786a2c3b2fd6604d80a : SUCCESS in 23m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d9018f4d95c04505b6bdb6ea042dd366 : SUCCESS in 5m 44s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/159f0d5a60204de5b0fd4e01913f83d2 : SUCCESS in 3m 51s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/01b61f598d084e69963fef41b0455619 : SUCCESS in 52m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a89a3aeeecc84ba19694a1b5c59de1e2 : SUCCESS in 54m 31s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/ef180412a6ae4a22a7e64776c5310c4b : SUCCESS in 1h 28m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b318883447c94cad9870ea1ec85d3c50 : SUCCESS in 55m 40s","accounts_in_message":[],"_revision_number":1},{"id":"683458ea5cf356a8d7d2d1bb8a9854bea074dcfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 04:23:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a39e32fcf246ed976ff19c37775ee9 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6420c07d2b6e43fe826ec47c69e4591e : SUCCESS in 3m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/da78131e8fc84100821fcf58929fb5f6 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94351eea1c5f4872823aac29187d91f6 : SUCCESS in 5m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/97199188561d4904abc249ff4a996095 : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6774d485093641faa6d5dd68ca14665d : SUCCESS in 5m 16s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/68dc57bc46884c2d9a2bd0aa0d4a0f49 : SUCCESS in 5m 53s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/58dbac3f77ca486c8dfa66a6bd36453d : SUCCESS in 8m 51s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9557224c62084267820183bfe37f43e9 : SUCCESS in 12m 59s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7fafc977cd7340de9e2133874c40477c : SUCCESS in 5m 33s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/e3286def129e4a8db7bc26dcb09d42cc : SUCCESS in 5m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/18bc62ca2f1f4b219598a5f25b8bdfe0 : SUCCESS in 12m 43s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1e17d83f631f45f5a6131fcc9b4023db : SUCCESS in 11m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/637c85bceb334a7db7933acdd93e08da : SUCCESS in 18m 30s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/76e6e66159a94db484ccf0ac2651a29c : SUCCESS in 11m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/22080e215683433798081ad02d944137 : SUCCESS in 25m 59s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4e4e4d0179464e5cb64de7ed1a7303cf : SUCCESS in 28m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b0690dfe0d6d431eab61daab334917e2 : SUCCESS in 16m 51s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/5fec13ebf5ff44ea8dc165d0e34d6b89 : SUCCESS in 14m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/238e58ff73ae4fd7ad0f86fb957ba77f : SUCCESS in 23m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8733735fc5bd48119cfcc03787400be1 : SUCCESS in 6m 08s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/592e7d9d2eb74d1382a4cbe727586fad : SUCCESS in 5m 56s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9d99bb55bea14c839e664cad54ec7d6e : SUCCESS in 45m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6270271451fa4182be05d4dd9d9d8e5f : SUCCESS in 1h 07m 14s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/44ec9057affb496f94c34be39b4be104 : SUCCESS in 1h 22m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6d5228241f1544b1a620db44135f45a3 : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9934193137634fb95385e1436a3a0ff1b88386f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 04:23:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bfd13f38015913d7de1af31f1e3edd687948b395","revisions":{"bfd13f38015913d7de1af31f1e3edd687948b395":{"kind":"REWORK","_number":1,"created":"2019-09-13 15:37:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/682090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/682090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/682090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/682090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/682090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/682090/1"}}},"commit":{"parents":[{"commit":"f8d0b078ffe32a9995f9b2c34721c4a569e7eaf5","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8d0b078ffe32a9995f9b2c34721c4a569e7eaf5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 15:37:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 15:37:13.000000000","tz":0},"subject":"update constraint for os-net-config to new release 8.5.1","message":"update constraint for os-net-config to new release 8.5.1\n\nChange-Id: I8ba50b9dc705498724a62de60fcc0803758535ca\nmeta:version: 8.5.1\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Change-Id: Id3a17cc2f110fac15cdd10537a5c97de78d2ca48\nmeta:release:Code-Review+1: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Andriy Shevchenko \u003chuang.zhiping@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfd13f38015913d7de1af31f1e3edd687948b395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfd13f38015913d7de1af31f1e3edd687948b395"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
