)]}'
{"id":"openstack%2Frequirements~682145","triplet_id":"openstack%2Frequirements~master~Ib0e9e910a852abe14153c5433da8edd226f161b8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib0e9e910a852abe14153c5433da8edd226f161b8","subject":"update constraint for tricircleclient to new release 0.6.0","status":"MERGED","created":"2019-09-13 19:32:51.000000000","updated":"2019-09-15 21:38:24.000000000","submitted":"2019-09-15 21:37:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682145-1568583420728-600a17fd","meta_rev_id":"555a37791ffe2a7260eed1743b67c7c3afe16b9b","_number":682145,"virtual_id_number":682145,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-15 21:37:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-15 01:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-15 01:33:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-15 01:33:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-15 21:37:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7a5ec700e03adde53add659673246cbc442d14b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-13 19:32:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71fa7537a7d1687e787869ea269848ffe88b307e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 03:39:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e6af4a3a26341b195da1eae288e14e1 : SUCCESS in 4m 28s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/411bd0b052704e40aff5820f612b0438 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef7b81ff862241ab8dd5ed33c054dda1 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09141e7bf6ba466fb0258db8223a053a : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe5a1dddc2424bae8be59cae10f42d47 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd908b013cc947f5b8f3b4a75c3dc504 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de727ecacc594e31b768cbe1b61d527a : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85ba43fce17d4b87aaf79a522cafe526 : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/869a3bd2c5fb4d6fb79cf10aaf68b48b : FAILURE in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ab27982a04c14826852adf4ed4468566 : FAILURE in 4m 35s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/27b1a2eec241499f92ee547258c0df54 : SUCCESS in 10m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4a9ab52cb9bb45cf9b424ce6adcca5f2 : SUCCESS in 10m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d18a7349ceef4ad9a1d1676fd00eb3b8 : SUCCESS in 5m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ccacbf7ff2984a9891b00ba5ce8bd564 : SUCCESS in 5m 04s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/16bb63f127544b6eb8e4532306186a8e : SUCCESS in 11m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8bd17ac237e74e4fb202e91e19cf8c49 : SUCCESS in 8m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e078f91ad09f480c9c03ddf901e34a48 : SUCCESS in 7m 02s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/626c242c0f8e4b479d1d68d000d0b399 : SUCCESS in 39m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2ccbdee911964ae394b9fd0dfa6ec7a0 : SUCCESS in 49m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f1550005e4144b8c8f1a1b1b88703c20 : SUCCESS in 29m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/84fcbbe6e32c4c7190e4864a655c61bb : SUCCESS in 19m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b00a43b5a6aa4933a68c67f4c9ca52c6 : SUCCESS in 13m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7b9f24de41e0499b8398076224065a40 : SUCCESS in 11m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/64f5ce47798e486db6d19d555c5922a2 : FAILURE in 36m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89dac3d6adf144208da8c220df2dd852 : SUCCESS in 18m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d0d535f4938c45e584b20cb67f966d77 : SUCCESS in 5m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a822877d7de64f71952acc9c1f62ace6 : SUCCESS in 9m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f7c103aa0ddb4b7196481a056dc2f596 : SUCCESS in 9m 11s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f1ab15e9d2934b9996c11f5f2d64620c : SUCCESS in 1h 10m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0484f0dc5499405b9ebc9a4b31bd20e3 : SUCCESS in 1h 37m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ba61fd42aa854f76beac84d5366edf1e : SUCCESS in 1h 33m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5373bd2405ea4cd3aa21a7c94c7df2e2 : SUCCESS in 1h 05m 53s","accounts_in_message":[],"_revision_number":1},{"id":"cd03049ab57fc3e2f69a8acdc7eb91ee2581045c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-14 03:42:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"542b99cf7c61a54316766b6cfb7cfff52c0a6896","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 05:51:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c952c3e021cd468391861603d7160eb3 : SUCCESS in 3m 08s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5c421629361f4a228291498659bdeb43 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31bc3b28a054ca09a6567b1c3f6f7d7 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6df1c9e1f02d4bcf80d0b74d924d8c90 : SUCCESS in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b6921bda837430a887470f6cf5ff29a : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d726adca1e4d4043add28dc9c60eb235 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97c95e39599a4e14ae63f402fc755ff2 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b674d4478b446368ab4a8c46f185f56 : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/aeaa6135c8ab444ea445c8e680474270 : FAILURE in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1d95ecc4a3ef460e838a7865422b64e3 : FAILURE in 3m 22s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/133f7cf94253491c8bd474cd58f0285b : SUCCESS in 13m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0cd390653bfb43929f54a1f215714ea5 : SUCCESS in 8m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f32effffb06f4418a0998efb9abce0eb : SUCCESS in 3m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d0dafda1d92546da9e254f8ed219fd6a : SUCCESS in 3m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/624e0893713649589d34e41a0ae34172 : SUCCESS in 12m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a191da3ae1044d2d874aea8200fc6a27 : SUCCESS in 7m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/657b39a8133a454fa90c3865e6697906 : SUCCESS in 5m 49s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/46fc965c52204b44a14d4438d5db064f : SUCCESS in 42m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7ffca12278ab48579a54d3cc4d75a957 : SUCCESS in 51m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/48cf519180054e839a20b52e83812fb7 : SUCCESS in 21m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fb4a9b4d0df645a286cdfdaad94116e2 : SUCCESS in 22m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/841ff088436e47bd9715b49d6508aaa5 : SUCCESS in 13m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8b85daf9e44d458685f6e763f3280ca2 : SUCCESS in 12m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/35b8c71627c44925bfc36f36c399c833 : SUCCESS in 11m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/acf366d8f48d4853bef7b0fabd7e7326 : SUCCESS in 19m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/51fc872073f3474fbc1080920513945e : SUCCESS in 5m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2a7c99e4429a48c3b8c177073caf4c26 : SUCCESS in 9m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ad3a1aa9fb44fcdbaff07fcd39cb36c : SUCCESS in 8m 22s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/759fd7dd9ebb493d806052674efb6b65 : SUCCESS in 45m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ba80ed6ba584ffab12c80d11bbb36b7 : SUCCESS in 1h 12m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9a632639652c456b83fe0301f83ae52a : SUCCESS in 1h 32m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/42b0f2b977cc40a5b513ec2bc8d2d7b6 : SUCCESS in 1h 12m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4a0c6fe76370a4118fd5bf63c2d7766f69aee1da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 01:33:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5fbcdf56f0db9c3093d82f512f014054738fd5f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 03:25:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cab59202f25435fa1ef3dbc3803c85b : SUCCESS in 5m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/89bcc8a0e2c24d6abb7e410d1ea6081e : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b098e165d7444f9c03eb4a11aa7e5f : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b270872ecbd04af790940a139efea3b7 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0723bf10fb0457092486c3e9de1f179 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e6d913c87934c79addae907abc8798f : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d70f4d397283485496976821678da007 : SUCCESS in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d523c1dfb444f2285249b8fbf337f85 : SUCCESS in 4m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7c91e24b5d014708bbab4441f0477adf : SUCCESS in 8m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/345741b95c6a47e8b92bc5e99d1f62d9 : SUCCESS in 7m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8ae70a73f80a41eba6ab15cbaa3e2af6 : SUCCESS in 11m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d8c7dd0d33bc4ac7997c02dd22c8c4a9 : SUCCESS in 10m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/dd9e99751b6b4a76ac3ad65ddf6cb1aa : SUCCESS in 5m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8809afd34ece4ea8af1dc85f4d3941bb : SUCCESS in 6m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ae0bb74c6e2341b497c98801cfaa7320 : SUCCESS in 11m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1cca208e42e34a75babe85e235f52022 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/51f708a8905f474c8c4360bf42fbbd55 : SUCCESS in 8m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f2abcd67afb344cb807f4c048dcff3ce : SUCCESS in 41m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/47897c3b83014b6491d55e8917aa9db6 : SUCCESS in 42m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/03f2c18d8ed343478762d64ab721d48f : SUCCESS in 29m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f6bd3d204168493d8a034fcb07705a0f : SUCCESS in 22m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ef8ce7e869304b9396c7c5a50c1d4793 : SUCCESS in 16m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f06473f2c2264bcabc89d3d4f452c20c : SUCCESS in 14m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f8f4bc563dbc48df96493c1b411f9553 : SUCCESS in 14m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9fbdb12eabf4bcba60d56d893932280 : SUCCESS in 20m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f14d3937c5f54aa6b8a835a32d2bfce8 : SUCCESS in 7m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f540bdc243ff4ac599782e93432b7f49 : SUCCESS in 9m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d6bf4abb25db46509094bf4fd91608e8 : SUCCESS in 9m 42s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ef6c7754802e461682d1a2f5183c049b : SUCCESS in 1h 00m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6e5953701914003b08b3f211223a385 : SUCCESS in 1h 06m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5017ac73bbb8408cab6387aa9f5cf4f9 : SUCCESS in 1h 40m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e3f1667526f544afa2bb1605b7d0466c : FAILURE in 40m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ab483e70660f75dc702b2c54907c5ac1742871fe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 03:34:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bf69d2b9ad50d9c0653f8bcf365bf88af5792999","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 05:17:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d490e2b779140ad8406e306d4067912 : SUCCESS in 3m 46s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c5c17f8764214c1987f7474b82f4d641 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e255514ae320465dbfd81c2aa4e1a47b : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66e53d9d602f4c7496684dd996dc4bec : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2564fe838637406797e5087511c0d8e7 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f62d6631eb2b42b3b75aba203d346282 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00c7b3cd70654da0974d17bf726d927f : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b796bf384e904443bacb7ada92cfb630 : SUCCESS in 2m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3734cd1e95ce4deb9896f028b7aa100a : SUCCESS in 6m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2f6b166390c14a96b88caea3a027644a : SUCCESS in 6m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9964c24cf2c94654a44732e255bdd059 : SUCCESS in 8m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0b9d0f1ce1834428bbb2b464d1e444c2 : SUCCESS in 10m 12s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/86dab94dad5d4e8f970e52f8713e7bdf : SUCCESS in 3m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0926fb345f17494080172a9edc91037a : SUCCESS in 4m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/58fdaa0548e0484aa452ba1852fffbd3 : SUCCESS in 12m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c10aba220b94fcdb385513db0b95119 : SUCCESS in 8m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d62c1af8a49944efa3227ebb688963cc : SUCCESS in 7m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b16fedac1e654f60b9143426fcdf124a : SUCCESS in 41m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a305c60976e246e390a4a6dac1b679f5 : SUCCESS in 46m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0bbf2977af164027b7c05a297a96720c : SUCCESS in 21m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bfd33a099d034d9baad460d3ce4c20b0 : SUCCESS in 26m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0457fefc51c74506a12059ecd053a201 : SUCCESS in 14m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3e703bff98d94395a2a3cc6c51999aa1 : SUCCESS in 15m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/25188647f5da44e0b890301c15ac3c6b : SUCCESS in 14m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9d4a682b836451182dda998986e60ab : SUCCESS in 21m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/81b668d6426e481e99c6e650436239a7 : SUCCESS in 5m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/403bc6895d944d75879ec725ddf6bd96 : SUCCESS in 11m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fa6a531ffd974f3680514621e2bf1d6a : SUCCESS in 11m 12s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/01fefa8ea6b44b2691051691a5acb03b : SUCCESS in 45m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d2dfad49d5345fa95b9eb29b90fff6e : SUCCESS in 58m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d09836f1134f4b8eb79a11d0e6a1113b : SUCCESS in 1h 37m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c9cf11c9b59a43bf9dd411114d9e12b6 : SUCCESS in 47m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f580a78526ec945b109013c5091b37f32efeb2fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 06:36:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73dee4bc30e3b1b44fff25922c2057eade48ddb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 11:12:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acf217e246e74a81bc8d5b5aaaf33d6c : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5a33aefc1cb4407a1b9d2f866e2fcc6 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19f5727b61864bda9006c59f4bdd1cc7 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d400df2bd833494794a0ca96884a8ac9 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0d03f31a6714ece9fa947063366b74a : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0126c9131ccb42f4a20ac53aa6bd12fe : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9e54a0ae5c447ceaf9a17de1734e462 : SUCCESS in 4m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8cea1d048653419ca6f923fdd42bfad8 : SUCCESS in 7m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5466ddcb9f114984a1eb85bf77cacbe1 : SUCCESS in 6m 07s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8e9d1a594ca44694adf70d97a1bffa9e : SUCCESS in 10m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f7d5123c3d9f4ec0a39f5e0ae91a2167 : SUCCESS in 10m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fc7ffec7303d4e71866a21969ed399e3 : SUCCESS in 5m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7a3d72cdc25542c5883583ae73a93bdc : SUCCESS in 5m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9ad1109fea264fbabf854e7f8c10fcaa : SUCCESS in 10m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5f498d11358d47cb9d909501d860a034 : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c3b2ae3af83047f09bc5092c62971c10 : SUCCESS in 7m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0090b7a0daf7433893a92691d4cc7c39 : TIMED_OUT in 1h 00m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0079bf154be24ae2884083c0c7ee49eb : SUCCESS in 41m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c59491628a944b9d86e2c6b62ee29d17 : SUCCESS in 27m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ddc0ccc7623747c68811482886469990 : SUCCESS in 24m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6eadbd270cba449cba25b438f1f45634 : SUCCESS in 15m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a5af62fe391e43c38a2caf11824fd9c1 : SUCCESS in 28m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/45dd33c210b742008306eb1920708715 : SUCCESS in 13m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b93e92184c44c5baafeb1a65cdc3cfc : SUCCESS in 21m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c10a7db471df415e82d0a00425e3da54 : SUCCESS in 6m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f66c6ae0c1834930a45d2c5d1c84efe7 : SUCCESS in 12m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/69e2b461e94940198c13c1235271617c : SUCCESS in 12m 34s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/78b1d86242ce4d27b12b5ef63d55c983 : SUCCESS in 54m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7eb1f96bb8c455d98870ea59d384656 : SUCCESS in 1h 17m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cf23af646c0046f4bf500362dd3295cc : SUCCESS in 1h 43m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c3e8ba40defb4285a9f261cd8886e525 : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0aa2a0373ab80ed3590cc15562472ae688e9a3cb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 15:26:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8f28547c9fde3a2bb29394a33f7eb6620a770431","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 17:23:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b5a09dd0434eec8e4530b9c35e9787 : SUCCESS in 2m 50s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4e38f577befc4233ba3ebc7554103012 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0cff494afc47adae7ab55c95db20b0 : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc1870c41fea47eda535c916e9a9acdd : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7d8924ece064d84b21cf72ee0dffe94 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f8420a2680d43478c4afadb44a3dc29 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0fc7ba67731a4a34878bb9e4f69ca64c : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ff33c6d8116447190c7bc4ea17548b8 : SUCCESS in 2m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/406b89ba312146c39be099c6597792d3 : SUCCESS in 10m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b4b9f45a5ae043179fc2a43d7ca443b2 : SUCCESS in 5m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0c8e50130588431f9691cd94922b2e1f : SUCCESS in 7m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cc187669bb044f26a18925dd4ee364ce : SUCCESS in 7m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a8b85e60170e4d23996a33e3a5a49e59 : SUCCESS in 4m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/acff28c328cf4b96be6652bc5d348e8a : SUCCESS in 4m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/73c115b667bc49a79bbb2e877a5af820 : SUCCESS in 10m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b398a4af9d864db3b8e73d398ae6bc92 : SUCCESS in 9m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/380f60ddc87543829c36a7d08c50e636 : SUCCESS in 7m 43s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ff502773def44ad398a3b7ceb04a3600 : SUCCESS in 42m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2e71f2c63705417cb7179aa03ed7efd0 : SUCCESS in 45m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/19c6c3dc49da4b4bb18e9a62bfb28239 : SUCCESS in 22m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bc4440741e0b46639edd1b96d84f64ff : SUCCESS in 21m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7cb554799c8443d3b9ca6ed5e4f90ba3 : SUCCESS in 12m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4d2af2198ec04eb7b073cb41dfb26269 : SUCCESS in 15m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e10f915f512b4df983b3ecc0fc822220 : SUCCESS in 11m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d404774a2d064654900a19e0336a6c2f : SUCCESS in 21m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/039202a21b3f4ece86cea73d2214d51f : SUCCESS in 4m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/34bd03e2e2af4583a9a708821f426fe4 : SUCCESS in 8m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/00cea1b4f9654c0eb928830aaf0ba31d : SUCCESS in 9m 15s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0fd3691f549a48a9b06489d38812e1b4 : SUCCESS in 44m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/32d9524364e94958aefff124f23184c9 : FAILURE in 1h 24m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8f9fae778dcb4123b392b6b51c69e3f8 : SUCCESS in 1h 51m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2523decaa215489099d7f795a0180855 : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a1d73c9f6bfe55768034acca999e101eea639799","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 17:29:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"155bdc09269f379c5a3640575048446ceb08f423","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 19:05:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d37e1cbf0e40e296c299b81dc806b1 : SUCCESS in 3m 10s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6ad9b9301517451c92f8e02c2a18b057 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8ca77d8bd324181ba2f60b264e58aa8 : SUCCESS in 2m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df800d1a57344ae197afc7a859fb46be : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a600a3951cf546da900346cc43edee82 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14fb9872e4904261883dd91a0c8dba9b : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b320338e00454d7285c7387047a6ea0c : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc9a997b29794629ba51e572cd8c12fc : SUCCESS in 2m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ecd28ca3e5fe4555b4e2759563143f26 : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/865407beaaa8480f957d07f3ce6aa948 : SUCCESS in 6m 20s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b72991e33c1e446c948f16dcf1231b94 : SUCCESS in 8m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b9b1d5f6ff0e4d7a83be87eb854fa995 : SUCCESS in 8m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b565e8fcdfe243f8a32fe92db685f022 : SUCCESS in 4m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0add1c04f2124a909a4ff8cab804cb25 : SUCCESS in 4m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bf70bf4068ac4af9b32e75ddf7066889 : SUCCESS in 10m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/48fa0eed83b8474d8f8d4113fcef4e9f : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0db21c88ed44bbf9c9e67f7a4ec7829 : SUCCESS in 6m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/db45514d3dc84085b3435fc9e0e54bd9 : SUCCESS in 51m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/35597bdffd7b43ffb1a913287185533c : SUCCESS in 44m 48s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a572d178b8ca49b3aac5201114065bd9 : SUCCESS in 21m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f444675c96ac43379b2565af723ba0f4 : SUCCESS in 22m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/db3e1f2205cb44ee84ef6e736a4d15e7 : SUCCESS in 14m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eb0d9f02e3884886852c4a5de84d83c3 : SUCCESS in 11m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/87adddd10cf442aabdaa1e281c35ad3d : SUCCESS in 13m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df8b8d47e7aa481da0a606e92e0b262c : SUCCESS in 19m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/70369853668748f1a6b836ac83cf9d63 : SUCCESS in 5m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ef399cbc9c584e659609ae301a9f8dbc : SUCCESS in 9m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5a6504a979474e6a9b1f5e8ff37ba89b : SUCCESS in 8m 22s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/10cbef6f245b4a7c86adaa464403ba87 : SUCCESS in 54m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/92947b4df5f34dd39ca402e8ceccca2d : SUCCESS in 1h 06m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/44b6ba8f7b1749aebcdf59473cc1ee12 : SUCCESS in 1h 30m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/194bf9cbf0bb4179ba4a5a87e52ec7d9 : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0f8b553d7c13b2bec79eb1b32a23e3c3743866c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 19:06:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52c804dcc12b97de628bd25a5fd2cef846897f68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 21:37:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19c5b6ab51364a73b11e3065d4d09906 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce1b685e21af44a5956d48d62dc24e27 : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aaafca44f645476591877643964d3b39 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7d3a2d715ac48d5bdacd9c053bab58f : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f08641e795204d4281bacc7e7960f695 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69b6d168c5e7436dab2a852232b97cf7 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82ce0450e3f447b9b480116ad0d5be8d : SUCCESS in 3m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5352152f43d945289c0257e1a8af7e41 : SUCCESS in 12m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9995f9d40d97463fb176ee75bb7e020b : SUCCESS in 7m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bed1576b6a754b3a9bc984ff8da52e53 : SUCCESS in 8m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cdad1fcc171f4c83a3fe4bcebae8771d : SUCCESS in 7m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d6488814f0af4276939af48fb673a4ab : SUCCESS in 4m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/af92a4afa1254b519637dd3b20ff80ed : SUCCESS in 4m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/39c2ad2a777a43c281d3097e2820ec8f : SUCCESS in 10m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4270c99db62a4c3284b46a1dc41d2f83 : SUCCESS in 7m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dfc73b9b320143c4bf5f9c503dbcb52a : SUCCESS in 7m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/aa2a9024df5447f7987de584cdff9e2a : SUCCESS in 39m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d0f1aa4b71e04de088906b1ccaf55760 : SUCCESS in 38m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f20e7af7dd5b457c8a2e9146bb7a29b6 : SUCCESS in 32m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2d3cd66605b34538ac90665c3779218b : SUCCESS in 23m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/68a056b803da4a148616f691f329bd04 : SUCCESS in 13m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1ee1111731444b77b2729cb34282811a : SUCCESS in 12m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/49ba2ee92fb4469e956793df1d5cc2cc : SUCCESS in 11m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/10b2740bc31b4dab85b128014c44c9fa : SUCCESS in 18m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b134f41b94c04aeb94bdc0b2e56f3b53 : SUCCESS in 5m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bf05d86f08a04441af22f84c74c84e81 : SUCCESS in 8m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/723fc0637c5644d69fa97fa70ab567ba : SUCCESS in 8m 17s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/514f0dd35b854f2aa64a6c487f941d0d : SUCCESS in 44m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/17064a5649fc49b798f3172bf33d0753 : SUCCESS in 1h 07m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9f92a0ebe7cd437d937eeca418fcf9c6 : SUCCESS in 1h 34m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3d19cb09c5fb40c8bc9edc4f4ebaa61d : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ce1aa656af172be1f3841ef4c609303b1390d7fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 21:37:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"08fb5807357965e249881192aafe13aeece21756","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 21:38:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a0f982ef604ca2991da5996304743c : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8bce4c534e472b69e4cb65d39fcf0596be7d8a0","revisions":{"a8bce4c534e472b69e4cb65d39fcf0596be7d8a0":{"kind":"REWORK","_number":1,"created":"2019-09-13 19:32:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/682145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/682145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/682145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/682145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/682145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/682145/1"}}},"commit":{"parents":[{"commit":"af3c9ff0053d71144ca8d04bfb32c4394d32cce3","subject":"Merge \"update constraint for python-keystoneclient to new release 3.21.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af3c9ff0053d71144ca8d04bfb32c4394d32cce3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 19:32:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 19:32:49.000000000","tz":0},"subject":"update constraint for tricircleclient to new release 0.6.0","message":"update constraint for tricircleclient to new release 0.6.0\n\nChange-Id: Ib0e9e910a852abe14153c5433da8edd226f161b8\nmeta:version: 0.6.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Commit: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Change-Id: Ic99fb368505bc58de880a2d38cf8d958e789b9b0\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8bce4c534e472b69e4cb65d39fcf0596be7d8a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8bce4c534e472b69e4cb65d39fcf0596be7d8a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
