)]}'
{"id":"openstack%2Frequirements~682180","triplet_id":"openstack%2Frequirements~master~I6519ed2269c8e8afb4c689851369d23f47998431","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6519ed2269c8e8afb4c689851369d23f47998431","subject":"update constraint for python-blazarclient to new release 2.2.1","status":"MERGED","created":"2019-09-13 22:13:53.000000000","updated":"2019-09-15 05:52:49.000000000","submitted":"2019-09-15 05:51:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682180-1568526675694-32d082bc","meta_rev_id":"847a480c090ea910f730b33c2d6c9ee75e769047","_number":682180,"virtual_id_number":682180,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-09-14 16:55:51.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-09-15 05:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-14 04:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-09-14 12:08:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-09-14 08:57:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-14 04:56:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-14 04:56:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-14 08:57:17.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-09-14 12:08:23.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-09-14 16:55:51.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-15 05:51:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"afcee3af32eedcd2d5087be7af19a8d2c91ec44f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-13 22:13:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52c8b5d89ae7701ddd5ef5829c6a7001b2f0822c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 04:27:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de11d0755741477491d010241fff5b5c : SUCCESS in 4m 13s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4b9f87fd43ee46b4a539a3f34ea0dc4d : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e827128f00574bdabe8be9d6cd1bf85d : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/253da7e823ac40e6b6bd1476290b1f76 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4c83615475f45498034508a57cbba9f : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae19977a89834675b747f20fdee1b918 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3973049e0c2543dfafbf4f6dc48a0988 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eec1c2920f664fc7821439e7f89163cc : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f36b9bc5264e4c109dd6da3709bdac91 : SUCCESS in 7m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ed150622ee32451b9b454da365037931 : SUCCESS in 6m 39s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5474c930fa254905a6db4cd5847197aa : SUCCESS in 10m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8f72a014ed984486bfac2dc62d18569d : SUCCESS in 7m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1116fccc705a4669b07dbea29d7078a5 : SUCCESS in 4m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/80a8cc93f46a48c3be390a0129230906 : SUCCESS in 5m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e71990be96f942a9ba4e59eba97bf8a4 : SUCCESS in 9m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/28ec5d62d6434bb0899d2f9e524522bc : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62e8dc091bb64997b005dc9ef494b8b9 : SUCCESS in 7m 02s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e667f53e7e294a4299b72690eba19d37 : SUCCESS in 56m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e21fcdcdea0e45c69d37c4e6f3d43a0d : SUCCESS in 40m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/04050bcfb7d54a6b9dd60d10d8e601fb : SUCCESS in 27m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4ca845fb8db64095b7efedcf11d0c2d0 : SUCCESS in 21m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e662d07f38bf47d5aa1b454bb4a06068 : SUCCESS in 13m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8ac93b313afc4887b83641299c0f9ec3 : SUCCESS in 13m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/db4e73bf00764fec84608fe0c0bc4dc9 : SUCCESS in 12m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e59c03121284785aa75dd4886a3a387 : SUCCESS in 19m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e389b055084427ca5f28256d409b88b : SUCCESS in 5m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cb063f4d59424a8fbd54282972e393ef : SUCCESS in 9m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cc178459d4724ae59d3f9ce61d5bedd5 : SUCCESS in 9m 06s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/42c4b1a16fc440a08c70bdb32cafe4df : SUCCESS in 1h 31m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a526ca0fd3cc4edea9476dca6330024c : SUCCESS in 1h 41m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e28bd3393bd541e4bc95242550719b32 : SUCCESS in 1h 56m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e20bfa91a8cc4631b50a3fce802c25c4 : SUCCESS in 52m 12s","accounts_in_message":[],"_revision_number":1},{"id":"768335b483a5d5e4e606ea6e26def731752a79d8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-14 04:56:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"245e95795ac32eddc82ece734118b8968186f7cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 05:12:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2988e37cf55f843166c11f94a35ae2ad21a8f798","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-09-14 08:57:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3823c4f63924b38f7e4c6cad40870c17a978562d","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-09-14 12:08:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"33294b4bc2f16a76c3507945ccac27b50bd5a2d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:15:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5277093fd064386a97af43e538da368 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4983b19aa5f24ae397aae22f1d5d58d1 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bea2f0e38c1b479f87c4fdde6028b32f : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6bec3f7288647e98960d802caeb98c2 : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/574af30710de4ab8af703dc8003e4bea : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d7817c1c9b2441da01d664622b2e417 : SUCCESS in 5m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a07aeda589c444a9bd90c03013d1ff4 : SUCCESS in 6m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8006107c5d5b4c399caee2ac040d81a1 : FAILURE in 8m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c360dea4cd1649f096f2e1a70ca43a00 : FAILURE in 10m 01s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c4198d37e887421d910f8239c13ea7e5 : SUCCESS in 10m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8cb70748b406404b85dbc86bf4575157 : SUCCESS in 8m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d2f049168f65424c9cc3f0d4b037fefd : SUCCESS in 5m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2e6389b50dbc481fb02010d60eb53aa5 : SUCCESS in 5m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/aaead14085ac44f7ba0a62514ea54ce7 : SUCCESS in 10m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/48068ab794ac45819f84ebd02da267c4 : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec979c8b170f4e7e8396415f5ff2b30f : SUCCESS in 8m 14s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/68d76e2151fb49c289f2978aceb2dc1a : SUCCESS in 43m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a61875b788614de9b242c15a1d4c146d : SUCCESS in 47m 42s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ed3a714b1d804b9bb95ae69c07b38b2a : SUCCESS in 23m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4b312bfbaab9459b9c0326bd00a56d50 : SUCCESS in 22m 48s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e6a6bf5e74624efb874395d26d4d4317 : SUCCESS in 15m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3f76bc3e5b544f9d92210b7307a70f10 : SUCCESS in 13m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/330a1d7a4f3d42bfb037a5afe0df5e2d : SUCCESS in 12m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f031ccd7db354ffb8500fc6777a18a0b : SUCCESS in 19m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ede2989bf784042a751f1a7764180f9 : SUCCESS in 5m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bf6d145ca8d84c3baadded8be86dc974 : SUCCESS in 8m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0c5ee79c7f5e4a21a2a69d470e9a46e2 : SUCCESS in 9m 42s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/877d639d809041278e90d0872abfa99d : SUCCESS in 59m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa8d1f1593a54198ae97c7278e3deb0a : SUCCESS in 1h 08m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a3e3575f310d45029007312fd47f90bf : SUCCESS in 1h 31m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7d53880fe9204bf88f0f9ae50ba96bdf : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8e2ac04960e92908fc2f6f6872852f30416e2e0f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-14 16:55:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"20b27fbe1d27b06edbc599a88617b6d27078d812","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 18:54:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ecbc15f54754493968a690c9c9e5532 : SUCCESS in 5m 01s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d5da483043d344d99cbc1fd84644af5d : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac3f1d7df9de474cb3d090cb674ee6b7 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a94a4b0181bf41ddbd07ddffb342c6f1 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/749da9e3f2c647b8b0b99f257a9e0ca7 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5108c1d42c804854b6625d122f9e9dd9 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3543998637db459da0f650177a468c65 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/26daee1aef9f4f1abaae226e530dfce5 : SUCCESS in 4m 37s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f9c48b9a072a40c4bb75cfa41ac1bc4a : SUCCESS in 6m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db20a1494f4d4e338c1b0dd6148999c7 : SUCCESS in 8m 17s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4c25c0930795459fa2dbf31d1ad9615f : SUCCESS in 11m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8171ca5cec6e408bbe3d2080567e37be : SUCCESS in 8m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0205d51415ed420ebf73fda162dca745 : SUCCESS in 5m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f1c1f3ee8c734f708f6afea02b3d2168 : SUCCESS in 4m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3a463af571f449158adcbd83dcbe41b2 : SUCCESS in 11m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9b31cac3c9814176bd4926fbe074989d : SUCCESS in 9m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1ead9766e4648ad9491bf56c0ef3833 : SUCCESS in 8m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/275ab945ec944e9aac87df7b65a1c1d3 : SUCCESS in 43m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6965d5645cbf4daa8b1a5f7d12636588 : SUCCESS in 41m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/470a1cec5520499f9415fef5ef863027 : SUCCESS in 23m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/12c6a231adf0474a8aab9b80188bd990 : SUCCESS in 29m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/137b82a6595c4268853f00ad10098593 : SUCCESS in 14m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f5141a5ba30548d992276e8a96958905 : SUCCESS in 14m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e7d4ff1ba95f47468437c965ae54ceba : SUCCESS in 12m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a6c360cc28604b4d81b2048e28349f4c : SUCCESS in 19m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d5b639f7897d44039de75a690d7b4808 : SUCCESS in 4m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3dbd25294bf6441cbc63218b90baac06 : SUCCESS in 9m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d7a3bc66727541298c0e0413396c4223 : SUCCESS in 9m 16s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1bf29fc194d540b6ac826cd934ef2491 : SUCCESS in 1h 04m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9fc28ac8c1ee4f52baaefb77970a52d4 : SUCCESS in 1h 13m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9398318f42174779815c3e52e4741957 : SUCCESS in 1h 51m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a505c33acc714561a6170f1ffde66bd7 : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0053c1274f1dcde65120921113cc91f4fb59a9f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 18:54:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30f208f966b71a855a850bb96742ee52bb965087","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 01:33:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c89323a0c4d984d0c4a0bf088c54472e8200805f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 03:23:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409e8aeace7043d4883a52e87cb316ba : SUCCESS in 4m 54s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c6d1b01275c343a68577d3cab8a8773e : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25f12deca39b421092d59045ba8d3152 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c22ee6f1a7b4ef8aa3d597959cc65a9 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0008b45dc5d84668bf189e9f970da429 : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c7112d020854726b955488eef208986 : SUCCESS in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f36048c4ece04e21b7802d60a2ea0fd7 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/27b1e3288b7c4bf2aa6c11c3a60dadf1 : SUCCESS in 4m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bd2ca8dc11024907b5392b45e8ca0a05 : SUCCESS in 8m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/28ac359e7789494ca300bc870d688b27 : SUCCESS in 8m 13s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6592fb3268d44ffeabcf491848292b5f : SUCCESS in 13m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9a111654e2cd4ebd8e3b2d595432b3c7 : SUCCESS in 9m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/70d0028430e64043b0038f2ce1f5b319 : SUCCESS in 5m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6bfe044ed3994ed396cbc178e27344f0 : SUCCESS in 4m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/584753b6cfd5427f87e1f8cea120594a : SUCCESS in 11m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fee92d5b86c947418eb8de8942816459 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23653406a0fd46e9bec1fa27cab3432d : SUCCESS in 7m 53s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/83efe5f681fb473cb71285a68a03c141 : SUCCESS in 48m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7cd3aebd51e74f41922fddafba5be04f : SUCCESS in 38m 42s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0ae18a864982405ea176144c927997ff : SUCCESS in 24m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/19f3cb948c934755a3ddf9de951b0092 : SUCCESS in 23m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1511272dfbd846fab006a13fc3d36f0e : SUCCESS in 14m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3cadb0f6a82b405cb1a9254fff0829aa : SUCCESS in 15m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1b94515d74f3475d82427c982cef2cb8 : SUCCESS in 14m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc9ff66b7117441bb4f3d673d9873b53 : SUCCESS in 20m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dc30cbfa268346c3976449619dc95f6e : SUCCESS in 6m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/80583254ab1340d1a16ee3b9bd113a21 : SUCCESS in 11m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3966fd8b014d452c950de85ffe623bdf : SUCCESS in 10m 03s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1031d421750a47d4a826cad4991eb301 : SUCCESS in 48m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a753c2d09cea493ba1bd23558a4e618c : SUCCESS in 1h 08m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/90f84b6c6d6746f0b9992585683ef850 : SUCCESS in 1h 35m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/67a7db0e100b4178927b6dd53d102a39 : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bc4bae9c373bbca307f96a5422a2656d378f1c15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 05:51:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce40ef88eb64d2a887cb9877bc1e43b : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ba66a0896e94271bcf1ebc0a8b2691e : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/80f4778baa7845c4abe5da92c94e8d3b : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d9d1a1b429b477ca5c9a107b6c12473 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fa436161bd544c7aae9c9313f6795bd : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d763cdc132cd4608bf88e6be8c234231 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/306759f7cc3f4aa98d63b5e955cb470e : SUCCESS in 6m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e1e9f2d44dda43b79721a0d961fcb7b3 : SUCCESS in 7m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/728ddae84d544c98b13c72571478fb60 : SUCCESS in 14m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5d6b781d419848ccac0ad6efe5b0e441 : SUCCESS in 8m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7c6a847d4d354b02b5af94205495339e : SUCCESS in 13m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/975d1e2a636f4fceb34a8480e3798a14 : SUCCESS in 4m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/548ef711cd334a3fb3428f985f712aa9 : SUCCESS in 8m 53s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/eaec633dd3fc405bab3f9e59ab2ea4c1 : SUCCESS in 10m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/76af921668034a4c91ffff3e56ab83fb : SUCCESS in 8m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4d67e24518cc4f3d8459d6ce0f4dd12b : SUCCESS in 6m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/987e7c58548d422fb6486adb161eb8ea : SUCCESS in 44m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/29925a9be9174e778f9a58d46f201972 : SUCCESS in 48m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/83544b6e8cdb4d5e8d51694f50c50f5d : SUCCESS in 22m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6a62a70e96e34829a572e879ff471382 : SUCCESS in 26m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/118be1f55c47409f892d885797286e70 : SUCCESS in 20m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c6e1f1770b0e46d8bf8d246b755e950b : SUCCESS in 14m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/956bad48d2a64c38bd52bbddf4d310b5 : SUCCESS in 19m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/242ca14ae2d14858bd162f7292750117 : SUCCESS in 25m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/751c4625e69c464aa896f22ee48909e5 : SUCCESS in 6m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/05a914692fa143ee8eb9985386b9f46b : SUCCESS in 10m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fdf5224b49f24e429ca1e2ec727eb81e : SUCCESS in 14m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/dd11cad490e84ee5bb05572b83169345 : SUCCESS in 56m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6320884deed4eb189a72639d0cf7713 : SUCCESS in 1h 15m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/661668304991475daf145bf98f546148 : SUCCESS in 1h 48m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/07a62a1f00214edaa721ece73ef2e779 : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":1},{"id":"531dccd81e432868a88f0269648b348c5b16b6a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 05:51:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4589f2064f75e56299b20a161e955231a7b42ccd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 05:52:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d4a05bc5694a45b8bfba199f8f78ba : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7832765b0ff126759ac531c7846d23174d7116d","revisions":{"b7832765b0ff126759ac531c7846d23174d7116d":{"kind":"REWORK","_number":1,"created":"2019-09-13 22:13:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/682180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/682180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/682180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/682180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/682180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/682180/1"}}},"commit":{"parents":[{"commit":"af3c9ff0053d71144ca8d04bfb32c4394d32cce3","subject":"Merge \"update constraint for python-keystoneclient to new release 3.21.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af3c9ff0053d71144ca8d04bfb32c4394d32cce3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 22:13:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 22:13:52.000000000","tz":0},"subject":"update constraint for python-blazarclient to new release 2.2.1","message":"update constraint for python-blazarclient to new release 2.2.1\n\nChange-Id: I6519ed2269c8e8afb4c689851369d23f47998431\nmeta:version: 2.2.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta:release:Commit: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta:release:Change-Id: I429bc5aebbc549bd917d000bd563b461e0d831ed\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7832765b0ff126759ac531c7846d23174d7116d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7832765b0ff126759ac531c7846d23174d7116d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
