)]}'
{"id":"openstack%2Frequirements~682938","triplet_id":"openstack%2Frequirements~master~I82ad405011b1619df690202346b5c2519b102d07","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I82ad405011b1619df690202346b5c2519b102d07","subject":"update constraint for openstacksdk to new release 0.36.0","status":"MERGED","created":"2019-09-18 16:00:44.000000000","updated":"2019-09-19 04:04:49.000000000","submitted":"2019-09-19 04:03:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682938-1568865820852-c3ff88a4","meta_rev_id":"e9f20a0eb5da79a1fedff74ce73ff73cfe0559ac","_number":682938,"virtual_id_number":682938,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-19 04:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-19 02:21:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-19 02:21:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 02:21:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-19 04:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9455402bc866f06c9abe38671c47001c81117aac","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-18 16:00:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40ae6a485fab78f0146daeb8ae6b5458c7f5f34b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 02:08:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f8803066889476cbcc5b71060d6d723 : SUCCESS in 4m 20s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b43051664dce44c4a013765edfbabe21 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3849eb1c51040fdbb0497acfce617de : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0846539be5984dd8a573cb6472c7f22a : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc1cc20e626a4c4c9c66c76e98fd211f : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/422116d87397421d8a81138e000ed92e : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b49288e17fd549108cd8590abf6154ef : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60862e5d63884248a65497929815ea64 : SUCCESS in 2m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/dead6176da8c4ce9a3b6443fcbb40e3f : SUCCESS in 6m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ba5b2a665845432ea27095bf30c8ee38 : SUCCESS in 6m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a5c729580a904c0880c1f30f0ee29234 : SUCCESS in 8m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cc837110dc264e53bcff1ad3d6823d08 : SUCCESS in 7m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f96e71e181104cbc8c0b2eae5f8fd304 : SUCCESS in 5m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/19f259a465f548449ae1d20136d2a838 : SUCCESS in 4m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8d1d03b69d2f4c94835c3558a5a6205e : SUCCESS in 10m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d1e787571e914242a057ef3d7c6a3214 : SUCCESS in 7m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4978e6b6f4504e26861fb96583a12693 : SUCCESS in 7m 53s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b95953a4b4a64677a6d7668a5b1a1b0a : SUCCESS in 14m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b50a406ae49945299c05a5ceaa7ecbe8 : SUCCESS in 14m 50s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/66cd6a4792e94d7e93e6939619465ec6 : SUCCESS in 25m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2f474e96fcde40a5b16130ad960349a1 : SUCCESS in 23m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a0e1451dfbab4f6ea3e9406a217759fa : SUCCESS in 17m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cbe0f6fc506f419bbfa24b279dddd0a6 : SUCCESS in 12m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/92245d410043479e8c26adceeaa34691 : SUCCESS in 12m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/15b29a708b8e4eafbbebff0875a0bc47 : SUCCESS in 21m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/26223dc9aa9e437b9bab1ebac2db0f39 : SUCCESS in 5m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e90d8e4324824f92a80363e39504e756 : SUCCESS in 8m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d14192d0f17c4be4bf0e6da415d7de79 : SUCCESS in 11m 23s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/90fa5267e5004808942dec2e3acb0ba0 : SUCCESS in 1h 03m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e0bb920155d495daafbdcf4c9901aae : SUCCESS in 1h 20m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a8d5fa9bd9874e468b197c04f384eb5e : SUCCESS in 1h 59m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e2b5c5b944de43e1bf9aad8d93f4db9b : SUCCESS in 1h 10m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f79c14926507176046b618c933376870593ea296","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-19 02:21:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b12c136e8ef0a2534077c3fb3012cd3d6bb97bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 02:21:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b384187cea720937f1251602769bd2f0168fe83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 04:03:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce68d7b89c2b44578c0b8d6a83b7977b : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cbc2cec49c24b51973d28f08a8ba831 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fda6a15ed03437da7bb08cb4da42fd6 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0800607a27b84c7cbbbdcb7614538203 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de595560556e45209258019b4f86ae6e : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38bbe98bf89442f2b10707160ccc9e47 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6d46bfd75084c2cb55c41dfea059b11 : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ee1662664a5e4c5ca1a25b5d5cc2da13 : SUCCESS in 9m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c56a47991196455087b13ea2f2c2de07 : SUCCESS in 9m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/92b98c4da65a4e9188fec4cbde7c6337 : SUCCESS in 8m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/34edf61353e24bc1a1b40e7a11005143 : SUCCESS in 11m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/66e005047f8d46678a5aaf0115b1bfbd : SUCCESS in 5m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/996df7c32d89498ea10882308667359b : SUCCESS in 5m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5f9609ea51da40fea933e7ac7ee342aa : SUCCESS in 11m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/99219b23cfd647f78b2a5e93ccb20ab2 : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8a89a49f65794ddb86385290f63d0854 : SUCCESS in 7m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9474668509144e5694a3fbeb6c1f0545 : SUCCESS in 16m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a1c61273149b4a92beeabca6b6807587 : SUCCESS in 12m 50s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fbd1a78604264c719618eebd273eb505 : SUCCESS in 32m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d89b6af6ecc64ea0884aba049b5e4f73 : SUCCESS in 20m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a599c5aa8c484f7f8c98d8359d04f8f3 : SUCCESS in 11m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fa190ca7a2aa447c99ff8fe161355dc1 : SUCCESS in 12m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/81bc75616a0a472d9b315c4343fde84c : SUCCESS in 12m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d437ab45a8034db9a02c4afd8f591a99 : SUCCESS in 18m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e35da011e1d4c07a935c792b30758c1 : SUCCESS in 5m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d7ac7e4e483f4c458780cf91ec7e1900 : SUCCESS in 10m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7b8afc022e7c4fdab3a0108e5efc0e0a : SUCCESS in 10m 26s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4eb2fb9dcbb84f25a4461c910c71cb9f : SUCCESS in 54m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d89b38ebcc84e45b9247384055aef4e : SUCCESS in 1h 17m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1799ceed487249aa87f3f78f7267108c : SUCCESS in 1h 33m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e78b1f95d783461a8c81208b6ad77219 : SUCCESS in 1h 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"58881eeceb36cf4c35343fa1f8737755956f7fd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 04:03:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"402a5873a9691c689d14a14d3b8e16839f3ec3a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 04:04:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/800d729e6802461b96df2aec9e8fe7cc : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"89499f382ccbaaad356c0386916c2357fdb3e378","revisions":{"89499f382ccbaaad356c0386916c2357fdb3e378":{"kind":"REWORK","_number":1,"created":"2019-09-18 16:00:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/682938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/682938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/682938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/682938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/682938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/682938/1"}}},"commit":{"parents":[{"commit":"c14b4c55d0d2e3e9605f0612032f40304d96a464","subject":"update constraint for os-win to new release 4.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c14b4c55d0d2e3e9605f0612032f40304d96a464"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-18 16:00:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-18 16:00:42.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.36.0","message":"update constraint for openstacksdk to new release 0.36.0\n\nChange-Id: I82ad405011b1619df690202346b5c2519b102d07\nmeta:version: 0.36.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Commit: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Change-Id: I2a9978d7c0523de646997941d2de0af5fef42414\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/89499f382ccbaaad356c0386916c2357fdb3e378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/89499f382ccbaaad356c0386916c2357fdb3e378"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
