)]}'
{"id":"openstack%2Frequirements~682979","triplet_id":"openstack%2Frequirements~master~I1e03b9115fdd6edd8e1ef9fa7cad89cbebf8ae70","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1e03b9115fdd6edd8e1ef9fa7cad89cbebf8ae70","subject":"update constraint for paunch to new release 5.2.0","status":"MERGED","created":"2019-09-18 19:42:24.000000000","updated":"2019-09-21 03:42:04.000000000","submitted":"2019-09-21 03:40:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682979-1569037247579-fe8b4c94","meta_rev_id":"eec3e04566e5a0f6c7609b5d4d7a4e6a5ae393e4","_number":682979,"virtual_id_number":682979,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-21 03:40:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-19 04:03:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-19 04:03:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 04:03:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-21 03:40:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"131ed5df90389f3de9f7285bbdf137d64aadf2f9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-18 19:42:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d230a042fac433ae33159654049f673aa931942b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 03:24:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a05fa290984e378be30fb49dcbf6ac : SUCCESS in 3m 25s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/728a3ccae54c4b10bc8af6e57b573bc3 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e63f0cc55c8b4a0ab0a09f8ce94951e2 : SUCCESS in 2m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/65d3009d9a8943ffaa15c8a456290c1d : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10511807aa7a40baacb3ae85601fd92c : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7e4083cb0f54ad1812a41a879151bf3 : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6de3124d83d41b590e45ebe99c299d4 : SUCCESS in 2m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/99d2b79af59f446597db8f7c5d569e81 : SUCCESS in 3m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/33cced3e4d864fbebb89b18e06e53699 : SUCCESS in 6m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8009a9c76455400c9d80f026b0070801 : SUCCESS in 6m 22s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2309b2104862481084e1c3dda3cbc515 : SUCCESS in 8m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fd14abfda36541cc8efd8b2828733384 : SUCCESS in 8m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/41b69e0089d5407aa9ae618b75c63618 : SUCCESS in 5m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/21e7440aee494ce4aacf436825521c46 : SUCCESS in 6m 05s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8666c7f0b0c04002aab18dfef30d4418 : SUCCESS in 10m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/969fde330e194fdfb527d6ac4478dd28 : SUCCESS in 8m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3fd3bcc5c49645769eb1b5715207a35f : SUCCESS in 7m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/70d25547febf455a94fa8a0cf30402ed : SUCCESS in 15m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c322f557995e440fb68ff961934899f4 : SUCCESS in 13m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a7e78bac48f84d1fa61b69b0165ad4d1 : SUCCESS in 23m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a9613c780d0248ca84e9dceb0fa77e85 : SUCCESS in 23m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/feeed5743fc84ae2aefd225eb225bfa4 : SUCCESS in 14m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/214765fca01248cbb4bc3268f216c4ec : SUCCESS in 11m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/34e23a97caf646b5a6c073cdc01d0b79 : SUCCESS in 12m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bb0b9fa2ccfa4bf085e0bec6cb6cf9ae : SUCCESS in 20m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7391f9505ae942fd8d7217542a46ee43 : SUCCESS in 5m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5f158c3fd4734ae5970bf3360af513d6 : SUCCESS in 11m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f847ce77ecb048cc9dcae9590cf58d3c : SUCCESS in 9m 38s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c2201aaf5d17459c9553edefd2e1bae7 : SUCCESS in 1h 07m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4c58322b8b749709fd7c75f6e64fde3 : SUCCESS in 1h 13m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/39d776bc132c48dcaa1eb2682d8dd841 : SUCCESS in 1h 48m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f1b40d6d33824daaac849ed051f8a6d3 : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c9f3731770233ac3b681bafd540479b4e9f37949","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-19 04:03:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22e1f6a6bc2b05f9e4a961a25d9fbf5dff25e777","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 04:03:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d2b1ecff9a198ba139b964a8be52737a151038c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 09:45:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/404d59a2a4cd404d9b622688c8bc1c8c : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c97fa964e0e84dbf9a4ddbecc90f33d4 : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46068b06ec434627acb5add69e7ce03d : SUCCESS in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8dd1947274e47a1b33ce56c1f980e36 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63cf631085ed4630bded222cdcac682f : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f8a934be8544a8f8a1a22af0a7c1d40 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/783342e8cbaf40eb9cb5c0c1da443b5d : SUCCESS in 5m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1c6e7674ad8e4cda9d3638e5304d8640 : SUCCESS in 6m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1c85e932fb1e49e88797f2f83e896040 : SUCCESS in 6m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b87b7bae17524a7ba48204e78d38879a : FAILURE in 16m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5f26ac545d35445f8f2fdd40d765fcd3 : SUCCESS in 9m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/694fdb8c7386412ebfec606c1a7db82c : SUCCESS in 6m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9d3582dde1ac43ff8937d2834479e248 : SUCCESS in 9m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b378642432864e74a0ce1eb0336ff225 : SUCCESS in 10m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c064d7f2a93f45208999ddcbaeaf85b0 : SUCCESS in 12m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ff8cff935a240cc8b8746d2dfeca553 : SUCCESS in 5m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9dd7b61ac5e144dca6390166bd7a004b : SUCCESS in 15m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/70dad18824664b91884d997b0d97624b : SUCCESS in 13m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5e82f8b5e90c46ba8a22617975d1703e : SUCCESS in 25m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ec965432260e46758672372079634c0b : SUCCESS in 21m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dc120abdaf174f779ff19a0201322fbe : SUCCESS in 13m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/54bd16acd7eb4108bd7b1fa8fa45a23f : SUCCESS in 11m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5b7a5e3c7cc24248be4517832b375697 : SUCCESS in 11m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b90b9c1bc35450999b99c83ad01510e : SUCCESS in 20m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eeff0a2ab4254e2c9bbd5d272260f32d : SUCCESS in 5m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/64b678ead0254c6abfff3e74232f2f44 : SUCCESS in 9m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/95318b789c6647d68e9fb2cd8c11fb1b : SUCCESS in 8m 28s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5130a462c5c84d319e8e156e1adb6f83 : SUCCESS in 1h 09m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ceb46173a66e4e0b8e156e4739bec022 : FAILURE in 59m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7ab48ee1eb8f4cc7a09b5e08ebddcab0 : SUCCESS in 2h 08m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/87cf29ad5b7e49278d960bb6fa1230f6 : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d728daabc5f1b669fe5123efd6f41c104527dea5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-19 14:56:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e3d8918658d39f5f1ce704175d6b15e4c86e9338","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 02:52:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4ad19cd75e24032a864fe104407f431 : SUCCESS in 3m 23s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c6f7422d42064c07b8e9be5c354aac3c : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d3bf56c113a47cbbcc6df8ca4fa982d : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8680126ecd84231bfcab9247e19e959 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/237fd5ddd8e8425dbac43dc69285854f : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca503f5450ac464985a4899a323dc434 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ab9bf73063041628e7543edba847a19 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e07614ef34f84a4782c2912e230af57b : SUCCESS in 2m 48s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/00be4c51978a4d1c990eb663a9680899 : SUCCESS in 6m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a07720d997314ff3bd844bfb7a0afe1d : SUCCESS in 6m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fae5e74cb16a4a0496606cd3ecc3db8b : SUCCESS in 9m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4df9a7dda14e4f8da98a899171a65501 : SUCCESS in 8m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/302de3d9d220493dab4f24e7a20c9180 : SUCCESS in 5m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e1b414a34ff245318ede03e8fa7aaa71 : SUCCESS in 5m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/664d098414c34ac98a2474c100b50faa : SUCCESS in 11m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c5e042e1ddf846e1afcada5630c6c0d2 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d90b380946e34637bec4281d326c42f6 : SUCCESS in 8m 21s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/eb064a35e1eb4747a90c078b5128dbcc : SUCCESS in 14m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/81f18d3800434b11b4dd511d1034028d : SUCCESS in 14m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2a14cf07a5fe45cf95e7da7b981c0262 : SUCCESS in 29m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a2f00553ec68480fba094fe49f50000e : SUCCESS in 22m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/325fa4292ae44e6d85d8ff0f5caa1a80 : SUCCESS in 14m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2d11ec54c18848e59a3d4640feeeea91 : SUCCESS in 13m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d3d95bf3b5744d46b21779641624184a : SUCCESS in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8c28701b8014d33ba9b8e30a1edecad : SUCCESS in 19m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a76ad059de014dfc813090c30cadb39b : SUCCESS in 5m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8badca04439549ae88a2cc92d2a464bb : SUCCESS in 8m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4a26c0b46bb24a9bb7413912d0f7483f : SUCCESS in 9m 42s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/48c0f24efbac471cb4f32c9e6a9577eb : SUCCESS in 57m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/875db1a4759e458082d5ecf7e828c398 : SUCCESS in 1h 08m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1912529ce4de44719627f3e9af368c87 : SUCCESS in 1h 35m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fa5594c0ce44458e9ad76c968b550edb : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":1},{"id":"75459cc7bc63c57745604ba9b9e3daf68b8cfa08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 02:52:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42e2897870c7343ea4225f0f9670c7902dfeaa05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 05:23:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d172ef81c1d64ea5bbd6844d978a7b51 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6de6c7f9d28e4e5880fcd4f2c5538609 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4235d4885fb5406eb49766c4eafe9fa3 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88f004ec8dc64fb7a7ebf7e2e6e98f0d : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcdcf0200b0c41ffb5b076e8bf430a13 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e3d943c0c6348f5b2f72888efd1c1e4 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2f028de038164c73849ff6d95e18a125 : SUCCESS in 4m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/63330a2ff0ab490eb986b3960aa8a47b : SUCCESS in 6m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/06799ed83d9540ebae045a1f72156d09 : SUCCESS in 7m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5c581f66afe74c989196e54c3118e28a : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fcd7d9cf34fe485aa2b949dee6b764cc : SUCCESS in 10m 05s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4f058ef946494926a18c4829a8f2d5e8 : SUCCESS in 5m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/85d5212e75af4f55b96c45f5deca3521 : SUCCESS in 4m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/71d14114bbf2430a878bd64dcd4fbe3e : SUCCESS in 13m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/df6b6b77160c400e86ee23e70b225e3e : SUCCESS in 8m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c010612e2df4f8eaf24517d817d64f8 : SUCCESS in 7m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/eb669d3742684e0489b76df4d5662d4c : SUCCESS in 18m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1d1895fd9df94f17b5babcae6c1823b1 : SUCCESS in 12m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b5cf18c075f64b7bb8a21aebf348a329 : SUCCESS in 27m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/64e14125d52247be8aafa21e4239fee0 : SUCCESS in 23m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b0b8a4714b4247c0aaef24bb2a766d6e : SUCCESS in 13m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/325339d73c60492697ecff698a401de0 : SUCCESS in 13m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/46d145b4497143ce830e6a19543c6445 : SUCCESS in 15m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6cbcbb72f47c40c2a75e0a84be7dc09f : SUCCESS in 20m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a278e23d3ef4e85a2a28b3887ae2174 : SUCCESS in 5m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7fdce24be8964b40bdd3e8f971ae971f : SUCCESS in 8m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8b74524457034f7292a8022503549fc3 : SUCCESS in 8m 59s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1f85f6ffc018465da7f20d4d1d6b39b6 : SUCCESS in 1h 00m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d3fa2ae3b1c4df4bc9731ad38c10d32 : SUCCESS in 1h 18m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/09197ffd33f643b19360ef6e5e4d4ead : SUCCESS in 1h 37m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/97ad947e18044f33888c1427bdf53fa0 : FAILURE in 1h 10m 45s","accounts_in_message":[],"_revision_number":1},{"id":"75eaf04ff8ff098480c5526710caa2a00b1d28be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-20 15:04:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c9080499b65c06bf1ff52cf8317c0a42ce6604d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:44:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f996645e9d82486081d094763dca1149 : SUCCESS in 4m 04s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7755eb7117f84c15b67caa58dbf2dc86 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c15ecfc59ff43ab869c4623af42647c : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3ba162d40b848b7b74c37fdb7cc9c27 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd3532c2e0024eb9918f85e99856308a : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f61570bba7224566a3b2326a2dba9c6a : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9704be3619df4b198c018c7c555ba3db : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e9c23ee4c2c4f41909733b5a20c3757 : SUCCESS in 2m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/80dfe5a8b79d4381b516b80f17739160 : SUCCESS in 6m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d8a71091ef0a4da187e4958ba1f88e70 : SUCCESS in 6m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/27d0c59e52bd48f997e3bda8501bb541 : SUCCESS in 9m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3721d59b706945398190747f2ead269a : SUCCESS in 8m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d08363c19f104747985236cd13dc1b08 : SUCCESS in 4m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/453a33c26a044fed87ca61ee35954c57 : SUCCESS in 4m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/33c854dcc9bb41d1addb9e82a8217a91 : SUCCESS in 11m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/799f48c71c2b454d84fef083f1e399ac : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e98940010494187ac532573ad1d6441 : SUCCESS in 7m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/44cf709d92c849f69aecc828e37cda2f : SUCCESS in 13m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0a4481b5dd9a455185f0b20f1a077ed8 : SUCCESS in 22m 13s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b05c89cca8d54a91a51e47b0b27bcbfe : SUCCESS in 21m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5a1802282adb4590aaf27000d66191c6 : SUCCESS in 20m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/135f1f15ae7a445d98480674ac3bbfa5 : SUCCESS in 12m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e53cfeb33e004a8ca689b9dabbc0e09b : SUCCESS in 12m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3c14d477aa964299b6044f0f5ae1e67d : SUCCESS in 10m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b002b278d1ef4b1e962f987263944917 : SUCCESS in 19m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/376300f0bdf24f4cb129f2e01acf5dcb : SUCCESS in 6m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1da62df9614842c18761ccf711a554a3 : SUCCESS in 8m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bff80f5dfa664e189cdf81669171bf36 : SUCCESS in 11m 11s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d04baead5e154c939d775a08e0e4a7fa : SUCCESS in 1h 11m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0887b5ad21414996a4a4da8ca6b842df : SUCCESS in 1h 00m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7df8036e9d654a7db17775075d4d9492 : SUCCESS in 1h 48m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/73c1d80abe0a4f7dba1d1dc013bb267a : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8b220b0fd657fc2f4cbf1e56acfa78bc23da467d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:45:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c78cc2a0aef5137249523a4e280baa88f3826674","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 03:40:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f269905e0d3247b3b7e2318e3460a22d : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71ffe980f5954f98b34359e25efc96ab : SUCCESS in 2m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa596e7247fd4f188d769c593cba54a3 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88b85a1561114e5dbc8222866c30de70 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/057eebcc233a4c8ebcd79871d97589f0 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c0e3afb1d5b4b56bcbe8002c10ce51b : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c18eaa0ba1f4deeb1c7ecb11fb0d5cb : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f7ff861511bb41a2a8aa7f75c6cf7344 : SUCCESS in 6m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/62dd2845524f45cc9819a46b26f4fd67 : SUCCESS in 8m 45s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a4fca8e22ac149d09b8741306bcd2941 : SUCCESS in 8m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5ca5b58a3520421e9a7b9f83cf321f2c : SUCCESS in 8m 21s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1a385d86588c41048a3f5113534444bd : SUCCESS in 4m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4b077ecdd66c4ccca6196ea63b6fd8ea : SUCCESS in 5m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c2e37997ad23498d9710d87d9c8da425 : SUCCESS in 9m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2b4643e9f50b42a3b612a7665e2021f6 : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/957ddfcd4acd48528afffc5e312eefc6 : SUCCESS in 9m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a424d970c4234023a8a2271e33ba7d9d : SUCCESS in 22m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5426ac0ce3504a33beade2179ac188e0 : SUCCESS in 15m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6b6c631141b74b4ea2f9ce48b80c6c32 : SUCCESS in 29m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/062d25828fc346e39ca7d363e56831fc : SUCCESS in 26m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b75efd1590be41e3a3802aa821e5ce7f : SUCCESS in 12m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3f1c2e47742c4233b856ccb5ad2cc271 : SUCCESS in 14m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ecdee7dfb3ef4d3db86069ab78c769b8 : SUCCESS in 16m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/02c49a6b19c9446183a1ed397392848e : SUCCESS in 18m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/342055fb660249b5817f3dcbcb5c4177 : SUCCESS in 5m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d33af559120846609e056dc953988a50 : SUCCESS in 9m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/da44a00ede544ac988790a35dae44d77 : SUCCESS in 8m 39s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9aeed56cdbb544f881d7d9085ed86ddb : SUCCESS in 1h 01m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/573dbbb6037546e0a9251c41b71782ee : SUCCESS in 1h 39m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a7b3767bd0194555acf12a2c504084f9 : SUCCESS in 1h 34m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cded5109b11d4c8c8a4e6ee0f9ef83c5 : SUCCESS in 1h 07m 20s","accounts_in_message":[],"_revision_number":1},{"id":"32a73cdc9119b9d63890c406ab89f33133008e23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 03:40:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b8a2dc1d187305c74d1d316c591a44a928890165","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 03:42:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/972720f29fd74517be2a2ae6e8446862 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65b1ff57d324da57a9419e1a97d34f6aa25a781a","revisions":{"65b1ff57d324da57a9419e1a97d34f6aa25a781a":{"kind":"REWORK","_number":1,"created":"2019-09-18 19:42:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/682979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/682979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/682979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/682979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/682979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/682979/1"}}},"commit":{"parents":[{"commit":"c14b4c55d0d2e3e9605f0612032f40304d96a464","subject":"update constraint for os-win to new release 4.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c14b4c55d0d2e3e9605f0612032f40304d96a464"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-18 19:42:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-18 19:42:22.000000000","tz":0},"subject":"update constraint for paunch to new release 5.2.0","message":"update constraint for paunch to new release 5.2.0\n\nChange-Id: I1e03b9115fdd6edd8e1ef9fa7cad89cbebf8ae70\nmeta:version: 5.2.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Change-Id: I7c003d3e285e86a174baef7c2749518972262932\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65b1ff57d324da57a9419e1a97d34f6aa25a781a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65b1ff57d324da57a9419e1a97d34f6aa25a781a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
