)]}'
{"id":"openstack%2Frequirements~682980","triplet_id":"openstack%2Frequirements~master~I0be0ff85dc8a5fdf326f076bf4b3430dedf0f5c0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0be0ff85dc8a5fdf326f076bf4b3430dedf0f5c0","subject":"update constraint for os-apply-config to new release 10.4.1","status":"MERGED","created":"2019-09-18 19:43:04.000000000","updated":"2019-09-22 09:02:32.000000000","submitted":"2019-09-22 09:01:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682980-1569142885602-c425b5ec","meta_rev_id":"b83ca55314bb72b96487413f9ad2ebee5b6cc7b2","_number":682980,"virtual_id_number":682980,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-22 09:01:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-19 16:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-19 16:21:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 16:21:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-22 09:01:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3af2b3b2038eb3f2ec5a01b31c71c2ac3011fb95","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-18 19:43:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1a54cc7ea7c481a95400cfa96536a34f7a1ca78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 03:11:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26710ca685d14051864186e84e7f4d89 : SUCCESS in 3m 32s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/06ac8fb2d5f94ed783b273e016f51af7 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a432979bf02541d0816b8e6d241e2f9c : SUCCESS in 2m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0180a2a577f42e89d38e552f26537e6 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b24a560fab44ce593c5c32b83e39d6b : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d260939cb39473e897192a78ee55618 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58dda290d4d740049e1efa280735e0a4 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a53d6cb8cdb47f0bb6effd5a27aa0d5 : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/856a4ff3a087431084bde40eed153ed1 : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/569dc1981a6447558569dee846b13db7 : SUCCESS in 6m 20s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d4d03342f72c401980cbe8fca53de1c1 : SUCCESS in 10m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c930c66bfa9a42f7a39173ede1b46f6a : SUCCESS in 9m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4c8fad8681b14803a35c63c4fc4f8872 : SUCCESS in 5m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a0f8f00c0098482ba38bf0ec738137ce : SUCCESS in 4m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/91b40d200fb14d85bdc90d76b74dc16a : SUCCESS in 10m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3af1a0a96e294bc3aaf4ecb36bb7d88a : SUCCESS in 8m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d09527eaff64298a2ac6967c0ba430d : SUCCESS in 7m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6bd5978980914c2683c58e7bbb024db5 : SUCCESS in 31m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/451808fe274a47a3a80fa8b17082603c : SUCCESS in 15m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5455b2ab3fa443318fa67c14efb265e7 : SUCCESS in 23m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/391ea9d008f44dd28d954791d7909b7b : SUCCESS in 21m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e57d49a2e331426f8a522261f6710c71 : SUCCESS in 12m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0ef6bb9fddb54611bef32534b8160c73 : SUCCESS in 12m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2575f59549cc4d0789227bdff6118a72 : SUCCESS in 13m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f82b26d54ac4f028fd45db33eab255f : SUCCESS in 19m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/99232b4976e242d28f027e210f46951d : SUCCESS in 6m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2a79d7d0bf814f33aa7929a7e1c1397c : SUCCESS in 8m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/db57858fa24141aaa506ed26546a3e86 : SUCCESS in 9m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/20b8b680cee143afb7019ecf0137e9ac : SUCCESS in 49m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/464540e710fe4f89948df32943ab1284 : SUCCESS in 1h 21m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/43564410ae9e41b9a0f5da232e35136f : SUCCESS in 1h 56m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/91fb54c83b04409bb334ecc07fa32ca5 : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":1},{"id":"269d529418330afded17784eca9308ad91fa1a18","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-19 04:03:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6aa8717f35bbbfe0994cc80a7373ce76263ebfde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 09:46:00.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"7d56876292cb942fc9e79b1e9888a255318905d1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-19 16:21:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2918630ac67f9fd79635c5a01c797dae66188d82","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-19 16:21:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ffa8ba4ba744902feaca72b0700786642b39042f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 04:12:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e6535788bc4543b7f4e7396be614ec : SUCCESS in 3m 29s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9eda45016a4442abbfea8873b7d5372b : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abf8a1b0d9c04c049c9ef56c763604b8 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71c7ca18417446f7b100ac413447ca9a : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/345beb3a72234dd59d0d1df9af3d8460 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b23dbe0b2bc47878df24c8070ce5a92 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90f093796c7c492ba82a60aa8546e279 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0fb8489b5db4dcd9ff1db7b5344f7fe : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a50ca627cf704edbb21bd40634c15168 : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39d4b7429dfd4fe982c878f477d749eb : SUCCESS in 6m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b6ea59c1b9eb490ba1a6f74f99667103 : SUCCESS in 10m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6a33e710a5b54f24abec0758ba9780f3 : SUCCESS in 8m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/98c3f94925de40b3991e045778c35d63 : SUCCESS in 6m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d06d6c5ce6014e2bb4ed14ceeb261e62 : FAILURE in 5m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2e94c386319341be8bf44f2a1d694c36 : SUCCESS in 10m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e00bc513a26f48e6bc225e1e8fc79a4f : SUCCESS in 9m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/38140be7b74b4d419a7724936169c668 : SUCCESS in 7m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fc7f136aa0f940c0ad9b46abc6701db8 : SUCCESS in 18m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/26c5c97ba5a64299b762c6f510dbc01d : SUCCESS in 13m 47s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3627671f03a04a05a7174b945a612635 : SUCCESS in 23m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/032369ca35d340dca7f738f16ee640e7 : SUCCESS in 23m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0723e8e87af74a6fa2b0c878f1aee4af : SUCCESS in 16m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e336fbd4e3774519b63dcdbb385406c1 : SUCCESS in 13m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3d31a6807d41472a9b1edede298b1fd0 : SUCCESS in 14m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67799f75e1fb4b9db555379e08512de2 : SUCCESS in 19m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/943e7df1dee14dd98d4b48e6356931fb : SUCCESS in 5m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/074ee7f666cb4ec397b63b514594ba49 : SUCCESS in 9m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/21bd6e6df83a4489b0e44539b076fbc1 : SUCCESS in 10m 07s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c0631406c36546bebca7c8b8dd2ed544 : SUCCESS in 1h 05m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/54d74952e3ba4906b179a16fdfab77ca : SUCCESS in 1h 11m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/73b93c3eb3ce4f75be1647865a2c78cc : SUCCESS in 1h 46m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/77336f357c1e4b85abe6af11ceecf20c : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":2},{"id":"749186e8a0c938ec5242c1702a5802281528afcb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-20 15:04:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c82941261642b1999a8a867ea37a6249100498df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:56:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b24e9df39be46fab39b019aad126a7d : SUCCESS in 3m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9f1ab44e90b449c7ad59dfb1288e4793 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11e1271db58b41208ebffcf3bc8d7cd5 : SUCCESS in 2m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6a7ee23f10944f1bbaa08f8fa49bd2d : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0818d086c47a4ab399d9efe022ab0f34 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed792dc82f9c4cc58113533854e5d9f3 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/628a404554f5438c9a26527db9e69435 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd0f08e020174c0cad10e392732b497c : SUCCESS in 2m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5f8abfd05e274effa440713ad8b62285 : SUCCESS in 13m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/91056478917a426380e443005d567e1c : SUCCESS in 7m 21s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/91efabefebaa49f0afa37ee675de4b33 : SUCCESS in 10m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a3537908e7654c05880b3b4f7770fce9 : SUCCESS in 9m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b20974079980429d8058bbbcd6248858 : SUCCESS in 4m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4fd6adfcb9154550b3cdcd28fcabc7d0 : SUCCESS in 5m 04s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/895f7423b0ad4967a5ab950beb65811d : SUCCESS in 13m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4f345b19aad14eedaff160c93d74c10f : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/456db7eeb341422095fb774e00557921 : SUCCESS in 8m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7a5c58bd5c7f4c84888fdc6260d5bf04 : SUCCESS in 24m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d7598bfb449c46dc99b7007b97a6c321 : SUCCESS in 15m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ef26f6a565e54966a520d9dcaf604006 : SUCCESS in 28m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/61b9ab6cf70d4379b6ca70c44baa5928 : SUCCESS in 20m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6dc3069ba13344c08128542261899726 : SUCCESS in 14m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6f5341b00a764fa48ea19085841ed74a : SUCCESS in 16m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/89087544afc4476783ba15c01db64d4b : SUCCESS in 15m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f241f5d05019477e873a696c6b062311 : SUCCESS in 18m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3df81b2f7bfe4a89ac109381d110b7bf : SUCCESS in 5m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6ebe81c49e6e4b52b0fde80780bd9eea : SUCCESS in 10m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/de5da57dd88541b4ad30a1213033bed5 : SUCCESS in 10m 22s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6c33bd660fbf49f5b429323d005da65a : SUCCESS in 1h 09m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/84bc1206e8934b9aaa74c6cc2960928a : SUCCESS in 1h 24m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f73fbad67a4a4a619641f5cc04477289 : SUCCESS in 1h 55m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a7b33c2d273f44459db83bb45eaa0fd6 : SUCCESS in 1h 06m 21s","accounts_in_message":[],"_revision_number":2},{"id":"317f0ab3e83cd2d38436e35a47112eec4553ef2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:57:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"173961295e11c65e52429566a3741357c9e5f040","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 04:12:09.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4b0cc645f654380b0566b88c95370f1 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b61538b44a8e454ea311e5e818271e23 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95ae65e044c346b69d0fccab17639855 : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91d77eba2ea2432184823990351d0c05 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1362a60eaf654c51a0eb3aa0887525d7 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93a25865b2894884af3298451bcf4c34 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/422b6641404b4791a68e033744f56ead : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1db683ad4991461b99a08c3ac36eee46 : SUCCESS in 7m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5b0cb2284df7443b8526442377adcacc : SUCCESS in 9m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/32f941b206484841874c81f8ccf1a686 : SUCCESS in 9m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4a60a4d12a9f4c369f662b1a3b782ffa : SUCCESS in 8m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/28365325db4a4b57ac488dd9280ed75c : SUCCESS in 4m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cca0d10d8f7d4ff3bbd05c1d880d020c : SUCCESS in 6m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a4d4edc5aa91436e80816c8f866ddb96 : SUCCESS in 10m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/75120d19a0414f909057daef3d030fa4 : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2720bc03cd2745c3a1261c6fd06b58e9 : SUCCESS in 7m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8df6e8714731426d94ae8aac0e5b669d : SUCCESS in 12m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/50b209a102834fdb83b20bcf4740d6ca : SUCCESS in 15m 49s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a6ccdb833f1645d48b341415309c613d : SUCCESS in 22m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8d91134e3e2a48059db8f7d0c7ba159e : SUCCESS in 25m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c3428d53a8164d5fb157110d1313c79a : SUCCESS in 11m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1271a641e7f94b49a42db648586c8a5f : SUCCESS in 13m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4f3785cb96144e70bcb9668dbbbb85d3 : SUCCESS in 14m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b72dba669c0240ecb24e7142356eb90d : SUCCESS in 21m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c931fad7f49f4ef2a264340ddf6da787 : SUCCESS in 6m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7c0aac3c752a42a9ae7eda68abfda534 : SUCCESS in 9m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c1bcb5a6b6704d74a492a72dd8d137cd : SUCCESS in 9m 52s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0e6bc7f4c82d435dbe196359d9e7bdad : SUCCESS in 1h 00m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/44c8e786a0954a7784ce1fa5eb7f1968 : SUCCESS in 1h 22m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d4be8cdf5ead44e2a246a39d8edc8311 : TIMED_OUT in 2h 08m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4e1367ccb89b4237a2696212bf168858 : FAILURE in 46m 41s","accounts_in_message":[],"_revision_number":2},{"id":"659c5fd78a67685515f92e35db03067e8049e943","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-21 04:16:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"524d52ff25700998bef0a3f0f5a1ce42c97473cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 06:33:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd4d0e3bd219432fbf71340c75446296 : SUCCESS in 3m 30s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3227d60aef95460eac1899b61572662f : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/515cd50b34c444a8a8720aeb15b02382 : SUCCESS in 2m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78e0d32b6bb343df8dd70128e92e174a : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edb5b941d4a14a8abd27fd6d2f959bcc : SUCCESS in 2m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8404159552e94c1b923d078485500e66 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8397a1f3c7d74755b870c82df0406d87 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c08c0e9e4f241529d3decfaea6edb54 : SUCCESS in 2m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3504a5f86ca2458d9c15758d897263d1 : SUCCESS in 8m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a6f3ba8d549441da955c89f01912fd4e : SUCCESS in 5m 48s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bda3aec657f04d49ad3ce27ba26299aa : SUCCESS in 10m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/62b8d1c57dec40c3ac42144ba8ea9c98 : SUCCESS in 9m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/56b1d5a76d164f279a345f21f0c11717 : SUCCESS in 4m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a3cb9445eec749c5be7c64985a2c52dd : SUCCESS in 4m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ce0d1acf6f964853b4646ebd0571d8fe : SUCCESS in 9m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/84c41ffb399348d5ace200de65eadb78 : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/188486f454ec48f39f0ca2df36b7f984 : SUCCESS in 6m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d22dff48479b43c9b6cc53b543de6145 : SUCCESS in 14m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9263faaa627a439887909625eaea3bf4 : SUCCESS in 12m 29s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/654348d450ae4616bcde824a9a1b5e02 : SUCCESS in 23m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/47da35b3e4ac46869d0f44c651c260d2 : SUCCESS in 31m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/701af79c58474956877f7ace479c60a4 : SUCCESS in 13m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dcfe3b7844ee4bfbbf709332f608cdfe : SUCCESS in 15m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2f7bc695ca7e44ab8ad4a89f955abb0b : SUCCESS in 17m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f3eb471a7654ea7b1201cc98446b3a9 : SUCCESS in 23m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fdc57e05fee4cc2ab93ec22256869a7 : SUCCESS in 5m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cf414d9b70cb4f8caccfaa17ba370531 : SUCCESS in 9m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4dafbba17b0945259d68edd94a1a42e6 : SUCCESS in 9m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d99e0ca535774aaf9898d84281dbb420 : SUCCESS in 58m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de0ef8d266be4198aec2eea11c636e1d : SUCCESS in 1h 19m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5290c5eff1174828a274e78aef76c029 : TIMED_OUT in 2h 07m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8bd0030b2f3046ceb273e00f73d26c6b : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":2},{"id":"48bad9f3c70ac410811c4de1ad9cd4189ed41f4b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-22 05:12:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"aaf31358d32552c9efc1349958ccc72c49016132","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 07:02:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef362690f40e49cea0c325e12a02f5c3 : SUCCESS in 4m 34s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d207b8b0afaf4f26968a98ccc57516e6 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aaa42ff6f2940768f124f922bee353c : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f69f570107c241239d8cde350b7ccf11 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faf90549f2b2412ea2e46c5440bf6fc5 : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcfc08ea96074e34a4ece709ef7a78e9 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ed2fcf9ba004dc9ab6f83a684c68660 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e0e0faecf4f6411e8971dc4b853aabe7 : SUCCESS in 4m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/39fd2f34d9ae4cbdb939952ea356ea16 : SUCCESS in 6m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ed5aff02328647bb8a343437b7046ef5 : SUCCESS in 8m 05s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/80c73a91540d4d3e91f1dba84c8c5956 : SUCCESS in 8m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/42e7aa0fb9bd4698909e1ac1cd100252 : SUCCESS in 9m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3c85d98525404b8b86d04a45831e2798 : SUCCESS in 5m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/918d11b7cf564840a8e630e234896df9 : SUCCESS in 5m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fcbc03389c1f43a78bf36cec3d09d588 : SUCCESS in 9m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0f0e38409ddc4dcd8eeaae24f2189f6a : SUCCESS in 8m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/91a101c2aaf04bf8ace86e4670250ee2 : SUCCESS in 8m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/55241659e14845b7bef1fe952603ae43 : SUCCESS in 14m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0f1c2da8ec3f496a90da8edf6bf1d99b : SUCCESS in 16m 50s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/87470d11243545b599d70164035961c8 : SUCCESS in 21m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b58861db0ee348a298a782250f02ab8f : SUCCESS in 21m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/626af6bcb4a64560aa6c7e63131b3bf6 : SUCCESS in 14m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0b204459a8f040b3ae606bae2805ef44 : SUCCESS in 13m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/490f7f0453f9469e922b4c4c2142d49e : SUCCESS in 14m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/233e4427288b47febdf7ea5879cc031c : SUCCESS in 19m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ea3625fa8ef4b94acc8b190f31d71fc : SUCCESS in 6m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/797f2862224d4484b6272e7dffd48c3c : SUCCESS in 12m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cfaad2af10d044c585c3d58a21feadcf : SUCCESS in 11m 31s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e0224d0a469e4b86af382ec846abf994 : SUCCESS in 1h 17m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/06c05d94d28d4130b87b52d4fb4bef92 : SUCCESS in 1h 09m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6217e4d4f6184b93a5fc9fede3cee016 : SUCCESS in 1h 44m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4127be80e9cf44d09198fd6010b141f7 : SUCCESS in 1h 19m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f34a6998e0b190771a37c7d3d0fb479b3ad828f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 07:02:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b9059f2df4f9a823cffb6e0f79ce8509d5b0cf53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 09:01:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f64670e09872b46ad5dcf03e7b7e8dfa9503e771","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 09:01:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9f4e81450b4097a9efbc32c4640c3c : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb46fd12e2b4626943b26d7c87f9288 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d559ced148194f89bd49a99a757b7fb7 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/712fe62e6cfc4c5ab623992b1140414e : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2104c88cf1b347cf99f615ae8b3a3437 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4348ab5d894b4eb2b7a16fb2833897e3 : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0af4e04f5a846b4b65506e496aad87d : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d893f25e25d3463b8abf28b7c2a40486 : SUCCESS in 7m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/560c4126619646818d637e4c79cadf43 : SUCCESS in 6m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ff6cce7c7d4740deb70445c9dd3de1b6 : SUCCESS in 13m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1b4a3283a41a446aa9d458af23900660 : SUCCESS in 9m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/04af29d0f55c43e3b47c93245b21377f : SUCCESS in 6m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/550547745194485cadd2032fc1234d32 : SUCCESS in 4m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4af89a0292d04a27857e207957c0cf2b : SUCCESS in 11m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a94a917df8d54c8ab6cb9ff9bf03121f : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b17526eda27b47288f2204d68a47377b : SUCCESS in 15m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6ba72fe976e44521ab4662de00dd3171 : SUCCESS in 15m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ca09e7743cc14d209f7e8becf08148b0 : SUCCESS in 15m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/41599d920e444dd6bf4508c60ca396f9 : SUCCESS in 24m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c1a83bf8cd3149389e5721a870696722 : SUCCESS in 25m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5ad3b491c6ee4a90ad76a558dab06a56 : SUCCESS in 15m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/667a8a1183914258a4ac8c20e732540c : SUCCESS in 15m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c3c199a92a9747babc3f4d720d94b113 : SUCCESS in 17m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b7fe4e7c68b3473dacdbb5139e5fdc71 : SUCCESS in 23m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b049ecc782145e7bbe38de3e0079f8f : SUCCESS in 6m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/678ee712fd7647e483ffeba19f0ac187 : SUCCESS in 10m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c3c0edb630724ab8b51ea05609070790 : SUCCESS in 12m 29s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/370ff56c53d946ee85b1bbe68b99667b : SUCCESS in 57m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b709799581174bcb875d1a778ca1bac7 : SUCCESS in 1h 08m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bc5dbc92032944509e27abc0ce5ed554 : SUCCESS in 1h 36m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2b2e24365a0742a1a89ecc221388dfc0 : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":2},{"id":"2fadd4aa6fb61d71f5b4b720655cafc2462e8e9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 09:02:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0ab1674314449af8676d5a20c13c466 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"57c9757c93f383a6b4440deca24ade61cc90a774","revisions":{"2f2781a5767559f0c8d98136ba2f831c1352bed0":{"kind":"REWORK","_number":1,"created":"2019-09-18 19:43:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/682980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/682980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/682980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/682980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/682980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/682980/1"}}},"commit":{"parents":[{"commit":"c14b4c55d0d2e3e9605f0612032f40304d96a464","subject":"update constraint for os-win to new release 4.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c14b4c55d0d2e3e9605f0612032f40304d96a464"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-18 19:43:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-18 19:43:03.000000000","tz":0},"subject":"update constraint for os-apply-config to new release 10.4.1","message":"update constraint for os-apply-config to new release 10.4.1\n\nChange-Id: I0be0ff85dc8a5fdf326f076bf4b3430dedf0f5c0\nmeta:version: 10.4.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Change-Id: I7c003d3e285e86a174baef7c2749518972262932\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2781a5767559f0c8d98136ba2f831c1352bed0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2781a5767559f0c8d98136ba2f831c1352bed0"}]},"branch":"refs/heads/master"},"57c9757c93f383a6b4440deca24ade61cc90a774":{"kind":"REWORK","_number":2,"created":"2019-09-19 16:21:11.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/80/682980/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/682980/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/682980/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/682980/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/682980/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/682980/2"}}},"commit":{"parents":[{"commit":"c86c9550a62df0ee37360ed611fc40b2f4ae0e87","subject":"Merge \"update constraint for tripleo-common to new release 11.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c86c9550a62df0ee37360ed611fc40b2f4ae0e87"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-18 19:43:03.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-09-19 16:20:50.000000000","tz":-300},"subject":"update constraint for os-apply-config to new release 10.4.1","message":"update constraint for os-apply-config to new release 10.4.1\n\nChange-Id: I0be0ff85dc8a5fdf326f076bf4b3430dedf0f5c0\nmeta:version: 10.4.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Change-Id: I7c003d3e285e86a174baef7c2749518972262932\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57c9757c93f383a6b4440deca24ade61cc90a774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57c9757c93f383a6b4440deca24ade61cc90a774"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
