)]}'
{"id":"openstack%2Frequirements~683108","triplet_id":"openstack%2Frequirements~stable%2Fstein~I4498605bb85be2cde8c1cc83937ed41d5a1d69b9","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I4498605bb85be2cde8c1cc83937ed41d5a1d69b9","subject":"update constraint for os-win to new release 4.2.1","status":"MERGED","created":"2019-09-19 10:44:55.000000000","updated":"2019-09-20 02:39:21.000000000","submitted":"2019-09-20 02:39:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683108-1568947161910-8c653045","meta_rev_id":"2da4db872280bdb3002f1f5d149e915cba154c67","_number":683108,"virtual_id_number":683108,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-20 02:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-20 01:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-20 01:02:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 01:02:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-20 02:39:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4cf92357ce0451da65f1bd98ce4b4ba7439bfc67","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-19 10:44:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7176c653deb5065fb90e01e7d112b056d8e957fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 23:19:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7b68ea3c5a0a476da3d487a6311f3f0b : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81980167a27949af8191676ab1a974a3 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe0a0f685a604371b67334377d1e606b : SUCCESS in 3m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/97dd7109a278493587dddc2d4f3cceac : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff54d1a158764263bf8f9653a729383b : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fc8771bd26144f31ac5f3f7585f592d6 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e5d4696f4fc54de99cae25d6fed6af58 : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/52c102f40e7948ac93c21e4faf659dea : SUCCESS in 11m 59s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/350453cd6cad4c66821bee36331f5749 : SUCCESS in 24m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b0146e6e7aec4272a37c3639e73d8d98 : SUCCESS in 10m 49s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/157c0513d0b04964972807077627bf98 : SUCCESS in 11m 03s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/c0bc62e644d244b4a0eada3219a8f79c : SUCCESS in 9m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/de5a83eec24c41c499920ecc34d1c8d4 : SUCCESS in 4m 43s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/f4c4a846cf0744c3b3f91165e2216d50 : SUCCESS in 6m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b98088aff25049cfbc3d86ab8684cce8 : SUCCESS in 11m 23s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/38a515e4f12640ebb676ee0212cadba8 : SUCCESS in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d347cc8e13654b5eb0d2f2fb1ab03751 : SUCCESS in 6m 58s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a035c2ddb7f54b769754057ee7c30997 : SUCCESS in 18m 48s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/81c55de34f8d4f9ba4ecb05cc0bf1e3a : SUCCESS in 24m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b31a1f5c9eb447d5a43772ccac1243ba : SUCCESS in 24m 50s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/1be4f53c0ab544059214bc963b30bbc9 : SUCCESS in 25m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2bedd287b8c34c34b9a89cf6cb52b90d : SUCCESS in 12m 50s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/819e326114e6408d8762ed7b2a9b36e6 : SUCCESS in 14m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8353f66a0ae74b0d941dd6dec4b050df : SUCCESS in 18m 30s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/d6d35bffdfef457ab5839197e5da6237 : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/21e1ba096c6049258750099a1c908a50 : SUCCESS in 9m 23s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/092668a693c142bab90b94a40cde083d : SUCCESS in 8m 03s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5fc0dd94a44c4fe397ceb655be76d10e : SUCCESS in 1h 08m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0082cf159ca7421fab2f0bcab1e533e2 : SUCCESS in 1h 13m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/debe6963daa646ec90de49616fe85ca3 : SUCCESS in 1h 31m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1a12330f146540549351ba7da6d4640f : SUCCESS in 1h 05m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5db75a7db50556f29ac2833e6ea5782aca89410d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-20 01:02:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f51b9aad6e6e92a157f8a306fe9ec676f024781","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 01:02:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"239102b1ff276f49ee29aff0d4136816f877b477","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 02:39:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5a99bdb179a0850d5fd90005414e9128d9c0dc44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 02:39:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ead89f985904255a845674898ffd546 : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afb66f78c5ef4234b78136cfc3d056b3 : SUCCESS in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7382c6795b244ef09706bcb382af5060 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c22a24b67cbe4354ac7474ff361f5ef5 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56285c7069d341b59e7a39328b80998d : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e059511dff74cae9332bea3a2340471 : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/35ee52d31d7f4b55975a281bc804947e : SUCCESS in 6m 22s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/0d560259964f4e83b09b33ee0355face : SUCCESS in 7m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9a5eb3d9482340da8088fbec69ab793f : SUCCESS in 9m 23s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/f91245526f0e4cc78a1b80ed1c05bf2e : SUCCESS in 9m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f99f57554e924f1ea4a0e022c49b30b5 : SUCCESS in 4m 37s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/3e7a5e8a180c481ab5dd3d6d4161223f : SUCCESS in 6m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a698379525724b44be93c4430294af87 : SUCCESS in 10m 54s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/294f5ccd4f84472aa8584ae0a4e7285c : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6f2858246cc423b93f637cd79299793 : SUCCESS in 9m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e5a3101c436e4c8ba91e993f227bcb51 : SUCCESS in 23m 46s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/daf47c4a5c5544b0bcf1712f21c4919a : SUCCESS in 25m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7689465c10ca4d9f9265e18d56786a50 : SUCCESS in 23m 26s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/be00e59ad7874b839ad7505fab8fefed : SUCCESS in 24m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6ce047bbd0e44d41b9c6291e674932f7 : SUCCESS in 16m 22s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/2037dccd911745658463fc21e29a02ae : SUCCESS in 14m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2a23e749b0544b7184c7c62458e684a2 : SUCCESS in 21m 30s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/f33434800473486d899460f5904061ce : SUCCESS in 6m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8a9c6aa8139842629084f143b3b9f2c3 : SUCCESS in 8m 07s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/08721822665842e49adb49d6e8d31c69 : SUCCESS in 7m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1a69d137406a49ae8d9f5ae6c0980761 : SUCCESS in 1h 02m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1271879ac8a24666b3aecbe8d2486ff2 : SUCCESS in 1h 03m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8b0b063df3bf422eaa6c1251bcb97378 : SUCCESS in 1h 28m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/69c1d2d2bad542f08e0ed3cc55578a18 : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb865f8601cbefe214d0aeece736b67c2446829c","revisions":{"eb865f8601cbefe214d0aeece736b67c2446829c":{"kind":"REWORK","_number":1,"created":"2019-09-19 10:44:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/683108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/683108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/683108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/683108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/683108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/683108/1"}}},"commit":{"parents":[{"commit":"dca6e6850d4a79a6ef52fabfde81b368ef28cbae","subject":"Merge \"update constraint for os-refresh-config to new release 10.2.1\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dca6e6850d4a79a6ef52fabfde81b368ef28cbae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-19 10:44:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-19 10:44:53.000000000","tz":0},"subject":"update constraint for os-win to new release 4.2.1","message":"update constraint for os-win to new release 4.2.1\n\nChange-Id: I4498605bb85be2cde8c1cc83937ed41d5a1d69b9\nmeta:version: 4.2.1\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Lucian Petrut \u003clpetrut@cloudbasesolutions.com\u003e\nmeta:release:Commit: Lucian Petrut \u003clpetrut@cloudbasesolutions.com\u003e\nmeta:release:Change-Id: I7fc8c89b3bb684ffdb434587125ab48fc68dd21d\nmeta:release:Code-Review+1: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb865f8601cbefe214d0aeece736b67c2446829c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb865f8601cbefe214d0aeece736b67c2446829c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
