)]}'
{"id":"openstack%2Frequirements~684192","triplet_id":"openstack%2Frequirements~master~Iec1f21e544b6efc17401c0cb9ff7472f020a7dc3","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Iec1f21e544b6efc17401c0cb9ff7472f020a7dc3","subject":"Skip alembic 1.2.0","status":"MERGED","created":"2019-09-24 05:07:31.000000000","updated":"2019-09-25 18:58:59.000000000","submitted":"2019-09-25 18:55:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684192-1569437721108-dadc74e4","meta_rev_id":"f273f9f20e772031d247b0bf4c03d41338ca5fe2","_number":684192,"virtual_id_number":684192,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 18:55:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-24 16:47:56.000000000","_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-24 17:06:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-25 17:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-25 16:29:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 16:29:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-25 17:31:05.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-09-25 18:55:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a592507ab56044664e6986ab75015ebaa9aef6e","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-24 05:07:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ec54fc7b4e3971db96080c2630a31365231abd8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-24 05:15:44.000000000","message":"Patch Set 1:\n\ncapping it is a very strong way to go here, are you sure it won\u0027t be fixed in 1.2.1?  It\u0027s also very late in the cycle to be changing reqs like this (let alone a core one)","accounts_in_message":[],"_revision_number":1},{"id":"101cb6aaf1e48db741d49399153d709f7078b8da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-24 05:24:37.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"f030c54936f1057da4006420833d170408ad7681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 06:54:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81c95491a69948efbf44bfb74cc1248c : SUCCESS in 3m 06s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/df8af167348048e89c72df3138af0425 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00a747370d2a4d539ab61b2a25584384 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae6edcbdac5a443e853e92352a68bb79 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2586c11fcf449bc8c114f91c166f73b : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b63158c8923c49ac89868ab1c5ba6e90 : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2073f9ca2b014a30b07c81e135c8b92f : SUCCESS in 2m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4aa032f68e6f40c5b3e05d83cf5195c0 : SUCCESS in 1h 14m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf72fec20c79491da66794da727a0413 : SUCCESS in 1h 03m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1777021a02624873abb0f32fba74341e : SUCCESS in 1h 43m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/041a37aa4f9a4988809430e3a40b3c3f : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":1},{"id":"39af8a08980c10d0b64d53a3ffb6c724a411bfa3","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-24 06:59:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7147eec9bc77deac82782e59bf7b399e259dd826","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-24 06:59:54.000000000","message":"Patch Set 1:\n\n\u003e capping it is a very strong way to go here, are you sure it won\u0027t\n \u003e be fixed in 1.2.1?  It\u0027s also very late in the cycle to be changing\n \u003e reqs like this (let alone a core one)\n\nUpdated to not allow 1.2.0 . I\u0027ll proposed a github PR to restore the 1.1.X behavior for the stamp() method. Let\u0027s see if this will be fixed for 1.2.1","accounts_in_message":[],"_revision_number":1},{"id":"fcdc1e58cd3ab1b87655052d34648bc94b16f299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 12:34:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b92dc2616c4b368583bbc0c22cd6e9 : SUCCESS in 3m 49s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a49a1a44131745fc98dbc6f81b9bb480 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d4df87b52b14a649ff5e212fee22cdc : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27a548c1fd08474eac35bb9ad728f042 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad91871abca2481f8c01edd48b2ff26f : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25de1f165cb44e2eb615161a5313d76d : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c1f4484657c4c9ba1584e2d38e859ac : SUCCESS in 4m 46s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/946b176b7b6448eca52f04a5d1ad012b : SUCCESS in 1h 00m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a48534df0594411eb745992a8ec1dfad : SUCCESS in 1h 26m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5562358909c64370a8ae1a73c8d37004 : SUCCESS in 2h 00m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2f340f954b4945a5a6b643ff096ba497 : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":2},{"id":"1df37e533e7b7399a1df75411547bb9d0864c324","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-24 16:47:56.000000000","message":"Patch Set 2:\n\nLooks like 1.2.1 will fix the problem. See https://gerrit.sqlalchemy.org/#/c/sqlalchemy/alembic/+/1477/","accounts_in_message":[],"_revision_number":2},{"id":"10e649051224e083772a18a43489ce9d2fd2299b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-24 17:06:54.000000000","message":"Patch Set 2: Code-Review+2\n\nWe should get this in now, before stuff gets rc tagged","accounts_in_message":[],"_revision_number":2},{"id":"02a3c297372b18213147623fc3757b4ed9f501dd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-25 16:29:54.000000000","message":"Patch Set 2: Workflow+1\n\ndone waiting","accounts_in_message":[],"_revision_number":2},{"id":"9019ecf7f0b01c38f1fa3231d7bc2decd5c08fc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 16:30:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"69618dd975ce912edbaf7be5fdb014efd467d02c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-09-25 17:31:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9ef4303df61a3c5cce79c50851039f5d6f7ea385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:55:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0b8ec9cba3b4ca99b387b44f19e8733 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96e0903935ce4846bc5cd59a95c53624 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/619a7b1afa23432bb68546d332359f80 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00f2564dc1e14a219f70421e7ef210ee : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a0bd7058d08436da52ca4d6696044ae : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a918d55c96a84cedaf06f40e1b9ad740 : SUCCESS in 5m 43s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/422c24fbb5b84d7a8616eb3952610052 : SUCCESS in 54m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/95e96d711f6146b984f1a01fadbfa9d4 : SUCCESS in 1h 26m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f45ffccaeb2244819497c65a1aa0a9ff : SUCCESS in 1h 53m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5409a2b3aa8244bfb82926aeb84b52f2 : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":2},{"id":"b1ed185dcc8d210e9c219436edb5c068471602ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:55:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f273f9f20e772031d247b0bf4c03d41338ca5fe2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:58:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac352a946be4dcc84483313cdaddb37 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a9472f74bdb73afe4a412519d0291d1ceb13851","revisions":{"f37dd65355de9667d385c7aba8ad38ae71adc7df":{"kind":"REWORK","_number":1,"created":"2019-09-24 05:07:31.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/92/684192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/684192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/684192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/684192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/684192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/684192/1"}}},"commit":{"parents":[{"commit":"bf19087eac2578f85bc7f2ef09e9e88b0eb610c5","subject":"Merge \"update constraint for os-apply-config to new release 10.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf19087eac2578f85bc7f2ef09e9e88b0eb610c5"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-09-24 05:06:02.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-09-24 05:07:29.000000000","tz":120},"subject":"Cap alembic to \u003c 1.2","message":"Cap alembic to \u003c 1.2\n\nVersion 1.2.0 changed the alembic.command.stamp() method which no\nlonger accepts a \"revision\" parameter (not it accepts a \"revisions\"\nparameter).\nUpstream bug is: https://github.com/sqlalchemy/alembic/issues/601\n\nChange-Id: Iec1f21e544b6efc17401c0cb9ff7472f020a7dc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f37dd65355de9667d385c7aba8ad38ae71adc7df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f37dd65355de9667d385c7aba8ad38ae71adc7df"}]},"branch":"refs/heads/master"},"8a9472f74bdb73afe4a412519d0291d1ceb13851":{"kind":"REWORK","_number":2,"created":"2019-09-24 06:59:14.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/92/684192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/684192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/684192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/684192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/684192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/684192/2"}}},"commit":{"parents":[{"commit":"bf19087eac2578f85bc7f2ef09e9e88b0eb610c5","subject":"Merge \"update constraint for os-apply-config to new release 10.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf19087eac2578f85bc7f2ef09e9e88b0eb610c5"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-09-24 05:06:02.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-09-24 06:58:44.000000000","tz":120},"subject":"Skip alembic 1.2.0","message":"Skip alembic 1.2.0\n\nVersion 1.2.0 changed the alembic.command.stamp() method which no\nlonger accepts a \"revision\" parameter (not it accepts a \"revisions\"\nparameter).\nUpstream bug is: https://github.com/sqlalchemy/alembic/issues/601\n\nChange-Id: Iec1f21e544b6efc17401c0cb9ff7472f020a7dc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a9472f74bdb73afe4a412519d0291d1ceb13851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a9472f74bdb73afe4a412519d0291d1ceb13851"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
