)]}'
{"id":"openstack%2Frequirements~684376","triplet_id":"openstack%2Frequirements~master~I1e69a66460627c47b9cb86c228fe2bdc08dc061e","project":"openstack/requirements","branch":"master","topic":"ironicclient","hashtags":[],"change_id":"I1e69a66460627c47b9cb86c228fe2bdc08dc061e","subject":"Block python-ironicclient 3.0.0","status":"MERGED","created":"2019-09-24 15:17:02.000000000","updated":"2019-09-25 18:57:02.000000000","submitted":"2019-09-25 18:55:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684376-1569437719183-e985b01a","meta_rev_id":"3e7059e50585977723618b9853594269ba6ff265","_number":684376,"virtual_id_number":684376,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 18:55:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-24 17:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-09-24 15:22:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":1,"date":"2019-09-25 08:19:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-09-24 15:20:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-25 16:29:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 15:20:41.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-09-24 15:22:10.000000000","updated_by":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"reviewer":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"state":"REVIEWER"},{"updated":"2019-09-25 08:19:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-25 16:29:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-25 18:55:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e4f48aafa2391c9279198b3a15b084b512c5ea8","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-24 15:17:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85c0d6507d1954fcca3ccc97ed14a628ea895aa9","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-09-24 15:20:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"24b335cc52e368071a83def603df48ee7558008c","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-09-24 15:22:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e66a8429094de2e903fc1c43d337e9196a0dd189","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-24 17:28:00.000000000","message":"Patch Set 1: Code-Review+2\n\nwe should get this out before RCs","accounts_in_message":[],"_revision_number":1},{"id":"71f3782aeb2e3971666d012f01e3c5b828eeafb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 21:52:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de36201adb2a41fb9e235ceaf02e205d : SUCCESS in 5m 44s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b6a71a7db1c042049ff4f64cbf076a84 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1421833e886c4190aad1d17c4ee93a52 : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6454ee01bb5d4714982f7d28ec7f0170 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1b4c832119f468687770a619d5ca6c1 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5edeb484ba0645de8ef0d50ff37111a7 : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4dc66b42f3414f7c96ddec4c4b6766cb : SUCCESS in 2m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1fcea52602a48d4a77dfd42557876ac : SUCCESS in 2m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c3e0dc311bf64fd9800d3cb21ff3eebe : SUCCESS in 6m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/67da0103ba9a401da79039f53db3b41c : SUCCESS in 17m 11s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2a67922c837344279b1ea994a963d1bb : SUCCESS in 9m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fe4ce66dbfd94d5fa8f9b506e3aa9019 : SUCCESS in 8m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9897751c1f374a39b362f13d844beeb3 : SUCCESS in 5m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5252ab2800c2430abfafb4d18ff13f8e : SUCCESS in 4m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/09c62574799a49b0b229e22aafed1bd9 : SUCCESS in 10m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/45fe400f64264d7e98cd034c1f753571 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c43bb6bad10b47d09a39044bd9d7ad5c : SUCCESS in 12m 33s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2202f48fba304706af30782ff6734d16 : SUCCESS in 12m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f3ac655bfa0d4e438bdd7a2f218d871e : SUCCESS in 14m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4dbfcfb04e894429bbc23b2bdab346ac : SUCCESS in 22m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5f3e7cb8a1f1474b9409760a1569ed51 : SUCCESS in 22m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/813f9c4945c74bcdb213fe75cdb22233 : SUCCESS in 13m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7f04b432ab7c47789cf2c393cc0d8ba1 : SUCCESS in 10m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/58f6a75f65f84f5584181c245ee4f034 : SUCCESS in 15m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/83f4413615514d6491a92d940dab39d8 : SUCCESS in 30m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7dc32aa48e70423bbffaa4bdb4571620 : SUCCESS in 5m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c10224e5bd8a4a0fae7059c0afd25a4f : SUCCESS in 9m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c8e3fe474c04420f96d6310417b7ecf4 : SUCCESS in 8m 20s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d67f973ab4744b71b991c578c9f594d6 : SUCCESS in 1h 00m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb4e397427d4481ba23a9b5b2f30cfef : SUCCESS in 1h 14m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f91ba89aad9d45b48458e41c3d7efc65 : SUCCESS in 1h 48m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2c6b57cf0e49477f8d6a57aa7423d001 : SUCCESS in 1h 01m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5968b981e60dcb81c045752207fc1a3979859a27","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-25 08:19:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"727e990667c3e7ea2f368976246d38337733256c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-25 16:26:55.000000000","message":"Patch Set 1:\n\nironicclient 3.1.0 has now been released.","accounts_in_message":[],"_revision_number":1},{"id":"d7a4ad70b1c4fe7f2a148efb0db993c70bc54509","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-25 16:29:44.000000000","message":"Patch Set 1: Workflow+1\n\ndone waiting","accounts_in_message":[],"_revision_number":1},{"id":"1bbd4ea4f46020b96f6ebae3ed1535c0b43a1488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 16:30:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26e0e1192e1d56cff0d2835890cac9c1d9bf53c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:55:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf432ed823f489d8d5a790289aea087 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7b68d89f72846ae92c742702f5b3e8c : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8866741f7fa4d2d92fb829ce4436784 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07d6e806d195406b8d6b0c15710dd709 : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2086f989c41741ca9384ab61be892822 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e02808cc71d0420dbf26f2dd6addf51c : SUCCESS in 8m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9d6a30b553b4442b272ae2de1c48851 : SUCCESS in 3m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d9709634ef124706a618765701f129a7 : SUCCESS in 9m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c62317ec1fac4fdeb4ad71f2679b8f79 : SUCCESS in 8m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1d13856153bd40eb94a97daae2d26750 : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/52595442263f48b38f6c49340cb9015e : SUCCESS in 10m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8aa81fcb269749b5afcb1ada478d6785 : SUCCESS in 4m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/925be61539a84b7eaca46f09ccf034b9 : SUCCESS in 7m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4ac4863c3b984411afbcb7cc432e8182 : SUCCESS in 11m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/54a9112a978a4b829f20c310a2f100f9 : SUCCESS in 9m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/12250a3a265643bca26234958487e9af : SUCCESS in 12m 35s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9b829381d6e74db2b64deaf8135718fd : SUCCESS in 15m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/90cdeecc2a1947feaf3631b165820d8e : SUCCESS in 13m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f6ce93c3fb864b93891e8dada865f98c : SUCCESS in 30m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7cd3e7d286d04f06aff3eaa3f59287b4 : SUCCESS in 23m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/60186b0755554727bd6223d9aa89c8c0 : SUCCESS in 15m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/39b1065a7f2b4646bc24440cba019742 : SUCCESS in 13m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f07f1c2abcef41929b17d136b249e934 : SUCCESS in 17m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/abd20dea555649118f33a1486bf679cf : SUCCESS in 22m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f349b0b1cec4b84a0551ad2062820e2 : SUCCESS in 11m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b58bfffc445e41009330babf5a7fdf79 : SUCCESS in 11m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/29906f29bbf04952af161f9f915bd5b3 : SUCCESS in 8m 39s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a5ac94a220e64791a9585d8f3e4ca2ec : SUCCESS in 1h 15m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a14825ba11b341c88e72ab895b388aa0 : SUCCESS in 1h 26m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cd31f06f9b8c46f2b6883b2a9840ff68 : SUCCESS in 1h 58m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f00e74a0b5434835a218689a01467830 : SUCCESS in 1h 07m 07s","accounts_in_message":[],"_revision_number":1},{"id":"72e7c95bbc01a19ea88d09a3a52d7b45e3f34ade","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:55:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5937baa9ec81d7b95bb35f5edb0aa8a40aa75a35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:57:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98320ccdc744e9ba5a3e1aaa5d54ed0 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da7a9a044833422997ce7007c02f623be20a26cc","revisions":{"da7a9a044833422997ce7007c02f623be20a26cc":{"kind":"REWORK","_number":1,"created":"2019-09-24 15:17:02.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/76/684376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/684376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/684376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/684376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/684376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/684376/1"}}},"commit":{"parents":[{"commit":"bf19087eac2578f85bc7f2ef09e9e88b0eb610c5","subject":"Merge \"update constraint for os-apply-config to new release 10.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf19087eac2578f85bc7f2ef09e9e88b0eb610c5"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-24 15:15:35.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-09-24 15:16:42.000000000","tz":120},"subject":"Block python-ironicclient 3.0.0","message":"Block python-ironicclient 3.0.0\n\nIt has a critical issue in no-auth mode, breaking bifrost and\nironic-inspector functional tests. See story 2006600 for details.\n\nChange-Id: I1e69a66460627c47b9cb86c228fe2bdc08dc061e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da7a9a044833422997ce7007c02f623be20a26cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da7a9a044833422997ce7007c02f623be20a26cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
