)]}'
{"id":"openstack%2Frequirements~686469","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ic578bc4d162aaad3c616cce14d5a7de8eefbd8f7","project":"openstack/requirements","branch":"stable/stein","topic":"taas-loci","hashtags":[],"change_id":"Ic578bc4d162aaad3c616cce14d5a7de8eefbd8f7","subject":"[Do-not-merge]: Bring in TaaS components to requirements","status":"ABANDONED","created":"2019-10-03 19:23:53.000000000","updated":"2020-02-10 06:05:21.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"55df6da6e116759550c2e30f1dccae0e9501c5b9","_number":686469,"virtual_id_number":686469,"owner":{"_account_id":27589,"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","username":"deepak.tiwari"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-03 20:54:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-2,"date":"2019-10-03 22:59:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2019-10-03 22:59:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 19:25:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-03 20:54:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-03 22:59:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"917868c0fd2de3d46512a096edf4e360e5d1df62","author":{"_account_id":27589,"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","username":"deepak.tiwari"},"date":"2019-10-03 19:23:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcd1bd519df670a93ac5259e5d3ca4076ef48085","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-03 19:24:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f089227ff1571b7817387246e38b70abe885571","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-03 19:25:01.000000000","message":"Patch Set 1:\n\nAnd be merged to master and backported (if appropriate for stable backport).","accounts_in_message":[],"_revision_number":1},{"id":"a9d21b35155597606abc4449ebe875e7d9c142c8","author":{"_account_id":27589,"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","username":"deepak.tiwari"},"date":"2019-10-03 19:27:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fbf527e10fd357eb88cddde6e4ffac9c1b9e4e22","author":{"_account_id":27589,"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","username":"deepak.tiwari"},"date":"2019-10-03 19:38:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd6d2de3df7d39e463b0f3178a8e5a4bd26a221f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 20:54:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6ad0eb6547264671a296742d6cb6ca80 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5a2cd3cee6146de9074a610e0090ea1 : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c4889c8042f4bdba7b559a8dd5df281 : SUCCESS in 3m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/36803bff72ac403293a2685e79adfbfe : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8d0efdef73d440197287c8efe5db709 : SUCCESS in 2m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6fcbbe1520c4479d82db26ed9368effd : FAILURE in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eca93826a6c442ffbf64845ce624e578 : SUCCESS in 2m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0077f9f27c894ddebb88aa404ae7a647 : FAILURE in 2m 42s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/74e4c34138ec48d59f149297acf8905b : FAILURE in 7m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0ebcf3b3488f46efbf8570de1c3c0572 : FAILURE in 2m 43s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a21fd8b6753e4709b4d2b7776458b251 : SUCCESS in 8m 25s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/a9595e366d1948f9b31566ecc08d761f : SUCCESS in 8m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5a706c7f3ff84789bb39714780703e2b : SUCCESS in 4m 29s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/f08f32e68b0e4835a48161b43e513c12 : SUCCESS in 4m 53s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c3ba2a4472b74edf8f2604a7abda9d99 : SUCCESS in 10m 24s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/c9b3aff43b7546d2bab548ba0050e7e6 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b7e5d6093d74333ba66026c4707fac6 : SUCCESS in 6m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/880b69ec16934d89bde85cb52ffbfff0 : SUCCESS in 24m 05s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/22ecc65477dc4cf4a0418c53d9237d9f : SUCCESS in 23m 02s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9a9919c1754845d48f31a63fe67d9922 : SUCCESS in 24m 37s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/0072c975863047d380f21e3a6c226b73 : SUCCESS in 23m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/523cdc6d6487423791483cf891bb9d4f : SUCCESS in 13m 39s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/3eed5e3502d149f89568f6d244475ea0 : SUCCESS in 13m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3031e051301b441cbdd8c97313cfb373 : SUCCESS in 19m 16s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/5e05217d972941dc8c82b57f74ca782c : SUCCESS in 4m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/905c6e249ab8491081d7ffa6356bd4b8 : SUCCESS in 8m 02s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/86036e06b23743b2b49a66a703917584 : SUCCESS in 7m 11s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/bda90bbd4ba2418da55e3ad0dcb3371a : FAILURE in 21m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e6424b5ac4945f297c9f1d0f4337e1b : SUCCESS in 1h 10m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a711714645814beeab740687b9b8c0a0 : FAILURE in 17m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4aaffa7bf3b14b17812a84074be8782f : SUCCESS in 59m 19s","accounts_in_message":[],"_revision_number":3},{"id":"2009a5cff2641ddfc1bc0c0d00fa4234c5d502ee","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-03 22:59:43.000000000","message":"Patch Set 3: Code-Review-2 Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"23b6ead753f9795714e4b90057a8c021854089dd","author":{"_account_id":27589,"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","username":"deepak.tiwari"},"date":"2020-02-10 06:05:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a7f7f1cd8635649b0d1ac720e64785b1230d92aa","revisions":{"4318a734bc40a05b6d5888b5d070be4cd794db1a":{"kind":"REWORK","_number":1,"created":"2019-10-03 19:23:53.000000000","uploader":{"_account_id":27589,"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","username":"deepak.tiwari"},"ref":"refs/changes/69/686469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/686469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/686469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/686469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/686469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/686469/1"}}},"commit":{"parents":[{"commit":"eb865f8601cbefe214d0aeece736b67c2446829c","subject":"update constraint for os-win to new release 4.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb865f8601cbefe214d0aeece736b67c2446829c"}]}],"author":{"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","date":"2019-10-03 19:19:32.000000000","tz":-300},"committer":{"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","date":"2019-10-03 19:19:32.000000000","tz":-300},"subject":"[Do-not-merge]: Bring in TaaS components to requirements","message":"[Do-not-merge]: Bring in TaaS components to requirements\n\nBring in TaaS components to requirements image.\n\nChange-Id: Ic578bc4d162aaad3c616cce14d5a7de8eefbd8f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4318a734bc40a05b6d5888b5d070be4cd794db1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4318a734bc40a05b6d5888b5d070be4cd794db1a"}]},"branch":"refs/heads/stable/stein"},"d9458bda54560694b9566fc6b5af32199e56fa1f":{"kind":"REWORK","_number":2,"created":"2019-10-03 19:27:33.000000000","uploader":{"_account_id":27589,"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","username":"deepak.tiwari"},"ref":"refs/changes/69/686469/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/686469/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/686469/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/686469/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/686469/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/686469/2"}}},"commit":{"parents":[{"commit":"eb865f8601cbefe214d0aeece736b67c2446829c","subject":"update constraint for os-win to new release 4.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb865f8601cbefe214d0aeece736b67c2446829c"}]}],"author":{"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","date":"2019-10-03 19:19:32.000000000","tz":-300},"committer":{"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","date":"2019-10-03 19:23:35.000000000","tz":-300},"subject":"[Do-not-merge]: Bring in TaaS components to requirements","message":"[Do-not-merge]: Bring in TaaS components to requirements\n\nBring in TaaS components to requirements image.\n\nChange-Id: Ic578bc4d162aaad3c616cce14d5a7de8eefbd8f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9458bda54560694b9566fc6b5af32199e56fa1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9458bda54560694b9566fc6b5af32199e56fa1f"}]},"branch":"refs/heads/stable/stein"},"a7f7f1cd8635649b0d1ac720e64785b1230d92aa":{"kind":"REWORK","_number":3,"created":"2019-10-03 19:38:26.000000000","uploader":{"_account_id":27589,"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","username":"deepak.tiwari"},"ref":"refs/changes/69/686469/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/686469/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/686469/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/686469/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/686469/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/686469/3"}}},"commit":{"parents":[{"commit":"eb865f8601cbefe214d0aeece736b67c2446829c","subject":"update constraint for os-win to new release 4.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb865f8601cbefe214d0aeece736b67c2446829c"}]}],"author":{"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","date":"2019-10-03 19:19:32.000000000","tz":-300},"committer":{"name":"Deepak Tiwari","email":"deepak.tiwari@aricent.com","date":"2019-10-03 19:34:31.000000000","tz":-300},"subject":"[Do-not-merge]: Bring in TaaS components to requirements","message":"[Do-not-merge]: Bring in TaaS components to requirements\n\nBring in TaaS components to requirements image.\n\nChange-Id: Ic578bc4d162aaad3c616cce14d5a7de8eefbd8f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7f7f1cd8635649b0d1ac720e64785b1230d92aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7f7f1cd8635649b0d1ac720e64785b1230d92aa"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
