)]}'
{"id":"openstack%2Frequirements~687064","triplet_id":"openstack%2Frequirements~master~Iab5cf83ea2125ad6d950743339eea791f4a1ad7e","project":"openstack/requirements","branch":"master","topic":"stable/train","hashtags":[],"change_id":"Iab5cf83ea2125ad6d950743339eea791f4a1ad7e","subject":"update constraint for tempest to new release 22.0.0","status":"MERGED","created":"2019-10-07 14:36:15.000000000","updated":"2019-10-07 21:58:58.000000000","submitted":"2019-10-07 21:57:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687064-1570485426212-01385176","meta_rev_id":"1621e6d67504e3ae20bc3b210b0379465753b2ad","_number":687064,"virtual_id_number":687064,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-07 21:57:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-07 19:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-07 19:54:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-07 19:54:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-07 21:57:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a814f0d82ab3631bd4488a643b29a38d56879f56","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-07 14:36:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f8ec001c81a7219b9ec5f27a084d443dc7dfe0b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-07 15:11:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nwill backport to stable/train","accounts_in_message":[],"_revision_number":1},{"id":"c927132bb215edae72a574f7f058a7311ea5d0c5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-07 15:12:05.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3cf9d92f42be13cd09b10995ba078bb5cce89357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 18:02:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6be5ee19423d437b8203e02108488068 : SUCCESS in 3m 21s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/69d302cfd3f743bea09bc5b829e24009 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e34dda2d45c43f9bf6dad3b98cd7180 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7f7fee87bc142b88552eef1bc7772b6 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b1be9d677764ed7b8f1a213ab4c0a99 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a7786bf77c944f8ae2d161d64192e26 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/995a476bfb64496aa0d382653415ba9d : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdb9e1c5a1a940b48a5dffe5af5f21f2 : SUCCESS in 3m 05s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c471d9c865944f0fb16cda4c013c6a2a : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6ccd2d526e164a60aeb09aebe4da05c1 : SUCCESS in 6m 39s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6cf85d7325754f4e9d06122b07c56fdd : SUCCESS in 8m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/876af00729114fe4822efbc8e3512af1 : SUCCESS in 7m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b0c04650e9e34f65970a92532b6dc2bf : SUCCESS in 4m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7c465cd234bd4b14b7515dce6932d1ff : SUCCESS in 4m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ee5af42d1f614202a2dfc6d554a859f0 : SUCCESS in 9m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ea5af637b332458f9ae2f976b43c35ac : SUCCESS in 8m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/677c7e47c6a5411c91210e5368188911 : SUCCESS in 7m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b8b2a758809240ea833c7af822658ded : SUCCESS in 14m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/94047f9a15814a61b1855f3128e3fe42 : SUCCESS in 16m 29s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f05ba7c587ba4d32ae54e4ee33e8e0c2 : SUCCESS in 24m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3f433a141ec540209f5f6c990ec57ebc : SUCCESS in 23m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/65d769cf426c4b62a6cdec3ba56e2f10 : SUCCESS in 12m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/27fbd311b39f4f1bb391e4f8e8186ca6 : SUCCESS in 12m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/590ceb1dc03b4d0fa9531bcf8cf56eb0 : SUCCESS in 11m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/925f5b3ff91b42499911e343f826b986 : SUCCESS in 19m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d074f9aa30e4af78ff6a49bb5dfae40 : SUCCESS in 6m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6dcc4e5123874c72b8eba69da7fa423b : SUCCESS in 12m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/60ad387403fb4e6fb79cb23eb9dfa2c5 : SUCCESS in 11m 30s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3176ec65adaa4e72a517b1d20b7dacb4 : SUCCESS in 1h 03m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f939a64a521d471eac88ae066e91af30 : SUCCESS in 1h 14m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d39ade7402994a03a7812942338147c3 : SUCCESS in 1h 41m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3e17d2914c87470191504e49d7779a67 : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":2},{"id":"ac5af6d72bf6537fc364365b28346cc8ddfbff2c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-07 19:54:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"491949995bb5f647e852c220184e21d585486644","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 19:55:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e5c079c602c0b8590def97071ab132ed3cb0e045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 21:57:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8633a125fe5b450a89dbf1e205049af9 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e1551b87ac4eb0a69eb1f8ce50a955 : SUCCESS in 2m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee552ab0849a43b0b15480fb049c8f8c : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/325b179e5458498c91f0d3e06a3d9f93 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8880490a8644ef2988b8a3d863d423b : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc7a8866f5924aad90c936ed9df36bf8 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc312ebe197546bf83c93768a2d727c1 : SUCCESS in 3m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5ccd41376bc1472d92f345286681f12b : SUCCESS in 8m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9f2fd9d3fa940d1945a2541681d98ae : SUCCESS in 9m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f1d26bc811414f58a6b9ff8ae222b07e : SUCCESS in 9m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ae2eaa0721a142958c0efde69d7dd006 : SUCCESS in 9m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0964363d86ca499181ed84c54dd83e83 : SUCCESS in 4m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4c12476e34134d2e88ffa6cdce56022c : SUCCESS in 4m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/723f72cb3b524af1985efb357cd5f907 : SUCCESS in 12m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5727d93dd730457aa61d6f09621bb957 : SUCCESS in 8m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f012b66785fc4f0d899033109cd2b7d7 : SUCCESS in 8m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c0381922ce8246f8ae4515f89a7018bd : SUCCESS in 16m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e07dcc4a36454b469a866f170757b98c : SUCCESS in 16m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f1e11db3bc2c413da9e1902dbe35ead4 : SUCCESS in 32m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/014af56838c54c0ab50fb407a113f5fa : SUCCESS in 23m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a4c5caa76dc243d79dab8d74cc506c1d : SUCCESS in 12m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/df21ab03f3bd405b8843780ebb847d91 : SUCCESS in 13m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bfde81f3b83e417bbe53c46e17b5ce43 : SUCCESS in 14m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8fcae5b90718459aaa9f82acf053d99b : SUCCESS in 18m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9759afbcad6e4aa9a52c363d4d68964d : SUCCESS in 6m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3de78d05bb9c4f1db586c66f2f2127e0 : SUCCESS in 8m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/57368be4f9a54d2499ac39f6c149a585 : SUCCESS in 11m 21s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/27606f6bcd96468f8cd5278fbcf0bbf8 : SUCCESS in 59m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/358bb4b2b2d843d4bacf2834851a5c55 : SUCCESS in 1h 11m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4d12c76536e6400ca2ee8fd4569b4314 : SUCCESS in 1h 24m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/005b99f97ddd4cfa92c196681b12f73e : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":2},{"id":"6fe55ca294dc37f860ad4d9802e83db0d6bcaf67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 21:57:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"362953d1f3fc836a250ce4ccb3163f9d2acf4d4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 21:58:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef5650633b834953bb27566a4e322ad2 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"530e8da854e7a450eb64b25841efd4204d56a2da","revisions":{"0873d3da3db3ad877151228ef55e3cce24e2193e":{"kind":"REWORK","_number":1,"created":"2019-10-07 14:36:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/687064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/687064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/687064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/687064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/687064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/687064/1"}}},"commit":{"parents":[{"commit":"aaf24f51199e298c89c73fcb1bbb4e488c52bde0","subject":"update constraint for glance_store to new release 1.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aaf24f51199e298c89c73fcb1bbb4e488c52bde0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-07 14:36:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-07 14:36:10.000000000","tz":0},"subject":"update constraint for tempest to new release 22.0.0","message":"update constraint for tempest to new release 22.0.0\n\nChange-Id: Iab5cf83ea2125ad6d950743339eea791f4a1ad7e\nmeta:version: 22.0.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Change-Id: I83aa445329f41c9c2f081483fcc283c6d4c4fb3d\nmeta:release:Code-Review+1: Douglas Viroel \u003cviroel@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0873d3da3db3ad877151228ef55e3cce24e2193e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0873d3da3db3ad877151228ef55e3cce24e2193e"}]},"branch":"refs/heads/master"},"530e8da854e7a450eb64b25841efd4204d56a2da":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-07 15:12:05.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/64/687064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/687064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/687064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/687064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/687064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/687064/2"}}},"commit":{"parents":[{"commit":"aaf24f51199e298c89c73fcb1bbb4e488c52bde0","subject":"update constraint for glance_store to new release 1.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aaf24f51199e298c89c73fcb1bbb4e488c52bde0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-07 14:36:10.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-07 15:11:40.000000000","tz":-300},"subject":"update constraint for tempest to new release 22.0.0","message":"update constraint for tempest to new release 22.0.0\n\nChange-Id: Iab5cf83ea2125ad6d950743339eea791f4a1ad7e\nmeta:version: 22.0.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Change-Id: I83aa445329f41c9c2f081483fcc283c6d4c4fb3d\nmeta:release:Code-Review+1: Douglas Viroel \u003cviroel@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 0873d3da3db3ad877151228ef55e3cce24e2193e)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/530e8da854e7a450eb64b25841efd4204d56a2da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/530e8da854e7a450eb64b25841efd4204d56a2da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
