)]}'
{"id":"openstack%2Frequirements~687080","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I354c5ba7f58c952e626463677605a9f5fdc0ddc9","project":"openstack/requirements","branch":"stable/train","topic":"train-in-train","hashtags":[],"change_id":"I354c5ba7f58c952e626463677605a9f5fdc0ddc9","subject":"update constraint for tempest to new release 22.0.0","status":"MERGED","created":"2019-10-07 15:21:46.000000000","updated":"2019-10-08 08:45:11.000000000","submitted":"2019-10-08 08:45:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687080-1570524311353-001e37fd","meta_rev_id":"fc90ede262a685e910e57285cc12196b8d0de021","_number":687080,"virtual_id_number":687080,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 08:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-07 19:54:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-08 00:56:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-07 19:54:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-10-07 15:21:46.000000000","_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-07 15:21:46.000000000","updated_by":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-08 00:56:11.000000000","updated_by":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"reviewer":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"state":"REVIEWER"},{"updated":"2019-10-08 08:45:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eabad595d7c57d1548706290ae62fa422ea97887","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-07 15:21:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79950f6499594a1072349d5fd8ebc8d45ef08569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 18:12:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5138f9980c1f4d8188118456dcb9c2b6 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/873ef1f264294131a48b7f3289979674 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be594f8e8d604448b3247f11fa227682 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0f1145e1add44edb85509cfecd0e7be : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/531517d1b8e9407e9d4984ac53af62cf : SUCCESS in 6m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76d94c1f1e6947dba4fb71850210fb48 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/475be0cb805c46568bd75c68a0db1784 : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/960765b6dca14668a438a44c7db5416a : SUCCESS in 11m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a174e2b1db7473f8d011d457bc4749c : SUCCESS in 6m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f4573a39f58f40579dd8d5a7d99f383e : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b7cb8743cd1941a7b18bd6c4309ac9eb : SUCCESS in 11m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fd1a11366ce34f2fb1435615d5a82138 : SUCCESS in 4m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/86413a0634a849648ce2cb254a68a0fa : SUCCESS in 5m 22s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0651e0d398424f8e839cf54fc09cd759 : SUCCESS in 11m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/17cff7d26a72412f9099cff23b4989aa : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/092884354bb147408443530f4fb27f7d : SUCCESS in 6m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/726197cd33cd4301947d1ee041dbd2c5 : SUCCESS in 14m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a897a84fc3ea48d09d4da81a9a2f9be7 : SUCCESS in 16m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/42824839e0234b5dabf0b9f067f8669b : SUCCESS in 25m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2c4d57006f1d4f6482979771df798b88 : SUCCESS in 25m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/60fba797289446458824a00d6157476b : SUCCESS in 17m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/95e2c9288c9a4b308f26c26c08ca0aff : SUCCESS in 17m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a0f6d68a2d764d7f904fa00e30d0bffb : SUCCESS in 21m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55cfdf82471a4cc09cfafb6adcae7492 : SUCCESS in 19m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f5a6b20c2664e279d6a2f09cf2f2f78 : SUCCESS in 4m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/70aa24a129df42e28e6a504cd691f2d0 : SUCCESS in 8m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4f40c3391bf349ff99c9a3a7a3492c4e : SUCCESS in 9m 30s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/65511e2d27f746f784ce4afa57113a8d : SUCCESS in 55m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/97f331a63bb24193b0ea66ec5c4a059f : SUCCESS in 1h 16m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d5bff80121bc4a05ad7844d41ae0cc68 : SUCCESS in 1h 50m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/703d481953f348e2bb64fe4463f69f3e : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0fdc7af7b44b2c078db844640629a6108d00237a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-07 19:54:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e51153c1f85a5889a5a78afb1e0b4d747f67cb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 19:55:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"511fb30461e197582446de9ef8121917ade91feb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 21:57:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab41ed6292474ee182490f7e3ab90cc7 : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2997a846b94c486e8113d9750ed23325 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5d35fd886ac448ab591ca1743bc9ddd : SUCCESS in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/057ab3b0d0a94413886b3d684913292a : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ba20d542e8748f6b2e2e8757041c16e : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d005b61856404260b3435aa68690c3f0 : SUCCESS in 3m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4efc2fbee6be4165b2296fb3eaaceb7e : SUCCESS in 8m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8c0e36618e834d128d7fc335fba5e8b1 : SUCCESS in 5m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c765cfa1385342efab66a6a4465b84d2 : SUCCESS in 8m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c3cca3a3bef04820a4dc5096b3680f1c : SUCCESS in 9m 45s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ffe4d834a5a147449c3d14f4fb9f95d5 : SUCCESS in 5m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4858123a0bcb44048d319b2c40dac66c : SUCCESS in 5m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3306ae830556422da9b787a9368e2359 : SUCCESS in 10m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ff39eb0b4e604d4b8e71cfb6e2c675cd : SUCCESS in 8m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23c0b00d8a5048c792e658f173f341bb : SUCCESS in 8m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cad94c25ab6f4a27a3d4360635f3cfe7 : SUCCESS in 15m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6f7d3aa0add340a48ccde1a8e41b6b83 : SUCCESS in 44m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/067f2c026c4042bbb0b04e093fe9e254 : SUCCESS in 23m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5f9c73977ccf42d2aefcccbda87a667b : SUCCESS in 20m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/aaf880dbbb7e4aa8a0b7474714b0204a : SUCCESS in 11m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/25f2fc1ebaf44451acb73e8769bf75d2 : SUCCESS in 12m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/79945cb899c84bb59ff3fcb4252465a1 : SUCCESS in 15m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e35e46edfa0143acbce20b6dbc995210 : FAILURE in 17m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c234c8295e3b4acca9f79989628d7b14 : SUCCESS in 6m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e714bb703dfb44abbac862ec7763d3df : SUCCESS in 10m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f9ce30c3be094a07bb08508253fb634e : SUCCESS in 8m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/90b16aa7ae7c4a80aa4d23dc81e78e4c : SUCCESS in 1h 07m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1c9dfbeec154669a52ad810b7cc2e17 : SUCCESS in 1h 07m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b1832166bf884aeb951b5a63d1c4e4b1 : SUCCESS in 1h 42m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/44e4a21a3e734213872e3c43b3156c6c : SUCCESS in 1h 08m 15s","accounts_in_message":[],"_revision_number":1},{"id":"472756c3a42441fe3d4cbd135b2e6fe8d4681610","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-07 21:58:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22a9620ed7976df809286ff6628b1f6e7cd15cae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 00:12:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5af24441c9154087b9cbb6bd2649b059 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ccad2794dac429d8bc4feeb03fb06aa : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1259eee0dc464247a0d29966577a7b43 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6501f0b2e68247279a6668bdf52a1333 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b70b74bd770488785d4679d71d91f8b : SUCCESS in 6m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38803728ec204efabcd36b958433dbc4 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f35cd1291f844142828f0ac49d374e1a : SUCCESS in 3m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/243ec380b4854ae2b7076b09ebcda0fc : SUCCESS in 8m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/df2afabb776b48f4afbd17b8c0e221e6 : SUCCESS in 8m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4c907c5f14a0467fb821d1ed4a63fc69 : SUCCESS in 10m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/49aa34d6a9684c4c9c61b46580899dcf : SUCCESS in 8m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/92bf802e372e4b8086a59f36c274accf : SUCCESS in 5m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b867fd7d850f40c390fb750e580ed3aa : SUCCESS in 6m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bf4d4b0dd5c3491c82dd035853f11dc1 : SUCCESS in 11m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d4997756a5e94abe90e423a893dd851c : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a51ec8854a2f4a36805e223885612c7c : SUCCESS in 6m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4991e46628cc49eaa8bb6bec916e55e8 : SUCCESS in 20m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3bca6444a3ad4a8f9a5ee3eb6ce85f26 : SUCCESS in 14m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b310352475734dd994e60f8fbf08be8f : SUCCESS in 24m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/873c55f046354050a16c4b3417da3183 : SUCCESS in 23m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4db4ffbc4bde481992a46af7c4d8e94b : SUCCESS in 14m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3cfe12f6577542738569b64255d519ca : SUCCESS in 11m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/000a8d09e864489a96d31b8461978953 : SUCCESS in 13m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/550b80cab7e44296a59847abaad0b0e3 : SUCCESS in 18m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba6adf648b0b447db2c2a2d5934063df : SUCCESS in 5m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/09528c2c60b74dad9acdde4e5bc9f294 : SUCCESS in 8m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/69c83a5d25a5476b97727de7831b1e26 : SUCCESS in 9m 08s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/44f58eddbe1c47db9ffea8658ce66ed2 : SUCCESS in 1h 02m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/21ea293af27f4099afe605e57fe38d3c : SUCCESS in 1h 07m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/893b58bc407f46e9bd1eb648a2714440 : SUCCESS in 2h 06m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/86c0c93bc0f7413bba90c069862e59e2 : SUCCESS in 1h 05m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c3c61906ff83100e47b7cc6ce104423a3d5f36c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 00:12:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dac29846c09104243c9ba99a800749d6fb95d8d8","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2019-10-08 00:56:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3fc81645f85a45e8792cfa6080057b8bb5db0b02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 02:48:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4307b4667fe4df1993e4a9754a4a9fb : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3331e41db064e55a66a267f59185bea : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0adc612e63a4a9a8cf8b3644cf769cb : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34093b8bb8bc43d4bd4957805e99bca9 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/917cefb5bd3443a8a9c5ac89912b44d0 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d6330ebe8ce34886a11bbf8be4a2f1c6 : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/59994019b37e4aeaa586fbb56acb16fe : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ffd8015803bb481ba7b225ea4ac6945f : SUCCESS in 6m 17s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e6e6e2fda3914e1ba84ef3ada07c1e1b : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0331a8a8f5554003bcdd5cd2c9bf1327 : SUCCESS in 8m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9b11c1dfe77a4e7c957f67af3af62fcb : SUCCESS in 4m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9a65bfe4041a48bfaf15ba9fbbb683d1 : SUCCESS in 4m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/60a02257662f4174a6acc41d9af48fb5 : SUCCESS in 10m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d52687a019164f6089c3cada71c73f91 : SUCCESS in 8m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2961a6c3be9a43fe844a17f3b65db09a : SUCCESS in 6m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9dae53fc00664583aa37b3e5a89d39bf : SUCCESS in 20m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a6b380bc55c348a88d2d7a587e63a779 : SUCCESS in 19m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0b573113c26f4e86b9b4200e62a84894 : SUCCESS in 24m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bdf53d7f4ecc4500b1f88ee9b7582191 : SUCCESS in 23m 54s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/324ddbe069ec4c08a5bdcc44fe3f7bf8 : SUCCESS in 12m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8bec72e49a81403f8cf62ebe63c62472 : SUCCESS in 14m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/504738e5a64849d3a85de90344a1a9a4 : SUCCESS in 13m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/48dee47e6b134f728cf2821c6b5bd980 : SUCCESS in 20m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1eb7b550779c45c7a920ea8ab062aa9b : SUCCESS in 4m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8df4f624846742b78be6277a7b8bfd73 : SUCCESS in 9m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7bbc8b38a2724b509e492573d4a36268 : SUCCESS in 9m 13s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/cad8b0b597a2415da1b968e5144763ac : FAILURE in 29m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec2b5f9bbaee4bf9bfa0ab77b96f24bf : SUCCESS in 1h 05m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/668921a8335b4ad4ba1e69cbcc0a6814 : SUCCESS in 1h 36m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1a2bee3f16674925be079e89cee62822 : SUCCESS in 55m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e47d5f95ac6dc5c0384a6124ce57a141d92f1634","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-08 02:56:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e13a9325e0348c55579e085e8a0f6b2424a2c0c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 04:57:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/41cf00149b1e40bf859298de5156444a : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f9d8cd0934046da87e3d3ec6a93fd28 : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d56060a2674643edb7678887047a1c8f : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d0390f0400a4b84b8375d516f7d367d : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf69560cdc574dc19ab7e83b4ac0465f : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fdd70083b66f435e9ca80738ecc7ae44 : SUCCESS in 7m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/52e6186bbf524690b1dd1b78b2b984da : SUCCESS in 2m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7e82cc917339462397d612605511a17e : SUCCESS in 5m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/821e6f938e0d48c0b8f49ebfb477f6ba : SUCCESS in 5m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4b46e06bbad843da80c269060e76bc99 : SUCCESS in 16m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c0fdf94c8b414f73ac7103b4a06a13c0 : TIMED_OUT in 40m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6334245564a4400b9e6f63e55f6610f0 : SUCCESS in 4m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3975e40d2ac549b297f0f3925662ee77 : SUCCESS in 4m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8e394cbf2d3941549e9a638cdd6e5d02 : SUCCESS in 11m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6f03fcad53104e5280ec5ca905f6ad48 : SUCCESS in 7m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/10c1670276874a6ba97052096217d404 : SUCCESS in 6m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ee915bacbf3341c78722614e042efe3c : SUCCESS in 22m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/def790725dec42e99cb914cbe2cd2fd9 : SUCCESS in 26m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7c035f0612d94478a8f64d5cb551e812 : SUCCESS in 20m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/666e94a6f2ff4609bf2492f2236e7864 : SUCCESS in 22m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/35549376b8f741f4bdba77ae13f0593b : SUCCESS in 13m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/287cefd735344847953fa4056a19f572 : SUCCESS in 12m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/27c4628bf00a482c9e80224c1b5f4d61 : SUCCESS in 17m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/58183cf15c2b44b99df80c0ae3aae740 : SUCCESS in 22m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f8f9a8ff8744877b03b98b0165db484 : SUCCESS in 4m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dd176ccbc1074f93b2e5a544c5594a49 : SUCCESS in 19m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bcb0765f54804b42857720bb8f355121 : SUCCESS in 8m 34s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/2f48791e6a1e48768c8e3336ad10c5e6 : SUCCESS in 1h 13m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9802c2d344524378bfa3faa2ca806e30 : SUCCESS in 1h 24m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2956fa30bf5c47db904679c309af36ae : SUCCESS in 1h 45m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7605893afa324441997b86edce604b2d : SUCCESS in 1h 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"32114bde406f20d7511249166290386b082f9e45","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-08 05:10:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5c038982e2b4501ed35c249b51033ac17d8bbe20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 06:56:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6815798f33cb4c6ba2e7bcde2134572d : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/258ee45a6e084a31a3edbfeaf8ac150f : SUCCESS in 2m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e4dc094603e44069a6914f5a0cdc2b9c : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d307d8dce804b1c8e70e71748fa5651 : SUCCESS in 2m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/deaf0bf960924ae5bf59d50caf155c93 : SUCCESS in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2a983d8993a496882746877303908fa : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bcf9675499374b649a5bee8c0f17f130 : SUCCESS in 2m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0d4a95eb300141da86915b297cf6995e : SUCCESS in 5m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/060fbadcc7a24387a8c746affcd312b8 : SUCCESS in 5m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/94b7acee9ec24e11a2d0729e26623a27 : SUCCESS in 8m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9ffedfdad1ec4e83a42b3204f51acbe8 : SUCCESS in 8m 45s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/24de913d92cf443dbabcd5ec4926d5ef : SUCCESS in 3m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4bc0a07f44c44afb9bed0c5a75977865 : SUCCESS in 3m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/00c7b08591a44c79acb661fdbdb52481 : SUCCESS in 9m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fab25aa1768a4f1dbda318bf0f4c6213 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a47c196dd8d1416f81cc7dd0118cabf4 : SUCCESS in 5m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7c363e00336b48b6ba7a6724305ffbaf : SUCCESS in 13m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/62c8c6bf045a465db769f06c7d930386 : SUCCESS in 11m 55s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b356f27bca874592aa671660014fdf28 : SUCCESS in 20m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/84b630a03f884226b59943fc005ebbb2 : SUCCESS in 19m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/fea5f894d77743ccac003f7338276dc3 : SUCCESS in 11m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/813e5be627044de8b27860cd2db762b0 : SUCCESS in 10m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/88a8edfb4ef64b25a72d62d36641fc68 : SUCCESS in 11m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1bb133471de847e18f3249c245d51ebd : SUCCESS in 17m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a494dc4cf29d4f91bd189d785fbb60a0 : SUCCESS in 4m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fda959a8446541eaa54a60da19a20873 : SUCCESS in 7m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4a96ebb9690349a2aaacb388b6d3f956 : SUCCESS in 7m 51s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/dc1ba6681425433081eb90405541e4de : SUCCESS in 56m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/59e68c679b634229a30feb13ad85929d : SUCCESS in 1h 05m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8073a91a3dab41bf9d341f0cd5255aed : SUCCESS in 1h 36m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/50bf618281764d6ab25365299cac63b6 : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1b68ec929dd8a479b9fd06af819b11574e002cd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 06:56:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffb695b213a4bfdb40a04ba58c5df75d9889c7e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 08:45:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6e7852aac7348d195b34d2534fae5f6 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e21b4301cd642fd9399eaaf4d0a55e5 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/192a81b1b65c44c5baf6dd5d0125da71 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19bc5523f2c44773aeefe33033241211 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e810c19759d94e0ca3d1dbc98363326f : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d2647a2c05d147cd8a4b696dad085685 : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e942bd2e730d467697f63efaa538fa32 : SUCCESS in 9m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c432910ceed042deab6e9d9d328f0a74 : SUCCESS in 6m 53s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/027e7b2748a54e2fbe29281c9c59613c : SUCCESS in 9m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a6f3f5330011478f9674fd66bb86aec6 : SUCCESS in 9m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3a45b586a39d426ea934f2bd0df421ac : SUCCESS in 5m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3f447da2a21741cead93578314137b22 : SUCCESS in 4m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/70a180a6e8af429db073d9a88c5973af : SUCCESS in 12m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e5f2a3b8024c49ffa22ac5160cecba61 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b7d7bc11d3c4c75ad6d6c5f2fa60fed : SUCCESS in 6m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d80c8ef84af3401ab7769c265b8864bf : SUCCESS in 13m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/42a5710383d7487eaa1113e45f3320b3 : SUCCESS in 13m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/932158c4b45b46cfae2c394af8b310b7 : SUCCESS in 24m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/87eededb070b4e509b42f847be4fd3d9 : SUCCESS in 24m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/58a63aa782fc475f92b1b5dedd485e8e : SUCCESS in 14m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eee5d08a6664424389f1c660acfe22de : SUCCESS in 12m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a16cdde59e4a45c58d570d88d3dec49b : SUCCESS in 12m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d30a71f681424296a0e6eb666a83a990 : SUCCESS in 24m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e76732841674285b6e3b7ca2ec0706e : SUCCESS in 6m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2334fb8841ca42ad84acedd2180eb426 : SUCCESS in 9m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/085a1d74d3ae48c48ba122e741799b19 : SUCCESS in 9m 39s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7d59febf469b41f1872c94c33f87862f : SUCCESS in 1h 01m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/af7796b10adc4f1684f6258dc8b7d272 : SUCCESS in 1h 04m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ef1e5c92dd4c4735bb2d0d5df95c1b9a : SUCCESS in 1h 42m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/44c1fd93480e4b12b72a60813bd2ebe3 : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d248fd009f7fbff78e729d30d68b95617d33ad21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 08:45:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b333795bc06d66c696da31409434ce369ceca3f5","revisions":{"b333795bc06d66c696da31409434ce369ceca3f5":{"kind":"REWORK","_number":1,"created":"2019-10-07 15:21:46.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/80/687080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/687080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/687080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/687080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/687080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/687080/1"}}},"commit":{"parents":[{"commit":"aaf24f51199e298c89c73fcb1bbb4e488c52bde0","subject":"update constraint for glance_store to new release 1.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aaf24f51199e298c89c73fcb1bbb4e488c52bde0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-07 14:36:10.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-07 15:21:08.000000000","tz":-300},"subject":"update constraint for tempest to new release 22.0.0","message":"update constraint for tempest to new release 22.0.0\n\nmeta:version: 22.0.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Change-Id: I83aa445329f41c9c2f081483fcc283c6d4c4fb3d\nmeta:release:Code-Review+1: Douglas Viroel \u003cviroel@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n\n(cherry picked from commit 0873d3da3db3ad877151228ef55e3cce24e2193e)\n\nChange-Id: I354c5ba7f58c952e626463677605a9f5fdc0ddc9\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b333795bc06d66c696da31409434ce369ceca3f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b333795bc06d66c696da31409434ce369ceca3f5"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
