)]}'
{"id":"openstack%2Frequirements~687214","triplet_id":"openstack%2Frequirements~master~I664ba64c9347ef681e1700236077c4944b129329","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I664ba64c9347ef681e1700236077c4944b129329","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2019-10-08 06:15:03.000000000","updated":"2019-10-09 06:21:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"005426b70132c2c477361465ea8f0d67da962a52","_number":687214,"virtual_id_number":687214,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 08:19:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0721c22111f4ac8441faabf74aa3f3ad132504b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-08 06:15:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eedfc15dc9c2ba1cdef774be7b1acba207571267","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 08:19:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a1f853375984a7d8a527caf7797be52 : SUCCESS in 4m 21s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c9eb0e26f867494fa4583b867ea3b0d0 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60e277804f9c419c928e3676444d4908 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8eb2adee30bb4e0d94599bc6202e0046 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e235fbccd4864f849e1246df9a35f8b2 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81b58ca2e885407a950eab368320f833 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c92bd7aa69ee47cab1c85c7770f43cbb : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bc04b7fd762d452aab760361b72cb3f0 : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8e7c669876a6482e8c00d25826bd1044 : FAILURE in 11m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5ed9281dbb9f4a93970247965f5d6317 : FAILURE in 10m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bb3a3112aae84e808416bbf5726899e6 : SUCCESS in 11m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/44583530abfc4e6fa353c5e6c871b062 : SUCCESS in 12m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/588a5832024c45cdabc8dbe399fad62f : SUCCESS in 4m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ff6d5d0ef30848558adedec0c67ef1d7 : SUCCESS in 6m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/eff8c67d95924989a144966622528126 : SUCCESS in 11m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/635b0f403d314add8d62f6ee3ebe6e97 : SUCCESS in 9m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d3a3de71fa24ec7be1b29b0306ad4ff : SUCCESS in 8m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/03c45c086b914da3940b95a3bd60d291 : SUCCESS in 15m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b625fe7f3141419ea4ee9b88d959983c : SUCCESS in 13m 40s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5fe4fedfa0f247fa945824535baeb809 : SUCCESS in 24m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/35a792fa725c4dd4a9a321c3911206bc : SUCCESS in 22m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/13f4f1c1c1dc4ab1bf39f0e7b3ebe828 : SUCCESS in 17m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9fe99ea5686141438cfb075ebd3ac2ba : SUCCESS in 11m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f35a56d4e35443e8af64c55904fed558 : SUCCESS in 16m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd4bf6148b1d40529858633e77ec1e40 : SUCCESS in 16m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6c40a349e28849cc9af2a3d72c328de7 : SUCCESS in 5m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/84fd4417bd424206a2ce59415122ae64 : SUCCESS in 8m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fdcc634d7ce24ee1941fc9d421bd343d : SUCCESS in 10m 25s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f9561f4d3d3446c5ba244991a25a1e6f : SUCCESS in 1h 52m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f9a159922e74ac48cb1abdce31531c5 : SUCCESS in 1h 14m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/220321bc44604a1699106cb99a69c526 : SUCCESS in 1h 33m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b64770e0d6214cb7a9bdbdf8ddace018 : SUCCESS in 1h 11m 56s","accounts_in_message":[],"_revision_number":1},{"id":"67fcf04b835ededcab40b513dfdd1fa2969c56c7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-09 06:17:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4fb3fc78fdeaaaa99951c81e4a81d7f1abcf4769","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-09 06:21:21.000000000","message":"Abandoned\n\nnah","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"00a3c732df58724f2d5e2a9b7a3805e09757c9c1","revisions":{"a62e404828a24aa4148ea6b23ec09eb6c9117d5f":{"kind":"REWORK","_number":1,"created":"2019-10-08 06:15:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/687214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/687214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/687214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/687214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/687214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/687214/1"}}},"commit":{"parents":[{"commit":"530e8da854e7a450eb64b25841efd4204d56a2da","subject":"update constraint for tempest to new release 22.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/530e8da854e7a450eb64b25841efd4204d56a2da"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-08 06:14:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-08 06:14:57.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I664ba64c9347ef681e1700236077c4944b129329\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a62e404828a24aa4148ea6b23ec09eb6c9117d5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a62e404828a24aa4148ea6b23ec09eb6c9117d5f"}]},"branch":"refs/heads/master"},"00a3c732df58724f2d5e2a9b7a3805e09757c9c1":{"kind":"REWORK","_number":2,"created":"2019-10-09 06:17:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/687214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/687214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/687214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/687214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/687214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/687214/2"}}},"commit":{"parents":[{"commit":"88eb85e374e222b116134d868bfaed4bcac55986","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88eb85e374e222b116134d868bfaed4bcac55986"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-09 06:17:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-09 06:17:36.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I664ba64c9347ef681e1700236077c4944b129329\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00a3c732df58724f2d5e2a9b7a3805e09757c9c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00a3c732df58724f2d5e2a9b7a3805e09757c9c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
