)]}'
{"id":"openstack%2Frequirements~687327","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I47f18334c5f15028c4f2bb7a8ba9dd46f2be8f4e","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I47f18334c5f15028c4f2bb7a8ba9dd46f2be8f4e","subject":"update constraint for oslo.utils to new release 3.41.2","status":"MERGED","created":"2019-10-08 15:06:12.000000000","updated":"2019-10-09 08:23:03.000000000","submitted":"2019-10-09 08:23:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687327-1570609383156-9cc037c6","meta_rev_id":"6fe3e5d8ab4b40fa317311abdb4d65d39747718c","_number":687327,"virtual_id_number":687327,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 08:23:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-08 15:18:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-09 04:51:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-09 04:51:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 15:18:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-09 04:51:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-09 08:23:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a67180c8c4a00deb6733e2c038193976224944f7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-08 15:06:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a940e870aa0df4356afe71497fce0cd230848414","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-08 15:18:14.000000000","message":"Patch Set 1: Code-Review+2\n\nFFE had been granted: http://lists.openstack.org/pipermail/openstack-discuss/2019-October/009985.html","accounts_in_message":[],"_revision_number":1},{"id":"ba136f8ef671ef43ade0c0ef51ed3d664e691645","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 16:39:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/22e0a554791e441994b6680ebcb078b8 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07fe902d2cc040b49d3cc5c1f9044a0a : SUCCESS in 2m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb7babe30ccb4ed0b48c79391271a74e : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/416b863359a74961b6093b44d847c10b : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97b7783e2dc144cbac28b20e51d356aa : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9f19df8af683450fa7007b08e7519ae5 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6387651061440d2987f3aa8e205f955 : SUCCESS in 2m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/792afd9ed3ed4730b51f0d9ce8b8069a : SUCCESS in 6m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/75705541587a4a98b63d732ae6d3bb14 : SUCCESS in 6m 47s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/83cde41f2da74c2597598a95ecf52086 : SUCCESS in 10m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b033525311c04527b21ac8f1e86e97a8 : SUCCESS in 10m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2df74e60d7f44e6fb0c84db8d790e8ef : SUCCESS in 4m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a9729c16c83c4ad29e016c93b8b52afe : SUCCESS in 6m 37s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b10e66845ee14358885a6c147de447f4 : SUCCESS in 10m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1f1d027f4e0446a6ad8f2d7b9429a402 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b94f7aa09d654e88a859b19887906d6a : SUCCESS in 10m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7cf320bcc29a4a8abb9d67f9b8b943e8 : SUCCESS in 23m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0fbb37ed423243d8a4e07c0dd20efe0e : SUCCESS in 22m 02s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3314f374412d4d7d89cc8c26f42b9448 : SUCCESS in 31m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d09332d30b344c2697abfcdb796e7c03 : SUCCESS in 26m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3e83fa6a784741229328b905912c0adb : SUCCESS in 13m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2e070f63a6c4478383026e4c54b995dd : SUCCESS in 12m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/954f5f7a4fd540918d58f826f22706d3 : SUCCESS in 17m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5013121327ed479aa3a93ae188711def : SUCCESS in 21m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7c2e411a0fad4260b3e777ccee6d4388 : SUCCESS in 5m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a04d6ed342bd4911b71fdbfbda36664b : SUCCESS in 10m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c37dd52a76044b61af6db59b5b42b564 : SUCCESS in 9m 45s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7f8f082776b34689a8157d6247fe7fd7 : SUCCESS in 1h 11m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/91ada102770f4841a0998de96f94a46a : FAILURE in 1h 26m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/624b98fc7d74478fab6a58a179757e2a : FAILURE in 1h 14m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b4b68d84dd3f45619e84fe149f7ff7a4 : SUCCESS in 1h 04m 34s","accounts_in_message":[],"_revision_number":1},{"id":"374b3a0e661fc59b2ac39fb7e84071c85ffaf724","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-08 16:44:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"017084eded54ff57244b3edb4d6498cecf9b86ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 20:50:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c8176c5a242c4652bdbd74d4de88281a : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c189bbaa394aed87dfd37e4cd26880 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2fbb2c110bb4fe18955ba07d77b9d69 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95db9519f12b434eab4377b8d517448c : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9ef9eadd1604ec58a7d888f17de6b86 : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47d4c4c502dc44e3b700e6f373da2e77 : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c09fd8940ede4a23a50a829e3d658f94 : SUCCESS in 5m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f261c264065f47a584293b8e3aa4e873 : SUCCESS in 10m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3efc6d4172f54638bf4208fc25140f12 : SUCCESS in 10m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3b31f6520c4e49d1b07ca60ec6bb8eea : SUCCESS in 10m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/af050d9218c24bffa26a8ff61c1e54d9 : SUCCESS in 9m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b5a867ac88024129856eb374ea48d036 : SUCCESS in 5m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a5db974e317d421885fa962697f843b7 : SUCCESS in 4m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/453dec009b694252859999c120c33725 : SUCCESS in 10m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6d2e55a3f77b4a64b947489db10a047c : SUCCESS in 9m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/03c19c6f0c404bad96d3be1646a7590f : SUCCESS in 9m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/13ae3c6460fe4864b1c85523dea5335e : SUCCESS in 18m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8f115e118f8a42e6ac0efda94473738d : SUCCESS in 14m 57s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a8d0edf517694e51b79f6391c4a84feb : SUCCESS in 26m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3c6a861e67c94a8ebd95059e235f703f : SUCCESS in 23m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/28e5fee2adc346c9adfafcfc85bcbc17 : SUCCESS in 18m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/31915c687b16444d886d966cbb96cd23 : SUCCESS in 16m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ba2a7516577d48c8bcb6a89274a011df : SUCCESS in 14m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d6233549f3f3471a941f985b470c22be : SUCCESS in 26m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2de42bda97d3427e84eef5347cdf1131 : SUCCESS in 6m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/68fd8bdc5ec542cd9ce162ad7f7669ab : SUCCESS in 9m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b5dca56227fa40d1ac3fa8d4c260e0ad : SUCCESS in 11m 08s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e425d434e1c3421da989e58eb5eb1c77 : FAILURE in 34m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a877137a336402484f1f747c923745d : SUCCESS in 1h 06m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e0c24b0cc82c40d094c43e9541c3a517 : SUCCESS in 2h 03m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/454f5ecaa88f483fb6145b88f611bbb6 : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":1},{"id":"fdc819d0ae14d3ef7f0614e1dbf13320388d9071","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-09 04:50:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ff53dce1330304fe7d2edbdef81d9fad6777a055","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-09 04:51:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"366bf93af08d335a99749765fb9e49887e9aa975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 06:26:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f39292542ff74e82b9a5dfc729d0e94b : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/372deb7c977445aab37e7f97d1878ec4 : SUCCESS in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1076342c74e1472aaafc3e4ab993d3ac : SUCCESS in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0131d451995649869c6cb882456b1bb9 : SUCCESS in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e690d84ea3141f19aa4a77f4c6c889b : SUCCESS in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/22dda5db7db548e4b92d1d8e26f05fc8 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d3504b8915614e95aaa85265397c00f9 : SUCCESS in 2m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d0a96c751ccb4307b9ff27acf94d1590 : SUCCESS in 6m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8ba814ef3c5b45c18c6c252255a2bac6 : SUCCESS in 5m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3b589c2aa88a42798a326a18405e1ea6 : SUCCESS in 8m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7928b7c5065643deaa44890c0458f349 : SUCCESS in 8m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/bc483db7be8e4af997dc4862edb23885 : SUCCESS in 3m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9a0eb58502344e8bb275ddf28ea3b69a : SUCCESS in 5m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/956188d887ef4645bd641b140833a636 : SUCCESS in 8m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4882317965404d3681375344747d4bdb : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3bb61bc84444d8c9a0c82690a979374 : SUCCESS in 7m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e96ab970546a46f2bfb75707572754b2 : SUCCESS in 16m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5a69363058504cac8ac930501d1cc2d2 : SUCCESS in 13m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2fb0f0e431364180a54299d7eacc82f0 : SUCCESS in 21m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a062db5348fa4c5886f845bd60f2324e : SUCCESS in 19m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c856f457b29d4a979d4354e60cea8fb0 : SUCCESS in 12m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/19f6867004fe44949cd482654769eb51 : SUCCESS in 12m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b6d8056182e24420a31d254e07e8eff4 : SUCCESS in 13m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/58686ad3c8ae404c9afb5bbad2a33c80 : SUCCESS in 17m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/642ed13666804e2f9f7337f3ccce7ea1 : SUCCESS in 4m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/75b924fe97024fb9866fbeb84272acff : SUCCESS in 10m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/031120768c23471cae58db32a11e721f : SUCCESS in 10m 30s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/07420f97ad374b46afac4b2a9b51ce14 : SUCCESS in 46m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/039a04c6fd714fb9949b99efdac09ada : SUCCESS in 1h 04m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/037afd5c704a455da8dbbef7c0b4cf52 : SUCCESS in 1h 29m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/774c429568c748b4b2b26ae59a2eb83d : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b1d9d6c0e9911e38daad4d4a5e3a15cd93e1c63f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 06:27:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7660df026facb667f9cc8e314b177cc7692491f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 08:23:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f045e7bbdd2a49ac9b435aca2aa37816 : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9099529a3674064b2a28271a19bddcc : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1508e014b0da444cb8b3ff79d1a901cd : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29598450ecac4c42923e3eeefbb2aab0 : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/42937395787a488290548a4a9e47a825 : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ac50095d5604d2ca53e7ef19a4aedc8 : SUCCESS in 4m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b884d17a13fe449da165135287aa6490 : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/81b7032741e34c88b127c9ae124aec85 : SUCCESS in 7m 22s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/dc8883b852b544f0a6b41164e999e5b8 : SUCCESS in 11m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/71ca058b6fab4179b211aee0fd2a0b7d : SUCCESS in 10m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2688c54707c24f81968454ace49b5bc8 : SUCCESS in 6m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e2a60e1b2f4047509ba5031438df5b4b : SUCCESS in 6m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6a58c28de7d3405e895bfe31dbdf3b70 : SUCCESS in 10m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ee9c87b66ef240038b85764f68f3918c : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4778cf498e114ddb8ecf33967022a1a7 : SUCCESS in 5m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2f67db527e3e4a7ebd22ae110be1dabc : SUCCESS in 14m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9142e671ec554d19a2eba9ec6600f524 : SUCCESS in 14m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/83e643e15a4a4dd994d8fb9376c8c335 : SUCCESS in 23m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fae55143743247d5afd73780a9a996d3 : SUCCESS in 22m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dc6ac041211546dba7d1349aefdadc4e : SUCCESS in 14m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ac26c16ffb5946a781d61ccd8c7b6d14 : SUCCESS in 12m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/887c37c4757c48efa0ddff36f5c406e7 : SUCCESS in 16m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3251b5517150419db2b50aff45790ca2 : SUCCESS in 26m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/053c537d025e4ec0a1d9926371af479e : SUCCESS in 6m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9be931794c674d639d0e55df88995c10 : SUCCESS in 9m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/61be660da0134d419b789c279d29ad23 : SUCCESS in 8m 50s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/afd7a0bdf2524e46b7c5e38de814d66b : SUCCESS in 1h 01m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/82bcc1ed36f84d7d82e68d47c9e845a5 : SUCCESS in 1h 15m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2981cbfcbe8540e6bea339a34302e2a8 : SUCCESS in 1h 34m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d7160c5ed9a1488eaa2bdbc3605f5a91 : SUCCESS in 1h 48m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a05133259a9eea5883719c058943d10ef4251045","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 08:23:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ae3cca1a1fde0e10c0770cfaf78550b1b3129a7","revisions":{"7ae3cca1a1fde0e10c0770cfaf78550b1b3129a7":{"kind":"REWORK","_number":1,"created":"2019-10-08 15:06:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/687327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/687327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/687327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/687327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/687327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/687327/1"}}},"commit":{"parents":[{"commit":"fcc85751820712408a2b234b5468d3e1ec302e8d","subject":"Merge \"update constraint for tempest to new release 22.0.0\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcc85751820712408a2b234b5468d3e1ec302e8d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-08 15:06:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-08 15:06:11.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 3.41.2","message":"update constraint for oslo.utils to new release 3.41.2\n\nChange-Id: I47f18334c5f15028c4f2bb7a8ba9dd46f2be8f4e\nmeta:version: 3.41.2\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: If9d119bfe8635b1bde4844958aa3f2f12fd5f7d2\nmeta:release:Code-Review+1: Natal Ngétal \u003chobbestigrou@erakis.eu\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ae3cca1a1fde0e10c0770cfaf78550b1b3129a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ae3cca1a1fde0e10c0770cfaf78550b1b3129a7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
