)]}'
{"id":"openstack%2Frequirements~688041","triplet_id":"openstack%2Frequirements~master~I6932370286a03669c603f9dd79a972e4a2ecf98a","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I6932370286a03669c603f9dd79a972e4a2ecf98a","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2019-10-11 06:37:43.000000000","updated":"2019-10-11 16:38:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"91af86bbf8727725fdb0d3f1f7048abf4936200c","_number":688041,"virtual_id_number":688041,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-11 08:29:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-11 08:29:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd835ab33374ff18e3c2b6b4217440a952ac2d86","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-11 06:37:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"791aa96c922c74c61f3d1e1601dc724e8b4271d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 08:29:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65357fc6e99f45089eaf9a0762ca5e3d : SUCCESS in 3m 02s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cb33293f261d47a1a51fc5526dcb2710 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb08616672b4176825f5b48ac3d4a30 : SUCCESS in 20m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf639f471d3f47039d2d041d9f6325d3 : SUCCESS in 23m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2927e0a63ca42c8a84fe64887757d15 : SUCCESS in 17m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1ab6c8e834d4f77b1b9057536052325 : SUCCESS in 25m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9766dcc33d1949a0abfd0b30c2c81619 : SUCCESS in 16m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b12432e8d8a4ca394da829897c78bab : SUCCESS in 2m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2220892a99964b2d959c612675fc2308 : FAILURE in 13m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/132cb0a6e21f4b05b080e2f58385fcb3 : FAILURE in 6m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9df3a5065dbc4d4389b41d2d2f13ce86 : SUCCESS in 11m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c3fc4c64bcd3474db04217814782bbf3 : SUCCESS in 9m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/10b1b78230444f40b88335dfb0e33174 : SUCCESS in 7m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/92636242098c497185fd474f42175976 : SUCCESS in 28m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/be07811b42b745b0b6854752ea32b9da : SUCCESS in 10m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d2c44a2d4c5d45839d3004fbe3c08dfe : SUCCESS in 31m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d5c2ba65f52949c7977e4704a7a6cf22 : SUCCESS in 35m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6933c514bb434006b2757361337bd46b : FAILURE in 14m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/60f2a3bdfa45405d8a5ff99b67e5c549 : SUCCESS in 25m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/434f4dac32b94603873cde78603f2f80 : SUCCESS in 21m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/815796fa973446b5a0ab1a7c89d0cede : SUCCESS in 20m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f2fdead5fab947f3b53654449b7beaea : SUCCESS in 14m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3bfb56a559b64ec18019a70fb07d060a : SUCCESS in 17m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d21ea6c7e74b4c19b9f77c25ab30390f : SUCCESS in 19m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf1d5366b321431281df8d125cbd61e6 : SUCCESS in 20m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8870966944a4100b13ffd33a1fb684e : SUCCESS in 6m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b0f5b3769a654fbdbdcd8fe5f7719d26 : SUCCESS in 11m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/86ce91c5d654449da6d3f5cb0cbb1f81 : SUCCESS in 8m 34s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1c2fcf61fa4b4d008282e57e1ed56fcb : SUCCESS in 52m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f26229ae3e2e42d590a0b05bd2a9d0d6 : SUCCESS in 1h 15m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ac7dcb0f0fff451784bc28d052c00d8a : SUCCESS in 1h 40m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9b21f81a2a93432c8152a50448d5ea71 : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":1},{"id":"dc4d16c49056865022e099f2e80fc9c38a8808a8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-11 16:38:01.000000000","message":"Abandoned\n\nnah","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa9a8a240436a2033476419f4fdba321e21560e2","revisions":{"fa9a8a240436a2033476419f4fdba321e21560e2":{"kind":"REWORK","_number":1,"created":"2019-10-11 06:37:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/688041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/688041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/688041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/688041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/688041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/688041/1"}}},"commit":{"parents":[{"commit":"722f91a10762f0fc291635c21beb886f1716c01c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/722f91a10762f0fc291635c21beb886f1716c01c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-11 06:37:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-11 06:37:38.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6932370286a03669c603f9dd79a972e4a2ecf98a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa9a8a240436a2033476419f4fdba321e21560e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa9a8a240436a2033476419f4fdba321e21560e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
