)]}'
{"id":"openstack%2Frequirements~688232","triplet_id":"openstack%2Frequirements~master~Ic73e5a2b195f1189ae0db54e0bc97fdde82abc98","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ic73e5a2b195f1189ae0db54e0bc97fdde82abc98","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-10-12 06:18:29.000000000","updated":"2019-10-13 02:34:47.000000000","submitted":"2019-10-13 02:33:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688232-1570934019133-f9e06b29","meta_rev_id":"e59c8b1dafa5406beacb500ada6c1b6b7c98db39","_number":688232,"virtual_id_number":688232,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-13 02:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-12 23:32:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-12 23:32:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-12 23:32:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-13 02:33:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0e25da10050c1ae551f7f9ebf096ce02725af20","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-12 06:18:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44c43420fff97356ca8aed1b3624524441d684d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 08:58:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9728faa43d7a40ae93a059f38f5eb661 : SUCCESS in 12m 50s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9160c15c49584dc6a2782cc77198fbc4 : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a5ad666421745018ffcb159a6c0f947 : SUCCESS in 12m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4de9c21ae40475c80432a1321970de3 : SUCCESS in 11m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2a0fa56374f4c8c9ff12c4aed562824 : SUCCESS in 13m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/08292262d759445aaebd514baf273343 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/624532dbf3984e099b1dcf1123aa4743 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cd8a021d782a4fe5adbccb669c4ee6ab : SUCCESS in 12m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/81ab3bdc4d7445939b20dbee31ca3fc1 : TIMED_OUT in 31m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4c9568a3657a40c39eb5b7c2edf66ea7 : FAILURE in 7m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/051b8c192b42426b9ee18a317f6deca3 : SUCCESS in 9m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bb8a8f47731e4a16bec5dce952c1428f : SUCCESS in 23m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/823e5eebd78b4faab521220ae71423b2 : SUCCESS in 15m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/46fd2a8dafb9492d8b09d5ebdb97984d : SUCCESS in 16m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b026e97ea4964f089021aa821c463360 : SUCCESS in 10m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8ef711b1ed6b4deb930867f61c7fcf16 : SUCCESS in 8m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94b335f6cd114a4690654a910830a45e : SUCCESS in 5m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/64ee3678810f4df787a548dcd382159d : SUCCESS in 14m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/06ffad06cb8e4c1985fe1f1bb572e586 : SUCCESS in 11m 14s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/38e06490f6714e9fb346f66d9b343697 : SUCCESS in 23m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/479960ebb78b44d1b4607b82184ef4b0 : SUCCESS in 31m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bf7ff04be56248dc887937662d9a7f0c : SUCCESS in 12m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eff179f468ee48a089e8f5c95190fec0 : SUCCESS in 12m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/eba45e37ac814ccf9d38c63df815df5f : SUCCESS in 19m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6af4b255052b4d06b750784bbcdd6314 : SUCCESS in 17m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddba1b4a10074712b3358c5f35d1a9c4 : SUCCESS in 4m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/60f0c05527814771883032a2dfd0d241 : SUCCESS in 10m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f3b57857a8214b2abc0e70f7a3042ae4 : SUCCESS in 9m 41s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/cae8e31b5c8b458d8bde8d1a2817728c : SUCCESS in 53m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed5b797a15eb4f0c9fdb5aa9f09a56b1 : SUCCESS in 1h 34m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6d03ae63926849eeae9462ccf6249cee : FAILURE in 1h 57m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/19f2954c142a491dbbdcf082f12bad44 : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e8ad8637d3a62de5fc4ed20c5955e6fde2f4a447","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-12 16:52:15.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":1},{"id":"1da3eb2cc5a81a11f887d9efc703f6e188406987","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-12 16:53:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9c84b62182c620ca319dad3a02b23c382213158","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 18:36:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86797cee318b4ebdacadd447ac6cdf44 : SUCCESS in 5m 16s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0554df0f79ac4a0894df401c0b462d98 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8058ffaf8fbb490182cf7b95391c040c : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c6a92d047fb44ae83c84b459b2fc38b : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1781d88afa7f467fb8a0004e70735d28 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28a57116e706445bb3b098000012b339 : SUCCESS in 6m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/755bc6c7618245568ecce8b8eef9e07d : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1fb3b09bcf3640b5a6fd1456e012464d : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3afe0da52f5542eebcfd9db2003d3f1e : SUCCESS in 8m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d533f83620dc405eba67408a25c5ac94 : SUCCESS in 6m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/45d424acfc62420da7032a2f7af27564 : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8837ac276574456885bd9b1228faebd4 : SUCCESS in 8m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c90a299fb024480fb72fc18221aa8040 : SUCCESS in 3m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3fb5197f19d14e3d9d25ad4d2c3acafe : SUCCESS in 3m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/dbe45eb20b9d4d5ea3c3ca69e114ea29 : SUCCESS in 8m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e21c0b124a95456fb76b1c8dbeccf821 : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a54786ad3f4148f894b49f40474952d0 : SUCCESS in 6m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/99e08db2924645a3bbf1fdc7723e3d72 : SUCCESS in 13m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ac09309205ab41c5b2d2139e1d1ec386 : SUCCESS in 14m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/429ea0b8434641168eb8a648a6b638df : SUCCESS in 32m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/07c5fa5dee82485a8cf2635f1197c76c : SUCCESS in 25m 13s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2fb1e138e4d549078d8922ed39b2a3b3 : SUCCESS in 11m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/20c10f81bfd04278861321529bb984bb : SUCCESS in 13m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/414157cb39d14327aa6282f17f25c326 : SUCCESS in 22m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7e67363179a493493bcd5558e7d81d8 : FAILURE in 16m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/725a65a30a5c4e4c9036eeb9635b8d8f : SUCCESS in 4m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/31aa648afca644eeb0935c28d53893f4 : SUCCESS in 8m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3e1a58fbe24b42128ffd2af3d119dbb8 : SUCCESS in 9m 15s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4eaa4d34944644b69efa5f6e5bc2d76d : SUCCESS in 45m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1497f35ce96f4fd8a604c20ec270927a : SUCCESS in 1h 04m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/636da2cc1a4e4c61aaae9fc28f66c4ef : SUCCESS in 1h 32m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/465c17188cd24cd782ff9e7f173d1157 : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2659ceac5b99483086a42cdc8ba3260eee633f8e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-12 23:32:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"75a1ee6138e39d5005ba8d1c78fe490cde510049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 01:03:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b57604c841954c36850fd3069ab8652c : SUCCESS in 2m 57s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7867a29084054d16b854244ab52e2e00 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc14a6de5f454054a4c05e4ba3e165e5 : SUCCESS in 2m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3d59db1c78d4d9591d6870a5a24d228 : SUCCESS in 2m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0612c8c5217c403196c37b839f18385b : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b84835a05b4c4f03a2b4d9accdcfa919 : SUCCESS in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de3af461205b4a6a8dec6b8b5225bde9 : SUCCESS in 2m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b625865e18c4c98b146718133fd8331 : SUCCESS in 2m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/652af7ab07e143dca2771a38865c9cab : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fbbf1138af3b4187bf50f4ce1afceb14 : SUCCESS in 5m 45s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ead73434187c443594e97a7e8cde335f : SUCCESS in 7m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2960061668f943fa8247af828131d805 : SUCCESS in 8m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f90ce87819c4482abeda4e1c99c83069 : SUCCESS in 3m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8819f5d094ee42d8aff02c8cdefb1c6a : SUCCESS in 4m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2ce251d5dc6944cfb78b2e3115b6acde : SUCCESS in 9m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a34aa488e59b4a62a5a93c8ccaf94b78 : SUCCESS in 8m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a85f6230564466aa194102ceb01c6b9 : SUCCESS in 6m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/db3953048f3244ee8a732af66f70a120 : SUCCESS in 12m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/42ca30a481a84f618482c7ce85a4e851 : SUCCESS in 11m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4980be39f713482e89e9353e47c40df6 : SUCCESS in 21m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/20548f4b94af4c4eb390081c326e329b : SUCCESS in 22m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8f226bafefc3466baf8f59fb3d6d5b30 : SUCCESS in 12m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6f04347dff7a4902bfd32d67bb37f209 : SUCCESS in 11m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d19b75edbb4b423a9a22578dc15079a6 : SUCCESS in 12m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82d145bbec7249f78581ac07b505dd38 : SUCCESS in 17m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b021c1cf03b4b7290ad0befd525f9c4 : SUCCESS in 4m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/06fa6d4c34144ce6bd0a139dc0116fc4 : SUCCESS in 8m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dc4f611b4395422aa6f773a491a812f9 : SUCCESS in 8m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/272b010b53b744e5b3fa54ccc4fc167f : SUCCESS in 55m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/98d54452ca5e432fa4e51f6a1b3d3afe : SUCCESS in 1h 02m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4ac44dc810c64c4a8901b09902a1737c : SUCCESS in 1h 24m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0f6b96b360b74875946944db1a501ab7 : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c933eff3cbf917fb0b418072168ca3c826f8c65a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 01:03:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a8648a1cf5313c1ce774db003e0353881bc9ea1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 02:33:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dcf21d3b06f43088a0a136042835417 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/003eb1b24ae5408f9229ea9349d76a90 : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7cfffc3aa6194b16a70bd10d9d9859d8 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc97f3e109da4143973f7298bee8bb40 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1e4e061345242f0897ccf3b1b36d063 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37042a2a48b1499fbe2714035d64dc16 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c0968127ca8b4db8bb179e242e523d00 : SUCCESS in 3m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4bb766890ee3431ca9e34dd6003a12f7 : SUCCESS in 7m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/41ec7ce09e6f461fb35523c4671bf365 : SUCCESS in 6m 21s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/553b202c977c4eaba123546cc44aea66 : SUCCESS in 7m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/14353189723d41d291e7368e6d5b2615 : SUCCESS in 9m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/69c2baa0f7fa4600acea66a2a5f210d1 : SUCCESS in 4m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8ddf14fd144141528e4d35a7d616399a : SUCCESS in 3m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/182568af977e4303b3b2a1b1b06b863d : SUCCESS in 8m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5dd61bd1f7a6440bb9533ff0c82497c3 : SUCCESS in 7m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4963af8e1963480996732e5fdaa2f9dd : SUCCESS in 7m 44s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/240a662c46dd4c8088b4effd2831aa6e : SUCCESS in 12m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/79745f048d844095ac7d33ddeae5e7e9 : SUCCESS in 11m 20s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/954d43d3ba0b490ba7d6732d71b55460 : SUCCESS in 20m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cf8f1efaf0e042c7b954bb7ac5588c42 : SUCCESS in 19m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/63f100fd1fe94ac3b46705d78c9f78d4 : SUCCESS in 16m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8e3f66a3061c47f58d9557c5c5530ca6 : SUCCESS in 12m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4945a56c10c14122b00c0ebd9b137fc6 : SUCCESS in 11m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/019eb772cc6c414aadf57f86c0f227fe : SUCCESS in 16m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/949c08bddaff433c9610d6195d8f8157 : SUCCESS in 4m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c04ae1bc42bf46e18621d6cc7b1bc0d3 : SUCCESS in 8m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8012574b5b4b487699088d5d59a3c0d9 : SUCCESS in 8m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/2ad3390530c14d71af4e02363d96c6d6 : SUCCESS in 53m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a5da9c1c4e44707b90c53c5850ccb84 : SUCCESS in 58m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/da4d85e6bcc14d1faf27d66fa9479fff : SUCCESS in 1h 24m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/92096f4d29be443e8709d59ab5c24ad1 : SUCCESS in 54m 35s","accounts_in_message":[],"_revision_number":2},{"id":"299d251a85f76c458b14ca9d8ff1f5ec84f2799d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 02:33:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e59c8b1dafa5406beacb500ada6c1b6b7c98db39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 02:34:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d7960ab3d844d06918e8b0731fbe81c : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b707a10f86e7a3dc77af5828c92451c2e54202ac","revisions":{"65549cb9f67268bc672db7291008a899e06b0736":{"kind":"REWORK","_number":1,"created":"2019-10-12 06:18:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/688232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/688232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/688232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/688232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/688232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/688232/1"}}},"commit":{"parents":[{"commit":"722f91a10762f0fc291635c21beb886f1716c01c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/722f91a10762f0fc291635c21beb886f1716c01c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-12 06:18:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-12 06:18:26.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic73e5a2b195f1189ae0db54e0bc97fdde82abc98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65549cb9f67268bc672db7291008a899e06b0736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65549cb9f67268bc672db7291008a899e06b0736"}]},"branch":"refs/heads/master"},"b707a10f86e7a3dc77af5828c92451c2e54202ac":{"kind":"REWORK","_number":2,"created":"2019-10-12 16:53:37.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/32/688232/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/688232/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/688232/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/688232/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/688232/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/688232/2"}}},"commit":{"parents":[{"commit":"722f91a10762f0fc291635c21beb886f1716c01c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/722f91a10762f0fc291635c21beb886f1716c01c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-12 06:18:26.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-12 16:53:18.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic73e5a2b195f1189ae0db54e0bc97fdde82abc98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b707a10f86e7a3dc77af5828c92451c2e54202ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b707a10f86e7a3dc77af5828c92451c2e54202ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
