)]}'
{"id":"openstack%2Frequirements~688296","triplet_id":"openstack%2Frequirements~master~I5abee6e993b6b6513a8f858781bfa94363fcc162","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I5abee6e993b6b6513a8f858781bfa94363fcc162","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-10-13 06:12:41.000000000","updated":"2019-10-14 22:07:43.000000000","submitted":"2019-10-14 22:06:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688296-1571090768092-f62c7b6a","meta_rev_id":"5332a7d33cd8e9297b9424dba57a9a92f687b48a","_number":688296,"virtual_id_number":688296,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 22:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-14 20:20:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-14 20:20:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 20:20:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-14 22:06:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08a06eda41ad16d392ac602dc74455de8725181b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-13 06:12:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"471d56a62038866029fab0a7f88e16910e3508a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 08:23:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bba9afd2630047de8b4f923948f3599f : SUCCESS in 4m 36s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f5360dafd07a45eeb396f0e8929c1d8d : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/756f842944a1497bb3a58a8641d2ed0f : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e210302581d460f959d94c190ce0e49 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c68d7cf2ad1f43a19e994a3781e432d9 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3a74cad316f46bbb302aec175eabd8a : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6bb5e0666f0b427bbe20fba926ca4ea2 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/95e84ca3e31648fabf1a97e4751523de : SUCCESS in 4m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bae2ab565a4c4b2793608f49647f8f73 : FAILURE in 6m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aa9ce6ec3f69478a8d27ae1091e1a111 : FAILURE in 7m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b84b4ab2a57740fbb05a9221d300a89c : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8cde0165e88d486c884f6f4b633a782c : SUCCESS in 7m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ea793a897e344e82b9483ee9910645ad : SUCCESS in 5m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/582320b3c57845949e8b3460d411943d : SUCCESS in 5m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e9537082454c4eb9966045e4405f050d : SUCCESS in 10m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/baa8be2453294f4db9842408964b700e : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ab61108d9f334bf6a33f463144ff3f8f : SUCCESS in 6m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c38513fa5fb64db091b541f9f18e0146 : SUCCESS in 15m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f3bc5b0fcd0740e3b2e5125117b3922c : SUCCESS in 15m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/87020e6b6bee419b9d34e3c2bd8d46a9 : SUCCESS in 23m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1c17712897ca429abc9cee82019e0cec : SUCCESS in 23m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c3cae7d46ae042dda5f381b9787fd2c0 : SUCCESS in 13m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/27f672d381324a8983131d1fe6067486 : SUCCESS in 14m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/521e8f3b6ffe4662b6475370e2d87b21 : SUCCESS in 14m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b0f6749216f74951a24aec58e11715aa : SUCCESS in 18m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7c709f9fb95047a5a0bda0c6f45f8241 : SUCCESS in 5m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/41a81a1adda34f82babf1d695ae8feb5 : SUCCESS in 10m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/928aeed6e93148e6802717c038bf2656 : SUCCESS in 11m 50s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/de1f1b45acec4845832cfba14a19ce51 : SUCCESS in 53m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/818ffecc570a428e8733f2bbaf2854a3 : SUCCESS in 1h 17m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2929ef9208794f189cf4233949d50007 : SUCCESS in 1h 52m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/537f93371ed64f58ace1a8e589d1070b : SUCCESS in 1h 06m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2381c417666f37e469c2531ac54705a28ac38ecf","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-14 06:37:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"790e0e6ed0f85e3834ccb45233078d3ad7d5af4b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-14 07:07:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5a8e9f26f9c8df9f6996d9eba71b32ebfaf62755","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-14 07:07:52.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"e9bee779aa66a619c22ec3bcd8bf0d5b2d46fd8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 08:49:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/693774ce29284982a62242c802f6a28d : SUCCESS in 3m 39s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b3c3ccab64524c19aff98ff77cefc7c2 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b277b33b2f482cbf1e6a8deb2b72ea : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a97e033142a4fd590bb722ca1836a54 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3becdec60b7e45c29c22409c6b0d7b1b : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42b49a6040c843bb82fb88676057f83f : SUCCESS in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ee9e94c0b444e348191e00cca8fe5f9 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ba304d3229b402b8136aeaedbdc3f7b : SUCCESS in 2m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/baddbfea27ff4a378fe75041afe7dc32 : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eaeaca500f85472ba9aeb4a180832d15 : SUCCESS in 6m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1f08ca6e99a5423a93c5115fabf585af : SUCCESS in 8m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/047894e3119548ff90ffe366dc7e1f5c : SUCCESS in 7m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f793eeb96f064cf1ad0dcb91ab4596e8 : SUCCESS in 4m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a67691b7ef904cae82f3295d28ae4a7a : SUCCESS in 3m 55s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2c2774c7db5e48fd8f776f8a8179fb77 : SUCCESS in 9m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3039636baf7d4402bfb16ef90530bd06 : SUCCESS in 9m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a0a49e194264ff1bd2d041b5fa6640c : SUCCESS in 6m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7cc31c31447b4e249013a8219e111982 : SUCCESS in 13m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/69e158d0509b4f7ca26a502df76b24df : SUCCESS in 13m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fe61e6fc31094779acab5f44b29e4ed3 : SUCCESS in 22m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0cd0552d50ec4616b066ad4b56f21ff5 : SUCCESS in 23m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/72a63dd0b31d4a77a6ddc65fa7d80a43 : SUCCESS in 12m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b8d3c02e73f54723807d453ac213122d : SUCCESS in 12m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5138d752b9514c74babc97ee33d35a09 : SUCCESS in 12m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/01ffaf8e5b2648bb9cc73852aad05a53 : SUCCESS in 17m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1446ab1db2c040ddb2bd60c9bcc39a8f : SUCCESS in 5m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/38d0cc207746421aa0fb19931e9cc09e : SUCCESS in 8m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/54bc1eca2cdf4df39e296da178997852 : SUCCESS in 9m 13s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d045d019c9584cb6950e57faa57fe170 : SUCCESS in 50m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2aad9746c2a41a6b0af5643482c6fb6 : SUCCESS in 1h 05m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d52eaf8eb6014cb09503b8fb46f38b1b : SUCCESS in 1h 36m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/632a5fd2530b41d2ba4c9fa047e44902 : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":3},{"id":"2382278d482665cb737c5f7419a0ee1aa50fe6a7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-14 20:20:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ad3f271587d5305502505159a87c3ecbb63002c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 20:21:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"320657b7212cc9b6f3a158274751b530850a6958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 22:06:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b585b988faed4cdd9f07be499031ed3e : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25ec67a4eef44d45bf093dce4e760c4a : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13a6be6dbc9842709b06cab330684c8c : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87b6a1f04fd14a4fb9cc1f8010914a30 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bb7feeef73e4727a0662f04f87d23b8 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f7a1a77608a445a9a8da9f34e103e87 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/18a95182f590461c93bbf1cde2b72f67 : SUCCESS in 3m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/03d0758cf6e04bce88d6221de8c58e4e : SUCCESS in 8m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1700b75eda824b068e0b00f7ad0c2a60 : SUCCESS in 6m 50s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ea6b89b752124effa935d63406165e8f : SUCCESS in 9m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8d5a7c447c3849549e62b0d1b605bb59 : SUCCESS in 9m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7d2f76cd312e4d5684bc014a1f525e5c : SUCCESS in 4m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/afde1237ae374a31bf51d1d5746a7ced : SUCCESS in 5m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5d62febc07954c7abb618835d42b72c7 : SUCCESS in 12m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/66dc2a7109e9472193247f8be7b56507 : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/efc6ccdc2a614a98a080bf2203a55193 : SUCCESS in 6m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/741fad50f01e46cb8d4868c99f58f42a : SUCCESS in 13m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c61906ff190049e28f99c97f1319671f : SUCCESS in 12m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e0e747c892394718bf2ee935b4e5fc53 : SUCCESS in 24m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/608798928dda467cb737b6ee85534cf8 : SUCCESS in 23m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5432f0d21963428597a32dc5b2f77e64 : SUCCESS in 13m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cb7fcdd0d9e14a39abf2e0b200476eb6 : SUCCESS in 20m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3c8fd8d211aa42eca09b4207a53c91a9 : SUCCESS in 22m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/031ba2c719cc433985b32d17bf07a422 : SUCCESS in 16m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3d93576072c841fbaa72d5a57a5f7d28 : SUCCESS in 5m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e38f1eda2f8b464aae5217a44cfa827b : SUCCESS in 9m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9e4c22914b3e4601a602c49ee826b1df : SUCCESS in 13m 36s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9c5fdc6690b1491ca9047c753dc86540 : SUCCESS in 54m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab9de7fbda374f2ea6c4e3297c74c304 : SUCCESS in 1h 05m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9fe9be1a94f1461a8418d358e07a6fb3 : SUCCESS in 1h 35m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/93557f68cada4d61a9a57cf589dbd8df : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":3},{"id":"8f68521677f234f7aa7353752d6afb7b3ab4afc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 22:06:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5332a7d33cd8e9297b9424dba57a9a92f687b48a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 22:07:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50fcc2d83ae54f7fa9239facabbc7e81 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"441c8325512cebae7be04037fbb8ab9b9262bc22","revisions":{"3905c8a521603c6623bcd3966ad8321294f3c85c":{"kind":"REWORK","_number":1,"created":"2019-10-13 06:12:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/688296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/688296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/688296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/688296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/688296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/688296/1"}}},"commit":{"parents":[{"commit":"b707a10f86e7a3dc77af5828c92451c2e54202ac","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b707a10f86e7a3dc77af5828c92451c2e54202ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-13 06:12:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-13 06:12:38.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I5abee6e993b6b6513a8f858781bfa94363fcc162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3905c8a521603c6623bcd3966ad8321294f3c85c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3905c8a521603c6623bcd3966ad8321294f3c85c"}]},"branch":"refs/heads/master"},"00eb3d16a15ca9a87f44ef8a8f5b7821101ef9bb":{"kind":"REWORK","_number":2,"created":"2019-10-14 06:37:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/688296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/688296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/688296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/688296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/688296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/688296/2"}}},"commit":{"parents":[{"commit":"b707a10f86e7a3dc77af5828c92451c2e54202ac","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b707a10f86e7a3dc77af5828c92451c2e54202ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-14 06:37:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-14 06:37:47.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I5abee6e993b6b6513a8f858781bfa94363fcc162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00eb3d16a15ca9a87f44ef8a8f5b7821101ef9bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00eb3d16a15ca9a87f44ef8a8f5b7821101ef9bb"}]},"branch":"refs/heads/master"},"441c8325512cebae7be04037fbb8ab9b9262bc22":{"kind":"REWORK","_number":3,"created":"2019-10-14 07:07:51.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/96/688296/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/688296/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/688296/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/688296/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/688296/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/688296/3"}}},"commit":{"parents":[{"commit":"b707a10f86e7a3dc77af5828c92451c2e54202ac","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b707a10f86e7a3dc77af5828c92451c2e54202ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-14 06:37:47.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-14 07:07:39.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I5abee6e993b6b6513a8f858781bfa94363fcc162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/441c8325512cebae7be04037fbb8ab9b9262bc22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/441c8325512cebae7be04037fbb8ab9b9262bc22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
