)]}'
{"id":"openstack%2Frequirements~688752","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I11388348d82e3be3bfe96d9d3b655e16a8b97bbf","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I11388348d82e3be3bfe96d9d3b655e16a8b97bbf","subject":"update constraint for tempest to new release 22.1.0","status":"MERGED","created":"2019-10-15 16:35:23.000000000","updated":"2019-10-15 22:56:53.000000000","submitted":"2019-10-15 22:56:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688752-1571180213973-01385176","meta_rev_id":"c4e370a0305443d59a585af69add8f8181478381","_number":688752,"virtual_id_number":688752,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-15 22:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-15 19:28:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-15 19:28:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 19:28:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-15 22:56:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2d5d43ad6339bcea59f5d60330f4a89ec9ec823","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-15 16:35:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da27ec30333cc12cfd4ad26fbae9a599075fd402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 19:22:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3a9e296b65a7435083426923c2aae71d : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ad23947d594ce2858052debe10a5b7 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32172706b4b94a75ab3af5e0a31ef0c7 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67245250436a4a83aad89ee862c07a88 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/055582cba89a4c4d9a4f4ecbe5518820 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1bfbf2aaeb504452b977115730b2be07 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ac3d2bad9594d1ebd701cc07a1fc2e4 : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bd7ebd2de5a6420795655745ae4ebd3c : SUCCESS in 8m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db29beec0ac247abab004d911783024c : SUCCESS in 9m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/373c5917b203412cbe24a9876438a488 : SUCCESS in 11m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/43094d929f394f24836a03ea8ab81b70 : SUCCESS in 7m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5f512e67af3b4d6d88bbf4565085fa9f : SUCCESS in 6m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/78798dabc376435bb1a699fb2b85ca12 : SUCCESS in 6m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1c6bd461bf24439893ef0618f26431a8 : SUCCESS in 12m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/96d2e87aa43249deb14b9049012f27cc : SUCCESS in 9m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a900af3fcf57414c9f04d53bf501fd08 : SUCCESS in 12m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4b4af6dd9d214950b76b42914759ed40 : SUCCESS in 14m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e085a5ddff91424f9ca9e7561a993b68 : SUCCESS in 14m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2604357400c34b2db656e5cdf90bef2b : SUCCESS in 26m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a21996c716f24f8cbc91376a813cceab : SUCCESS in 23m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b94e1b0498b7411bb20ed6095aeecb34 : SUCCESS in 15m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c1b851c8b72746b2ab63e89117224045 : SUCCESS in 16m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8e2f735ab6734ffa8f28d59b8b56b715 : SUCCESS in 14m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ffdbb53abcb490bb18d64820e0b6187 : SUCCESS in 19m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e61a5bfda46436b92f8f6de2e032f98 : SUCCESS in 7m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d4eb948a07304393935e7ff71904fb1e : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6a3efff61f544d83827cb741b874e046 : SUCCESS in 8m 37s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/47efb8611993443b84080457aa179d45 : SUCCESS in 56m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89b6e2890f0841d2ab03aa47b62d5db5 : SUCCESS in 1h 09m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/323f7baba24e4fbea56a982dcaf5cfde : POST_FAILURE in 1h 47m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/609e97c316c243e281f6670458e71d4b : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4bbc8a769671fa2b0ca23dbf82e170eed53b0c9e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-15 19:28:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6b318a56fb82b33fd98ff321506c619c92533909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 21:06:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/fbcbb7800e8d42039b85f37be1f97d24 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/880809e6bdd74759943ea219f2a23061 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c067711b237e4e5f8709e83a56fa4a40 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/399290ef706c43d59a560b0a2fe9ab7c : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf192997a4074b689134f4b7d126fd21 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2218f0f24065427893eefd8bee10e4ce : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68abfa0eb0764de6804982ee95424c7a : SUCCESS in 3m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8faf403d0a2b49aeaf1c642e9ccc0bd0 : SUCCESS in 5m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3b912d171fbc40c29f342bab08b5640f : SUCCESS in 6m 09s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/16395da39e22408e8125caa7b0ae8a10 : SUCCESS in 12m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7940751d27d0498aa5624acabdf56f0f : SUCCESS in 9m 21s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0ec2e576f3e74ca9a0ed0832cdb12d6e : SUCCESS in 4m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8255304300fd4bd990fe6765d7f5ce99 : SUCCESS in 4m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b9931001b4e1458e814618951c03f92e : SUCCESS in 12m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3d62172231d74cdabe95ea737437b9fb : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1362e36123e74fbe9fa69fecccdde134 : SUCCESS in 7m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/26d7782745af4e3089ffd5532b317357 : SUCCESS in 22m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1cfff4aab2b541b0ae5068367a8cd154 : SUCCESS in 23m 14s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b8918dd0d8a8470fb4d53e2b2a4c32c3 : SUCCESS in 24m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1b38d03d7faa4e6482f4a7b18b714b1e : SUCCESS in 23m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e5f118c8ac764e6fbe0b4da43f950bc8 : SUCCESS in 12m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dc5abbc5d2c947c1b17b4c057519ee75 : SUCCESS in 16m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4cd25851f32544d8a0b15d6e0216ca19 : SUCCESS in 16m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2bac25312462400dadcdcb99e8588932 : SUCCESS in 25m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b5ad840df8441bfbe2f1784920ec42c : SUCCESS in 5m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b4bf0768eaf848a89a55f366af42fe84 : SUCCESS in 9m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/11e4a5e095a249bb8b9409c1a2d66816 : SUCCESS in 9m 56s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/cacf953d9f514f59abc4aebc6c476858 : SUCCESS in 1h 11m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f6e4fee70394a0aba0c8abc99ae3ce5 : SUCCESS in 1h 28m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e24db208da3f41018693088ccdaef68a : SUCCESS in 1h 30m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4a6a13552d0c487fa7e0d0ab1208538f : SUCCESS in 1h 08m 58s","accounts_in_message":[],"_revision_number":1},{"id":"84fb37785ae1ff5114d2aba479fdc5f932a171a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 21:08:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a731a1ae849de070cd391c26338df3da9a563f0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:56:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa80f91721d24087acaf10a476bd6a11 : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba2d2f73ebdb4000b353be5b7dcd619c : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fb1672048ac446bab8f1a3855868906 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd1e648948ef4f4cb2992034d5a78a68 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7a08db2233b465dbf306559ab8e60d9 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fdf5b8fe97714c64bc4110237d9681e6 : SUCCESS in 2m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/69febd762a884a1481e1b8bba67b23bf : SUCCESS in 7m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9cb833859ac744ad8e7ea1d0aec77d1f : SUCCESS in 6m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/df593850072e4e8a90f3d5790fc3e162 : SUCCESS in 8m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fdafeee063be4d33af5bf88ffa8fe6ff : SUCCESS in 8m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0ccde1a45c604a709be3285df5300257 : SUCCESS in 4m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8c68c3143a504829955a82aca0c88312 : SUCCESS in 4m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ffaf7c6270be4e609ea55e43fa4bbf09 : SUCCESS in 10m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c6b7e820b250458cb99abf1615d666f4 : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a546f1fda2d94b9cb6c0c01b36038d9a : SUCCESS in 6m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ecdfbd8498b74f569c2ef181dc7577cb : SUCCESS in 18m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c2a140096ce8467280e2362d8c9739e6 : SUCCESS in 13m 08s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6915efa0997b499fae07f178776ae9fa : SUCCESS in 22m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/37b5d7d0636c42cab70c5f568a05fa18 : SUCCESS in 28m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ccec0897dcf040dc9d7c4c3efe14b306 : SUCCESS in 14m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7299569675114648a7e56be6938bdcf7 : SUCCESS in 12m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f10c061b172e4255b8e810d69f8e7fc5 : SUCCESS in 12m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a46314bfd74c439abbd602c1eeb0ab10 : SUCCESS in 22m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f74129c709c40eebd3ae8b187f2debe : SUCCESS in 5m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/96a5c9f166074f92b3d936ce6403200b : SUCCESS in 9m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7155f39900c74e3992266cfa73b69a70 : SUCCESS in 8m 53s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/bdbc0024252d440c858ce1e17202a9b1 : SUCCESS in 55m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8837f1fefca24801882acb22c333920e : SUCCESS in 1h 07m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/62be590683a2465b8dbdbca0ca027f00 : SUCCESS in 1h 33m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6eecdcad01564998a7302697050f7f54 : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":1},{"id":"12c8a306dabfe08f37ecdba3a10f7355d5cc495a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:56:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7790dc55d28cd3e1cb3c9d42d0e7c91b581a742a","revisions":{"7790dc55d28cd3e1cb3c9d42d0e7c91b581a742a":{"kind":"REWORK","_number":1,"created":"2019-10-15 16:35:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/688752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/688752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/688752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/688752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/688752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/688752/1"}}},"commit":{"parents":[{"commit":"03f2a88a302a146257d6239f3421ae911b3fb4c2","subject":"update constraint for ceilometer to new release 13.0.0.0rc2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03f2a88a302a146257d6239f3421ae911b3fb4c2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-15 16:35:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-15 16:35:21.000000000","tz":0},"subject":"update constraint for tempest to new release 22.1.0","message":"update constraint for tempest to new release 22.1.0\n\nChange-Id: I11388348d82e3be3bfe96d9d3b655e16a8b97bbf\nmeta:version: 22.1.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Change-Id: I6d049ab6a1b3a69b3aab185dcf13109b70d2a541\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7790dc55d28cd3e1cb3c9d42d0e7c91b581a742a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7790dc55d28cd3e1cb3c9d42d0e7c91b581a742a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
