)]}'
{"id":"openstack%2Frequirements~689079","triplet_id":"openstack%2Frequirements~master~I7b3a409a7b76b1d5ee63149143887168d8db24df","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I7b3a409a7b76b1d5ee63149143887168d8db24df","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-10-17 06:12:12.000000000","updated":"2019-10-18 07:32:20.000000000","submitted":"2019-10-18 07:31:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689079-1571383866763-9e252c24","meta_rev_id":"e166cdff7910d49d6f78ca393187ef46c39b14a4","_number":689079,"virtual_id_number":689079,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 07:31:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-18 02:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-18 02:05:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 02:05:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-18 07:31:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be91cd8ca50928153b722529e964409bdf77cca0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-17 06:12:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9cf2d93009ab5fb9c7a195a27ec6a98e94ef9a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 08:00:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97e6d82339984cbe8f2872bde0448c54 : SUCCESS in 3m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7ab8acbb98414f5da7497ba6e81fba93 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcd2bb276b384482a5014ba3ce6d103f : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/127f03ef6ecc49979dffba76d062c176 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca37e736ecaa49b898ad3831567b6bae : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abff1fabd70d4d01a8e7602806f232ab : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82cd79110eec452aafa5609d376690d1 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f3047bf759f43049f260b3b8c4e90e4 : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c879dcc999294917b0df71054a8c8d44 : FAILURE in 10m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/978d612d6c6840ff9524937e54c9a984 : FAILURE in 7m 22s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/efbee280976944cc8a442e1824ecabe7 : SUCCESS in 9m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2d7698858ae9496a9f8e20bca7615c42 : FAILURE in 9m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f01c3e06b16d47f393757d4306ce4f9f : SUCCESS in 4m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bfe27eeb48214716a91239fa64b170f3 : FAILURE in 4m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/66d0991bccfb403294ff0e01959fc71c : SUCCESS in 9m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/10a6a82b4ef241b59018e5b97264bf1c : POST_FAILURE in 31m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5dd71a9112e1422f84155199a694640d : SUCCESS in 8m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/065195b4d9f6494c8809d87b74cc3cc3 : SUCCESS in 12m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9f3ecd14c0f7427a9266d0c5c59431b2 : SUCCESS in 14m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/49b26b0aac9a41bc9484b5943646dd1f : SUCCESS in 22m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/080bf1e788e847d3a0d9866068fb37ae : SUCCESS in 24m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a68b6b9d31c1475f831c2600d22f712f : SUCCESS in 15m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9af22223dfc94f159f0a2340a2b15275 : FAILURE in 12m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1c3ff504f8f1404ba897848344ef5c9e : FAILURE in 13m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5af65a9f63b74e4eaef3b6bbe149ea3d : SUCCESS in 18m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d15e00bba444c4fa7e8eaf0bc761cfe : SUCCESS in 6m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/43a7220dc77947049e29c3b89ddd443c : SUCCESS in 10m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/391ef3db0b894ee5ac883d3b5be1cca9 : SUCCESS in 9m 28s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5013711936f54624b93d016b222affab : SUCCESS in 1h 18m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4dce9db8c5f4d8592852e54fdb69fe3 : FAILURE in 1h 00m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/51ffdd955c31414690476e5c3219b75b : SUCCESS in 1h 38m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e2cdf186a7094b84865230098137912b : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2cec508d1b1a1cbc57f9d584534f44f862a3d40b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-17 16:51:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b44e36825c4e93acfd697c7fc4505b8ac3007359","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-17 16:52:32.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"b46a322a35e66868925697e9025554f26d27ff5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 18:33:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed1d75da206463e98f67468f6d7d623 : SUCCESS in 4m 38s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a702d5dd389347198f9d0568c6bf43c6 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e9305513b60463cbeb8e43f7d1fa50c : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/81ba7e330d99411481f1e725f2276b72 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8547dd107664017943fef4c2eb4e243 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/175dfec5417442de80a64b268c252fe5 : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/565e60fc36674080a05836278d0e87cd : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/05cdd9a2d01044f08d6304cb25188704 : SUCCESS in 4m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/03cf26100c43415692e55e4714999b78 : SUCCESS in 9m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0a844d0f88bb4854a42a694155fcbe8a : SUCCESS in 7m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f912d73e2e9e42979cd99b5d5737f10a : SUCCESS in 8m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0667b80927754be1abf4396eec7b4ce7 : FAILURE in 9m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4db0f497a13d4937a3eb423d45811c9c : SUCCESS in 5m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c82a08b7dcd54cc1800350742cd192dd : FAILURE in 5m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7faadbd49b294ce297a49904758ea2f6 : SUCCESS in 10m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e48c8ed65e0840f08f3cb5a19cf905a8 : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/85a61d5894954324aad5d690869db14d : SUCCESS in 6m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/21478ef34eb44cdb8cb1814267085dbf : SUCCESS in 13m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7c3c95ff7148494487c0fd716a45d10a : SUCCESS in 15m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8cd3884285484ed29ad62fd89586fe12 : SUCCESS in 24m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0c69c65061704e9f85ea605ffe10f06a : SUCCESS in 20m 45s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/57d30e3ee0e04b5da0aa3d928b7aeb78 : SUCCESS in 13m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/81892c169c5d495082f675fca23ec24e : FAILURE in 12m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/460086f517a5422e93a8e3b104702d05 : FAILURE in 14m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eae9947ad9f040b9a14265c895cef205 : SUCCESS in 17m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5c3a71e0112a4401b9cc883e7d0d1539 : SUCCESS in 5m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/eb09f235c7a342519db5dd62b9e6fec7 : SUCCESS in 10m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fb0ec95f995543deba00aa03812f8ef4 : SUCCESS in 14m 55s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5a56cb055d1a408b9d758921bfb7315b : SUCCESS in 44m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4687cffb411345d1a29a91a2e9c7d161 : FAILURE in 1h 06m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/45686652cc9242318dd3cd4fca662eba : SUCCESS in 1h 35m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e16ece3f93ef4d71a1a580c90d25a36f : SUCCESS in 57m 36s","accounts_in_message":[],"_revision_number":2},{"id":"e12ff555e974f862fb0381faa35f47e6b7c0d7aa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-17 19:36:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"26b215cf748cab9ff94fb0e6ec3bcbe37905b4fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 21:23:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc6b08a59a0417baf31d5cfc110badc : SUCCESS in 3m 26s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2f71e0f17a5f4461b1fc8aa65dd92a8c : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3175f479a474f2ab4e951b0cb474d5a : SUCCESS in 2m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d6ae4d028b1c4c398d2b9143dba96c59 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a348439fcb14f8ab5c139d8a2238625 : SUCCESS in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12ce12806c57454aa7179e4b1c5526a8 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca83c3d617d544a2b6a507a3fb3d449e : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/705c9ff993ff446e85b2c7094b14583c : SUCCESS in 2m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3459298605d843d0ab44ef1e76bc5fc5 : SUCCESS in 7m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6e298907695d449fa06affbe0846b716 : SUCCESS in 6m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/212001ab261240d186501811406c2366 : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1c23278f50ad4c33b592f87ab086cd12 : FAILURE in 8m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e8f7e2228c2b4f7fa63da08eecd6d7ff : SUCCESS in 4m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0853f72b4b1e4fac8f3b6a9f5c167c4a : FAILURE in 4m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/eb5cf3c2b426477bacca4bfd87a57d43 : SUCCESS in 9m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bb366af5aa4c4f0f872a7086a7887c7e : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dde33a2589454ce9bc558a50d6695e11 : SUCCESS in 6m 17s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0f45e6a1bc2b4119a8b5d924f27f8351 : SUCCESS in 14m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/26d2efa835a246e39d4b92c6d91f9a77 : SUCCESS in 12m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b1d1a78480574646a3b9fc241bc3bba5 : SUCCESS in 21m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/217617118b3241c19b808d55a8d63a2e : SUCCESS in 24m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/63da6874ebfd4d419dfded1589013452 : SUCCESS in 11m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fb411eb15f2e422095d27f54c92affde : FAILURE in 11m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bc5b38115fea4e2f9f1f61ea8da068f2 : FAILURE in 13m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f5f2675ec624fa3a22cead124e4b082 : SUCCESS in 19m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3243d32ff0f4ba9af778ebd6b3cd674 : SUCCESS in 4m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/215292a3db0241d6a166660b94f4022d : SUCCESS in 7m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/768d4e3ed55f4b30a55cc548e176e459 : SUCCESS in 7m 56s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/02858076332d471b8ac18228a75be5e2 : SUCCESS in 58m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebacc7d5d4d74fbf927af94bd0540d76 : FAILURE in 1h 10m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bde72eec81be4a4bb730f1b68f707b42 : SUCCESS in 1h 36m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c9a78e29dbaf402ab84671d1bd39fa60 : SUCCESS in 1h 39m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d247e4e379b97fb2d1121a2a663bb783cc6bc333","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-17 22:15:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c03ae56816afe00b089b608b9bd00fa8a092eab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 23:57:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6f8fc8716d4884bdeb631025af7904 : SUCCESS in 3m 39s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/43dfc28dbf094c4282777e29698d2e01 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58144fe740ad47de8e38d3e3939858ca : SUCCESS in 12m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a7197f9c6f048e5be0d2836c3f16c3e : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/772885d3a20f4233a3d88d36a57282ea : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b57605df2be34d1aa063fa9b7c7a7043 : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4ba199c785d44dcadfd70fba9ecfa37 : SUCCESS in 10m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7769761eb5d24fb5aeccdd71f4f541a5 : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1f00953787c54efb96c0c7a6352ed75f : SUCCESS in 7m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5c1853367a5346f7aa32946fb50343c5 : SUCCESS in 6m 47s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/13b8d466779d40bf85527d04d8048477 : SUCCESS in 9m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/81627dc37056438081a696ecd8276e8e : FAILURE in 8m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/96f9dc44983743828af67f8977eda04f : SUCCESS in 11m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f8c423b999794341b19bfdf39915b92b : FAILURE in 8m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/48df59b47c64440aa14503f7fcab0766 : SUCCESS in 9m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0cb3ff5986b946b7aa9293f7bd79ab94 : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b94684b53aff42a1a1db59749df811dc : SUCCESS in 9m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/def212d8ad9c4b6cb8774462b2c7d8f7 : SUCCESS in 20m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/42a1faac660249eeadafe52501712327 : SUCCESS in 19m 38s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6065cfa478d143e6a912c8b7c23b5469 : SUCCESS in 22m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6dec5e62510c4f1e8fcfe15881dc9208 : SUCCESS in 21m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/00e233eb2a1c462995eaf2f18d83d4a3 : SUCCESS in 13m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/19ce55045f3349ed846d79326f34cd75 : FAILURE in 12m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4737d7003f2c4f06a47ab202709c1d1b : FAILURE in 13m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc77cff6b6424191b12c2a38cf18e515 : SUCCESS in 18m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1bb2fee4a8b347f79236be22614f3fd8 : SUCCESS in 5m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a3c7ed35c1b14be398b7f92439903133 : SUCCESS in 13m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/40078c2bdf8346e7b40fe9bab31f99a2 : SUCCESS in 10m 09s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c8eb9c989d174e3b932082723883973e : SUCCESS in 1h 02m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d168cf487de3498e8e15b310ee28c333 : FAILURE in 1h 09m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6761e28331164b92bab5683b3ae59a6c : SUCCESS in 1h 36m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/47b6d8fc238441e68a0ab083c4536356 : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5a19bb345626bd05bc6a2de05125a2ba5c5c4308","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-18 00:04:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"69373379d6549cfac60c7b378e71279050bf5e99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 01:42:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01a1202b46ca4f6ba4a0a8c09528e591 : SUCCESS in 2m 52s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8886c78733d94c9d81d19842ec9835b0 : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fa873b86256496eb707255d44e346b9 : SUCCESS in 9m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39bdc4191f804ffbb40e99b47e0317f6 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f1841d6b4094ad7bb24feb3d2b331b2 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97ce2dd8802b4c1aa81eaf54677577b1 : SUCCESS in 8m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1bf80d2596e54f4fabad73554b46a689 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c84447b2fa8a49e7aabff777ed26487e : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5c4f0ea140a94f2c8c9ced5deeeefcfa : SUCCESS in 12m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/94a57c8ec7594e8bb66816d0c86dff5a : SUCCESS in 11m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fca0d1d11f214c52be3c627b11652410 : SUCCESS in 15m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a45014fbe3b644e3929fa3b4ee53e4a9 : SUCCESS in 9m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0407a3b7e32e4e00a50593f32f968334 : SUCCESS in 9m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f7dcd18b9b824955af0124166e7f278d : SUCCESS in 4m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b4be85ee7a354a19a141034e86d7cf36 : SUCCESS in 8m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/17b19f63ce1e4039a62b7202789df3bb : SUCCESS in 8m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d661accc7dc49dc9fd6e06387aaa3a9 : SUCCESS in 8m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/654bc1fd161b4bcf902b8fa39c400137 : SUCCESS in 13m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1de453ce00c3446fa2766091df494b6c : SUCCESS in 14m 47s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ae8063f786144d669260e49f9bb43fc3 : SUCCESS in 20m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eb5c5fada09e4fa6afdeb3a90c8da990 : SUCCESS in 20m 50s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/fb3a759b05aa412a85691474d281de04 : SUCCESS in 14m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a554346771d444ca8c22c64b81ee00ae : SUCCESS in 12m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cc1bbd5be1de4162989cbf0f51c74a91 : SUCCESS in 12m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c2a86578367444daf2bbdbdfa5f5f76 : SUCCESS in 17m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43e02f908d064288b729f41ca4250807 : SUCCESS in 5m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bf4c72df84514a9ea0699682d6a07f84 : SUCCESS in 8m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cc648d814628433d9b5f575f070d450e : SUCCESS in 9m 07s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/66f247f09c2a4f8bb0e0651d00831b6e : SUCCESS in 52m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec47db2afb194b0b973cf65343cd8038 : SUCCESS in 1h 07m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0c64bf2ebec342c9ba545ec7867f268f : SUCCESS in 1h 32m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/18b2a660bd41448e8f3fbcebf2594650 : SUCCESS in 51m 03s","accounts_in_message":[],"_revision_number":3},{"id":"96ded49f8e65613def79c88499cee450ff3bb7eb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-18 02:05:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9eb5a36bb7e68458746ecb4f16bd5ce8ff3457f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 02:05:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e585cee378b0d38ac32a239b87e6414d91bae61d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 03:35:44.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dafaa31b9c194748885c03356601afea : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11252849937a40359d63eeeb5704936a : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67c263453bd14846b43efc6064fea291 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95f08129662b4335ae497ffd27cafaf5 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/263313675f0c48cd8c3702f60db5cd28 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3df0a756428a4876b54313ebd1d3ad15 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/23a0bbde49bd4a59a59bebde8dead119 : SUCCESS in 2m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e589546d15544117b9fac543fac30973 : SUCCESS in 6m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1e2e21c447f94f4faf7734dcd2a9301a : SUCCESS in 7m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a286e8c32d0847e398226c23f82fe04d : SUCCESS in 8m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/39407206b6004dd88f545ede3b1a7d55 : FAILURE in 10m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4d03615ccc904694a42295a0a919fb1b : SUCCESS in 4m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/335d95a3c89a4779a23ab4096027df9e : SUCCESS in 4m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9526d0396a814caa963a37d43f4e627d : SUCCESS in 13m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ec9b97f043694cee9ded730984abfe79 : SUCCESS in 9m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0866b54eace94152bb26bd3fdb098694 : SUCCESS in 8m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/02192483417d4abf9e77edbab3b0aa13 : SUCCESS in 18m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6e5f272d8ca64964a5ac289138317e09 : SUCCESS in 17m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/72bbf997cc3c4d61b753cbac28b26cb6 : SUCCESS in 24m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/82fea9fffe0b4b17ad3f7305c2d41120 : SUCCESS in 29m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1ee801fcf01c45fb8ba7d92e8b7763ff : SUCCESS in 16m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/10706d1d75b24205a7dbb7981db264c0 : SUCCESS in 15m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f2c62a777a364ed2b5a62cfc685c7e34 : SUCCESS in 18m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cbb76f14c4344ce196695d089c931b3e : SUCCESS in 24m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fd27829057604d45829f3cc8e5ebdd87 : SUCCESS in 5m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2f2f85274c5a4e43955751533e79c061 : SUCCESS in 7m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/62cc6990217849f295f86940a231b88a : SUCCESS in 9m 53s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3fac3d4030a0407a965647d40fb11260 : SUCCESS in 56m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bfd9bf8bf276468baafd26b913e7f429 : SUCCESS in 1h 07m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/58b4064b8b584ba48e6f6325d99a20ce : SUCCESS in 1h 28m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5c6b5696170d45bfbf1a6a3673614b36 : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":3},{"id":"14e7ee35baf0462ef23a1cd9359b540c9d347f19","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-18 04:05:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"98be1d543efff5efa52bb9f82bcd4e25c4ab69d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 05:55:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/366c431b762d4b5c8400c9ad3e3a9dd9 : SUCCESS in 3m 04s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d22defca90604242bfae86c427011fca : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abc7a641d0d545b5ac362d76d39824b7 : SUCCESS in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da0be124c1a244208ff7c0009ed8e27f : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a47903a49c249318218c0b8771ed9d2 : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7261c79320464f869a577b2830766492 : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1a1766cc9c74294984d44d6a99216ba : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a179278f847d4a65acf7edb67272c93d : SUCCESS in 2m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1478770167354ea8a20e639b845840da : SUCCESS in 6m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/57f5444792474ecc819c32ce844ce777 : SUCCESS in 5m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2c333cc43d1841e09104f90b42b3c863 : SUCCESS in 8m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0ec9d9726bb1497fabb6167ab3d432f0 : SUCCESS in 8m 05s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3148cd4e31fa477bb361c30bf2ffd095 : SUCCESS in 4m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ef835618d90f450f85f7b0bc8d45d6fc : SUCCESS in 4m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/134ece1f24b8487f8787d2bfeba147b4 : SUCCESS in 10m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7ea9763264ce44cf94c268232c98b413 : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/872bb691bd524cf2a4fa14c425b55424 : SUCCESS in 6m 36s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bea0d0e82f9e4b24a4c524afb9a57918 : SUCCESS in 11m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9ae3d391cf3f4793b440b76be701bcb0 : SUCCESS in 11m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2283abd24d5d43d4b5c7fdc5f4b9a4ac : SUCCESS in 21m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7ee6a4be7ed64ff5849352e3c7f0290e : SUCCESS in 21m 45s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3356a530d5e14cda94e603b7aab91bab : SUCCESS in 13m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7a52a7543d1d4697a8751d2a7034754e : SUCCESS in 10m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e2065e4d74e14e5f8e470854eff4b569 : SUCCESS in 12m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6c09e195532483aad190059797303f4 : SUCCESS in 17m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e0fc691e53034a32ac1c264e580b601d : SUCCESS in 4m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2d0b76e6600046efb56d4562cd395992 : SUCCESS in 8m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d06fd82008ab486db0343dba684c972e : SUCCESS in 8m 18s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/028c863d0e2c4a4fad9223c8d8eb22bb : SUCCESS in 53m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc09a9d91c384ad1b592247aa17ee81f : SUCCESS in 1h 04m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0eb89baaeb6049309130d01b659fd058 : SUCCESS in 1h 43m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a963fc93d54e47589d14b8a34566d1c4 : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":3},{"id":"1a52f2240912f8b3b122fc1296d5799c742a7818","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 05:55:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"31d0cf495a37a3591099c2a5cf108099eeb3bf72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 07:31:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a3605184fdaa42a925c308e4b57e8859e233ce52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 07:31:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca7730c422384278ac6c931dc8a1a7e3 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/457908891290440fbb0a097a161a237a : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f5d8b7204034780936d75fdfcdd42af : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/738ab17c47624fb58ab44316ecc17726 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d93b7575bc44f5b9f2f7d8cdbd474b4 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7eb16130c5b64e1fb44fb2bc5c998357 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17e9d0713b3f4836980570797a61f3f8 : SUCCESS in 2m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6270b33a07794af1b518fcd8c129d398 : SUCCESS in 6m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/48aaa57d6bba4284bfed4df9c9624ccb : SUCCESS in 6m 43s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/dad1404e975940c9a55a178b517d48e0 : SUCCESS in 9m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/825ec37c4e3a489099d4cd312a0358a8 : SUCCESS in 9m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c4d0c949d5614ea89d5a5ac4bb1a3ce5 : SUCCESS in 4m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d5f0aa7cddd642dfaa19d3bd37cdeaba : SUCCESS in 5m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d39c020fbc9d4450ba65aa84d0796c5c : SUCCESS in 10m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/04e1bf3014a9487881193a15a3d15f51 : SUCCESS in 8m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6f854b7a486048bd99348c08b6f8b737 : SUCCESS in 6m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/308d7bfaed3543b6b7c0b42a0f61ff85 : SUCCESS in 14m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d958c0fca4cb4bb9968562447e2aa415 : SUCCESS in 21m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ff5e5914ed9648b6a04b6e98aaaad6bf : SUCCESS in 21m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d1f44595ea864acfa5a2e393d5c2deb0 : SUCCESS in 21m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3625ca48ec524adab3107a14fb6032e4 : SUCCESS in 17m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9f35f82ac945499996203b7aa4d8089e : SUCCESS in 11m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b806cbd942b34bc0acfdf49519ffb705 : SUCCESS in 15m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ae78a002c664561862a1534ed46ed8f : SUCCESS in 17m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/164ee8aa3b2041aca77e6069bba34d3d : SUCCESS in 4m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/643a1740b2d04e2395a188d9ab598b43 : SUCCESS in 9m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bcfd87a851844f6682700646d3e5f96d : SUCCESS in 8m 59s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/bf1f1175e91b4a609938ac763d1e3ff9 : SUCCESS in 53m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa256a1f5e0a4359b2b138a1f06b0c17 : SUCCESS in 1h 27m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/59467478a1704e58929933d9cb16c10d : SUCCESS in 1h 32m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/de86050e1a5346a399bcc0a3e2774e1d : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":3},{"id":"e166cdff7910d49d6f78ca393187ef46c39b14a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 07:32:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5da6bafca82a4ac0a525a7d932a6a47c : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9f0d9ef478e506334494a906f89bb29ba64f7e59","revisions":{"356807a51563ca268eaa31f2a7fad3dcd560ba9d":{"kind":"REWORK","_number":1,"created":"2019-10-17 06:12:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/689079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/689079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/689079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/689079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/689079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/689079/1"}}},"commit":{"parents":[{"commit":"e51596e97a3ed0fc72c4174c3b7ac485268839c8","subject":"Merge \"update constraint for os-traits to new release 1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e51596e97a3ed0fc72c4174c3b7ac485268839c8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-17 06:12:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-17 06:12:10.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b3a409a7b76b1d5ee63149143887168d8db24df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/356807a51563ca268eaa31f2a7fad3dcd560ba9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/356807a51563ca268eaa31f2a7fad3dcd560ba9d"}]},"branch":"refs/heads/master"},"124a8176c19885b1577d03982ff825c09769cbc4":{"kind":"REWORK","_number":2,"created":"2019-10-17 16:51:50.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/79/689079/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/689079/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/689079/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/689079/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/689079/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/689079/2"}}},"commit":{"parents":[{"commit":"e51596e97a3ed0fc72c4174c3b7ac485268839c8","subject":"Merge \"update constraint for os-traits to new release 1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e51596e97a3ed0fc72c4174c3b7ac485268839c8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-17 06:12:10.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-17 16:51:23.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b3a409a7b76b1d5ee63149143887168d8db24df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/124a8176c19885b1577d03982ff825c09769cbc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/124a8176c19885b1577d03982ff825c09769cbc4"}]},"branch":"refs/heads/master"},"9f0d9ef478e506334494a906f89bb29ba64f7e59":{"kind":"REWORK","_number":3,"created":"2019-10-18 00:04:16.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/79/689079/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/689079/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/689079/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/689079/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/689079/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/689079/3"}}},"commit":{"parents":[{"commit":"e51596e97a3ed0fc72c4174c3b7ac485268839c8","subject":"Merge \"update constraint for os-traits to new release 1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e51596e97a3ed0fc72c4174c3b7ac485268839c8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-17 06:12:10.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-18 00:03:43.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\ntry holding back networkx\n\nChange-Id: I7b3a409a7b76b1d5ee63149143887168d8db24df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f0d9ef478e506334494a906f89bb29ba64f7e59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f0d9ef478e506334494a906f89bb29ba64f7e59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
