)]}'
{"id":"openstack%2Frequirements~689140","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Icfd235850b47b38a3d8348693ce3ab8dfbe54c8c","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Icfd235850b47b38a3d8348693ce3ab8dfbe54c8c","subject":"update constraint for tripleo-common to new release 11.3.0","status":"MERGED","created":"2019-10-17 12:03:46.000000000","updated":"2019-10-18 05:06:33.000000000","submitted":"2019-10-18 05:06:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689140-1571375193778-283fd268","meta_rev_id":"01244ad2a706f510117f98093e75cf7a4664b730","_number":689140,"virtual_id_number":689140,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 05:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-17 16:52:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-18 02:51:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 02:51:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-18 05:06:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff421eb4fb49cb59fc1f497d524c52c655753431","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-17 12:03:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dea8e881585481e21d41763a8422a778266ec79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 14:08:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/02a81753aa1d4cbf88c70142ae3940a0 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf96cef7bf04eba880c897e0ee0e29d : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ce494b32403462baee0a1cadce55b26 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6af75a197caa4191a6b21f0bc4418f5d : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/adbc835d7ed6438ca19656da23f4ce96 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f235c3b12c804f07be1a23de23769948 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5514709d92743039afe5f0f25ed4637 : SUCCESS in 2m 48s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/95dd8d7df61a42a1a4eb11ec36d4544d : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7aab5351c6844cdbbe1ac098528a15a6 : SUCCESS in 6m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2f7944fa4f6a459cb5576907d1583fa3 : SUCCESS in 8m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cf04682eb2874cddb721c3133c986acc : SUCCESS in 7m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7406f2a3c4834419b3b83410d3dcbb3a : SUCCESS in 4m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0df5d47070fa40169425bed19ed95b6d : SUCCESS in 4m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ac8e9c8f72e349bfb5554f5be5a3e579 : SUCCESS in 11m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/81ce5b9a460a4d9e914f23771b485e07 : SUCCESS in 9m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/552192cf569b47829d8634077c9ade12 : SUCCESS in 8m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a75a5f2998e74d6c81aaffbee8d176ae : SUCCESS in 12m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ee1c40b9274d42a1a4e91a9df0121392 : SUCCESS in 16m 02s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/914878c2074d48d18c3fdf4fdd9dcc1c : SUCCESS in 26m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b8d596ed9a214040a035a23f4d6033d2 : SUCCESS in 21m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/229ae21d18a94742b342d1b1aaabd8b4 : SUCCESS in 12m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1ba30adff6cf4fc4b1e4cece83b4e595 : SUCCESS in 12m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b986ab3698684eda840a50a30d49e59c : SUCCESS in 12m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dda0448b315b44738164c776da30e482 : SUCCESS in 20m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6db1c93f7144ab4b7929fef170caf10 : SUCCESS in 5m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/aac46b564cd0469caa950372984a0a25 : SUCCESS in 9m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/775d8bfe354242cfae5b18f821d58acb : SUCCESS in 9m 14s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/fa2cee93f6164d03ae78d1c757bd925a : SUCCESS in 1h 01m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f13dd5909324e3b8c1b0f144ffb4775 : SUCCESS in 1h 09m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/329cedf20a9741399c283160c4f4c3a2 : SUCCESS in 2h 03m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a44e497ed5494861bc0774cbd95bca8f : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":1},{"id":"31a8f8b6c01ccd9275f17964a46e5625a5e513ac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-17 16:52:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"54ac7a2848678e08d7e45d9bf4ce0dc16bc1ce83","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-18 02:51:01.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"217fd1ae9e9168bae50572d2f0ff985ed3834d54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 02:51:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbb39c0b6286580d454649fe2592941729645d12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 05:06:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9a1ed68487e4692aeac8e0637cc58b1 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6617bb322e924711a00f158717c7202e : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/000ab21d25204fad956643b1b6e67276 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/682cfb17cb004d59b5bf010652b87139 : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c47a8ce6cf2948d5ac10ec33910c331c : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2155447dd0c417d89a85d0f88937778 : SUCCESS in 2m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f087fa2099bc414a959374ac223431b8 : SUCCESS in 6m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5f0627e16dc34865b7984f4262c2ec78 : SUCCESS in 6m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/50abd397b68740028272d9bf2a961d30 : SUCCESS in 8m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5a1730fd4a974168849ae298eff17ffb : SUCCESS in 8m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4a98b3d9507d48a9a48388fb2fd379c4 : SUCCESS in 4m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7cfd73a8474745b88f5b31203016bfc2 : SUCCESS in 4m 05s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a86ececd53e340a2a7790c703fd0427b : SUCCESS in 9m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5b8972a6bebe410ea5479c440f69d0b1 : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53f95be58f76419e9c0f210ff80db2d1 : SUCCESS in 6m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ade812c8b5c7479597f082bef968217e : SUCCESS in 13m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/85866b2f42f349969d6aac4684b0b4f0 : SUCCESS in 13m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7cf0c0cdb8bc481a9f140bd9710f8531 : SUCCESS in 22m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/77ab3b8d3f4945e9954b43acdc9465a7 : SUCCESS in 20m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f5af54ff181840109b3161424c65b1b0 : SUCCESS in 12m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/df5788836a4141f0846f7007f57093b0 : SUCCESS in 12m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0e79c9ff10e744e89671726da6afbe9b : SUCCESS in 11m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a04964232fa4186bf4c5fa925a32ffb : SUCCESS in 18m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/260f7759e96a4feb9d25e1a467489e25 : SUCCESS in 4m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/480278005f3f49f1bb22058f44a13987 : SUCCESS in 8m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d578881881fd4d43b8cec029ffa4e124 : SUCCESS in 8m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/fc5ed068d4d8438a843b46502a4cd812 : SUCCESS in 55m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6abe1e73826649f49a4f21829a7d3ba4 : SUCCESS in 59m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f1d91204a77e465c96fd7819158714b6 : SUCCESS in 1h 36m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9c1caaa239f04eb1b603004053854852 : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c7dc3ff27368211592f15c393ab6922490ff3300","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 05:06:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09d7fe323187645bff016cd64106cdb368def727","revisions":{"09d7fe323187645bff016cd64106cdb368def727":{"kind":"REWORK","_number":1,"created":"2019-10-17 12:03:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/689140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/689140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/689140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/689140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/689140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/689140/1"}}},"commit":{"parents":[{"commit":"5d2dc701ab2fe7e87d78e4d96eb444766eef6725","subject":"update constraint for ceilometer to new release 13.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d2dc701ab2fe7e87d78e4d96eb444766eef6725"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-17 12:03:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-17 12:03:44.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 11.3.0","message":"update constraint for tripleo-common to new release 11.3.0\n\nChange-Id: Icfd235850b47b38a3d8348693ce3ab8dfbe54c8c\nmeta:version: 11.3.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: Ifee25a01b6296c542feebe6d611ec057729f8fbb\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09d7fe323187645bff016cd64106cdb368def727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09d7fe323187645bff016cd64106cdb368def727"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
