)]}'
{"id":"openstack%2Frequirements~689221","triplet_id":"openstack%2Frequirements~master~I001a45de5484e040006e3941a285e99ef5095f4d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I001a45de5484e040006e3941a285e99ef5095f4d","subject":"update constraint for os-traits to new release 1.1.0","status":"MERGED","created":"2019-10-17 17:09:17.000000000","updated":"2019-10-18 00:25:21.000000000","submitted":"2019-10-18 00:23:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689221-1571358188759-9e252c24","meta_rev_id":"026fd20237c39471ee1e06abc8d10f54e1236700","_number":689221,"virtual_id_number":689221,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 00:23:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-17 17:19:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-10-17 17:18:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-17 17:19:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-17 17:18:42.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-10-17 17:19:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-18 00:23:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55729e94980df104c8169a2f30a1d3e00c259268","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-17 17:09:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90641bce8cb7596c58a153e472d83d153f27c313","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-17 17:18:42.000000000","message":"Patch Set 1: Code-Review+1\n\nDo it!","accounts_in_message":[],"_revision_number":1},{"id":"7a6341a5e477aed34b53dd736615a3abb3398a70","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-17 17:19:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"818951ef5417153c2e4ee899ab32a2b200679d9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 19:09:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f12fc2967e774c628ace94a99743b038 : SUCCESS in 3m 32s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6c822df712ad4f75a1fd1168cb7803d1 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f61ed6462b44f72a5be6c944d7c26de : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0fc11f09daf14d2f8284b1e719b66dbf : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eacce6117e534e288005b040fa556031 : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15efc3fb833a4f68b58336baa0984315 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da34c9d39cff4918aa758a7a89b7be6f : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78a2d6db0f064acb80c4094e65d1ceeb : SUCCESS in 12m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/964e9c2b81684fdb9cb862ef38b16222 : SUCCESS in 7m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/49ac7cece73042629177e76119970e63 : SUCCESS in 15m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f025c7f6d5164f74937f56c2ae70275b : SUCCESS in 8m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9d436f9d4f7648c3b3c5b1a7e81c8b67 : SUCCESS in 17m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8b3a882862424900952c24ec24814773 : SUCCESS in 7m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/efadc40f5fa5449b88b1e5ab09a79dde : SUCCESS in 12m 30s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f07774e2a68f48a8b5be24ede81d4944 : SUCCESS in 8m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d0bffe96eb174c5d8f6a65f478efb850 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa05172fe5a94a94af7a6de62f27030f : SUCCESS in 7m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3a056121f06b4c128e8f12714f669e9b : SUCCESS in 11m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b35dcf187515434687564db06f8cfff1 : SUCCESS in 25m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/90cdd3bdbb1b4c6bbb4bb12927f202f4 : SUCCESS in 23m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ba67c5c1448f47099ce583705c1b3874 : SUCCESS in 27m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/136ac962b18e4ad1b3df954771bdf15d : SUCCESS in 17m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2d3e2112af524398a35f1d11544f5382 : FAILURE in 14m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f7acd6edb3f2467ebaf6d5d3fd866c8d : FAILURE in 14m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/85299b92b6304de692b3cbb1b6f27e84 : SUCCESS in 18m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b38f004a69654bd0adc93c5d24ff05b7 : SUCCESS in 5m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/657d1467986045039b93d6cebea012b7 : SUCCESS in 10m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c942328d781e49f0a7cbe64b05db0e6c : SUCCESS in 10m 33s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7aad43ec8fcc41df9853e3a954c1540b : SUCCESS in 1h 00m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8ff3bdfd4c84f5d8559c4d15625ae24 : SUCCESS in 1h 14m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/49deaf76e3a94b658860e0df9959368e : SUCCESS in 1h 53m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5213092c30084f01a4a296fe346522dd : SUCCESS in 1h 02m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5558a292ebd4088bda6fcba8a27f9fbb54563023","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-17 19:35:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ab70ee1acb0be05ce152ce896135971609e12650","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-17 19:44:56.000000000","message":"Patch Set 1:\n\nbah, this appears to have just missed https://review.opendev.org/#/c/689152/ merging in nova :(","accounts_in_message":[],"_revision_number":1},{"id":"43cd4940aae5b67c045d8b72cd5f95f7b5f8597a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 21:44:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/611e975ca3b2481f93a2c6ae00c0cea9 : SUCCESS in 4m 40s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/82eccff294794bb5a48a38af0bd30396 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22203f54f4e04a009d0be7c6a2899590 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8215c8c013346bca7c47264752e9535 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28336d6d67514ae7a56dc4af1ef537a6 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9671acd07f941119b5ac78079c61736 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/905da0bb76f14ebf8b2bf0e407e5f49b : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5b118f9a7cd84bc2a7a5dae7b128ac95 : SUCCESS in 3m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/eafc5ec7a06b4fe7b20dffaa8e26a490 : SUCCESS in 7m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/edd4cec86e274e3db4415e8ed7a6845f : SUCCESS in 7m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0ff057b4b8aa49b9b69f12a0952a93b6 : SUCCESS in 11m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/59fb5ab5e129437ab92734012be27dce : SUCCESS in 11m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/bc1cec1e81214f0abaa12acd3f67f4e8 : SUCCESS in 4m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a45345fc1fae48f19f78fa11e31575ec : SUCCESS in 4m 53s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/274206753e0a43b58a87a89ec700aee2 : SUCCESS in 12m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6ee122275b844b9d8da56aa94da96a0a : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/590a68b1a2b1457a9e130819104da056 : SUCCESS in 8m 17s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f8aded74cea549189c69d33186c79cb6 : SUCCESS in 16m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5eb18bd797124c259feb669fbb5fe2c0 : SUCCESS in 18m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2e302171963a4d2083e96538103ea0e6 : SUCCESS in 35m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/37b93edc8e194e5aa25669317c8a5a45 : SUCCESS in 27m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0a8396a26c81475aa89238772ca0ffd2 : SUCCESS in 19m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0861493f5bef49d5aff2487ab412e61d : SUCCESS in 18m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a449d7644b50440b920e91ca0db02a59 : SUCCESS in 18m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1c2dd5721a47419c88a7cc6e85e2bba4 : SUCCESS in 18m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ff3263f0ce44eef8acfe693e084efeb : SUCCESS in 4m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/89ee0da3a19d4faab139ad6ab0b5a883 : SUCCESS in 12m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/887eb6ea3fc0467399ca9e54442c4ba3 : SUCCESS in 8m 42s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5c58adfdf2ad435b99e0f0268da8c7ad : SUCCESS in 1h 09m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e54b9a888dbe4740b412c81663266ce5 : SUCCESS in 1h 16m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9cb77df16d2e4f6caebbfe341d7ff395 : SUCCESS in 2h 06m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/640451d8379647d8a669da2d09340729 : SUCCESS in 1h 25m 00s","accounts_in_message":[],"_revision_number":1},{"id":"196a40ffe7a86fcddc3d05a3ea4b8c26a7fe3d8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 21:45:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"615363f249fd67b3a6c984a1e8279603b0aeaf64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 00:23:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de20f4d24d7409abccef2fd977b44e4 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/044be9ec84d940e98abc527bbd193f5b : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e9c5927eb764364aca9bfe55ffb01ba : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58cf00e69024484e8a3bff093fda27ba : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86c8eb354a1e4b8fa9b2f591cb1a3f73 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/42f56947c0fb4d738e7630ae8380a6db : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/121243b4ce0d45bf8e4e7f5713d154ee : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2b69eb024b934a038ce8adfaf70b7546 : SUCCESS in 7m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ad7fc86543ed47c391c2ba3ca7090b39 : SUCCESS in 7m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3cf694b7cdac44d2b84ca37d8b6cf06d : SUCCESS in 12m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bc0eb29a0a6147899d1ecfc807b43769 : SUCCESS in 8m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f4a78072cbac47e89c0868deb01e3f99 : SUCCESS in 5m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/08e5e085d7b14e52b7a435abef41fc86 : SUCCESS in 4m 55s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8b3db27bed1c427aaeab7697151084de : SUCCESS in 10m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/87075e123b70476998aa96d1703b2679 : SUCCESS in 7m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2de5aa39f6e439a98cf0b19d38af925 : SUCCESS in 8m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3491d27c73e84ddfaadeb662637dc2b9 : SUCCESS in 17m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c37936b298cb4480a26d6ca01b29eaa1 : SUCCESS in 16m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/05eb2f2bb90849acb27ccccf3dbd9338 : SUCCESS in 20m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eb4c6887a10a47e795ebb22b54a6b31f : SUCCESS in 33m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/aa11a4af676a47eeaa14f036783b7ea5 : SUCCESS in 14m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/da43ebe49f5b40438112e1d2869185ec : SUCCESS in 14m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fc0823f20ef0496fbda2d81ed43eaf1c : SUCCESS in 12m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3811c8230a234fd28668738a44dd6b6d : SUCCESS in 17m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90291250c7344aab9dfcaf2f23a97520 : SUCCESS in 5m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/da16a009c82a46479be754dbcd6bcead : SUCCESS in 8m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/759437505ee14b498c04dd560a32893c : SUCCESS in 9m 31s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/03fc7d9019b2422cbbf2ab1af8292897 : SUCCESS in 1h 00m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9725acfaa7b7447a8dee39149ceb295e : SUCCESS in 1h 08m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f9c5e5de26db4c25b9a5d4f6612aeec9 : SUCCESS in 1h 26m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0d837edb3d6445029347e502ac4fe5b6 : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f437674fc84be002236824aff8f545729e290383","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 00:23:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2216aaa2d74af7f018571f46067fe581445a9d96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 00:25:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a852ee6329a4583a2e22df965d04609 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e865306829841672da3189994a00aa6124678c59","revisions":{"e865306829841672da3189994a00aa6124678c59":{"kind":"REWORK","_number":1,"created":"2019-10-17 17:09:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/21/689221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/21/689221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/689221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/689221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/689221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/21/689221/1"}}},"commit":{"parents":[{"commit":"9806c0bbfda9504e5447d11757ff08bac1d90658","subject":"Merge \"Switch to opensuse-15 nodeset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9806c0bbfda9504e5447d11757ff08bac1d90658"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-17 17:09:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-17 17:09:14.000000000","tz":0},"subject":"update constraint for os-traits to new release 1.1.0","message":"update constraint for os-traits to new release 1.1.0\n\nChange-Id: I001a45de5484e040006e3941a285e99ef5095f4d\nmeta:version: 1.1.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Commit: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Change-Id: I23519891bb68899a641ed5dd5bd74cfc0cdd9758\nmeta:release:Code-Review+1: Chris Dent \u003ccdent@anticdent.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e865306829841672da3189994a00aa6124678c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e865306829841672da3189994a00aa6124678c59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
