)]}'
{"id":"openstack%2Frequirements~689790","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I233ec7b78638c84ab0608da652ca97b14c33f8a3","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I233ec7b78638c84ab0608da652ca97b14c33f8a3","subject":"update constraint for os-net-config to new release 11.3.0","status":"MERGED","created":"2019-10-21 14:30:02.000000000","updated":"2019-10-22 04:30:34.000000000","submitted":"2019-10-22 04:30:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689790-1571718634532-34f7ffe6","meta_rev_id":"d87794b3a9707a410b3dc49874a2752ad77c092b","_number":689790,"virtual_id_number":689790,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 04:30:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-22 02:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-22 02:11:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 02:11:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-22 04:30:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68b80e7c27ff7c366ae5c44b4777d4450d943fac","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-21 14:30:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae0c6ab289a1be1f8501b9cedff96a4b18b05e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 17:04:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0f888107640d41e8be673d9290fad46f : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38fab95a428a469f9fb14094521f1f0a : SUCCESS in 2m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ac12c3f222c428a95de593f6d38b5b5 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/844dad45c62847d0a8e7d75164af749f : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39cfadec1fa7474b8abd2e8c0cce91a4 : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d16ce0f1ea294f07b2530b25ac634e40 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56907ab34f4647ac93c4ad6acffc83b7 : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/96dd52af8f6a4addb1efb026a3cafcc5 : SUCCESS in 7m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3c4279123cc840cba783e577c5b5cb07 : SUCCESS in 7m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c40d1b4a465140c8a40ddc15d21ed70c : SUCCESS in 14m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/30f69ea4db484a9bb5f9935b2e3a7d75 : SUCCESS in 9m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/20f1d11a71254c049a71811f1d2e92f7 : SUCCESS in 5m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1cee0408007244fb98f5666915a73b31 : SUCCESS in 4m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/81b1ed68f1b34ad89c155b8085a6394a : SUCCESS in 12m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/53e17ef643734b8abf78b792f8e53d1c : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a8ec4336811471aa2285d097a4b6b96 : SUCCESS in 7m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/51f88800f2e54572afd43f7bb8830c0f : SUCCESS in 13m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f3149a23c8214df6b07035dbdfd0b498 : SUCCESS in 12m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7aba9867b8014e5d95c80251a3e46f04 : SUCCESS in 23m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e945feec084d498ba1a96e79f173d774 : SUCCESS in 23m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ae0585b287d0424d812ae9fbd666eea3 : SUCCESS in 13m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eaff267b27a245b59f6a7e01aa2331a1 : SUCCESS in 12m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bee62db93c2640fea8c8ebdeb7efc34c : SUCCESS in 16m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dfc8293be5ca4559bb6e87a9766f6fdb : SUCCESS in 19m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ee49aa3df7d4b57a3e3447000fe3ffe : SUCCESS in 5m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/79dd5b76b7dd43418081a3afdaa5f3f5 : SUCCESS in 9m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a6fa5f5714ae4145a7a01a55755f41b6 : SUCCESS in 9m 37s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f478394d179e46eabca7783158864a47 : SUCCESS in 58m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/27572c3dbb004b8fa9dbf2c75f0cf109 : SUCCESS in 1h 06m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5a92fd78b0cb48a4a193edf2c049714b : SUCCESS in 1h 42m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/03f2e73129e84a038e21689664bccd5c : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b49aa402766971db6a54cca3de48a08df059c319","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-22 02:11:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9af448f092c6985b76fdae1b06a534d3e55bb20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 02:12:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27147c07d295d1370d313c9e46d40f592be7bc9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 04:30:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7dba0c6e63666df0d23029e1165f9859f241f557","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 04:30:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d48ca49897f8431095432fdf04bd4b80 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eca6669b68d648c8a916662228a03818 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9a4570494c942d6b36589d64d630350 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/812c35f812904bc680e77ca07b063eda : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0cb9e5cc4ea34e2686ff9cc9f88d55ef : SUCCESS in 15m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6836bd3c427d42059789f44335179dcb : SUCCESS in 11m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0178a8f012ff4dbab4ecd9a23ceb70cd : SUCCESS in 7m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9de36e34a628425096f0f11f740c59e0 : SUCCESS in 8m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/073a7e04399d4c14a7bfb6a3ccb9df0d : SUCCESS in 9m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/031237949e6e40c08593cf8b4f9f9913 : SUCCESS in 11m 12s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7a8e771fb9574714a19a4c3cd9cc786f : SUCCESS in 5m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/594279a23bf84329bf27175646e5b094 : SUCCESS in 5m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/82011244320549f487a6bbf062759bf0 : SUCCESS in 9m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7b039da7985049318977b0236ed5c45d : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a42fe5a013b4a96a370932891fed32f : SUCCESS in 8m 21s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/232feb5145b64c3388741bc339aa34c7 : SUCCESS in 20m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/97560820e82c4f07b0348cf6d52455fa : SUCCESS in 27m 20s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/11bc8f567c444fdebd8f3665112f7164 : SUCCESS in 22m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3ba98624023541a7aecae6be0171ad86 : SUCCESS in 19m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d3b330a0b6a54ce58597edd9314cc3b2 : SUCCESS in 13m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dac80dd1be254d498082e786f7f1e144 : SUCCESS in 12m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d508f93cf6ed4d759c7702e28f02f7a0 : SUCCESS in 18m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9cf6a683a354c6ca6c10b1e99584989 : SUCCESS in 18m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6485a3db2e404774a740d66f99d327f0 : SUCCESS in 6m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ae6bc8c02b424c80b256b9eb6f1e498b : SUCCESS in 12m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9dd6c97cd37b46d1bfeed9397d829dde : SUCCESS in 11m 45s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5a1af34c932842b1b0461fdfe82d08c9 : SUCCESS in 53m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/02d3f5c6c86f47a9b5e647482d19407e : SUCCESS in 53m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/50a419c414f84e7d8f73a45887caeddf : SUCCESS in 2h 05m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/aea4e09b24cd4e2ca2b432f91fa0b486 : SUCCESS in 50m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bce7a8492d1a80dd7dd521277178378f75730a9d","revisions":{"bce7a8492d1a80dd7dd521277178378f75730a9d":{"kind":"REWORK","_number":1,"created":"2019-10-21 14:30:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/689790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/689790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/689790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/689790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/689790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/689790/1"}}},"commit":{"parents":[{"commit":"41f6fb2ddf26179c54f6f2c21b88f0e58ab59872","subject":"Merge \"update constraint for tripleo-common to new release 11.3.0\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41f6fb2ddf26179c54f6f2c21b88f0e58ab59872"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-21 14:30:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-21 14:30:00.000000000","tz":0},"subject":"update constraint for os-net-config to new release 11.3.0","message":"update constraint for os-net-config to new release 11.3.0\n\nChange-Id: I233ec7b78638c84ab0608da652ca97b14c33f8a3\nmeta:version: 11.3.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: I240496b0032577d8dc97837dba7bc43a4f5cc2d3\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bce7a8492d1a80dd7dd521277178378f75730a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bce7a8492d1a80dd7dd521277178378f75730a9d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
