)]}'
{"id":"openstack%2Frequirements~689794","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I727cdb0d87ecb2f2fbde8ad2f59453f352f1a30f","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I727cdb0d87ecb2f2fbde8ad2f59453f352f1a30f","subject":"update constraint for os-refresh-config to new release 10.4.0","status":"MERGED","created":"2019-10-21 14:32:33.000000000","updated":"2019-10-22 08:31:06.000000000","submitted":"2019-10-22 08:31:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689794-1571733066750-34f7ffe6","meta_rev_id":"6b6984c6a454f64a34e7c140f04ed50885d65616","_number":689794,"virtual_id_number":689794,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 08:31:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-22 04:25:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-22 04:25:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 04:25:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-22 08:31:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2daf2273f6e8d4447f37a54dbd157fe5a5c42b05","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-21 14:32:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a042144e7aa429c0c49ae21529f3cd941fac7f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 17:16:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a0d8616a8fa7466c87d65b4f61dd3dc4 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b6423314f874b69805512eaf38ac456 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84390d5f0939419ebd93cc79af1eaf2d : POST_FAILURE in 14m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97693c24df5648a5b15088bee65fd09c : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b8ecd938b9143bdbad237c0b397a99c : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4edaca98a17b4ed595a1dc36c0327129 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12931dcba8984c318db0ba56522cdda7 : SUCCESS in 4m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7ccaeea336224e58805b1e10a8074847 : TIMED_OUT in 34m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/04e01853fa694298be1440d90a4b73cd : SUCCESS in 6m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c4b7472d46b8468da19f2cc46b78cebf : SUCCESS in 10m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dafc861314b6411b871752a3226b7f18 : SUCCESS in 9m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5d507276ae0342d5a8b2e766ec604828 : SUCCESS in 4m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9d5777f4c60d4c2896d2244dd865418c : SUCCESS in 5m 10s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/029a2ee6834d48e7ac80f2de02a0fec3 : SUCCESS in 8m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7dc173cb648148a3a329f28d578a2d95 : SUCCESS in 8m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/058e5dc7932f488fa95c5b8f7531d37b : SUCCESS in 8m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/292baebf9dca429aa0111e3a0bd55947 : SUCCESS in 24m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4667ddda675f4927a7cf89d7b2891686 : SUCCESS in 22m 29s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/490ab858e88e420a9cdf06d8db3b2103 : SUCCESS in 20m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8ff71c9c555d494e8bd511fc53a6eeb6 : SUCCESS in 34m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9fc67ea917184bf7a8f88d0be6bb1166 : SUCCESS in 20m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8dd1316daa74468995b4fe26b9e62e2a : SUCCESS in 16m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b458e6607d684a4eaf562ff017264bd1 : SUCCESS in 15m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd6afd0b25c143209a60c3d54ad167c9 : SUCCESS in 19m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ca649da4aa24dcba9bb28ddb40ba8b7 : SUCCESS in 5m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d7636fe011184da1a438e131972e7445 : SUCCESS in 9m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1d369ce4bd30495284d315a5c516da61 : SUCCESS in 9m 52s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4358e010978e4ed288bd7cb63519109b : SUCCESS in 1h 01m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a210576d463440de886fc9005047bf5d : FAILURE in 1h 21m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5ab248a8f8204cfba2fde2dfe760c9b3 : SUCCESS in 1h 43m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6985d92c622743cfa2f7adc371dd9470 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ae101ef6f1bee22ced1ebb03db137b865eaa6324","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-22 02:11:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a8eac0292d00b1ccc53ce885c5c6db1e312322c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 04:20:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e697dc75c59e4b719cab74367c20780d : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ca9de90b4448ec93c743ec233282f3 : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aea6cd493592450bb5d82b78afc242fa : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/623caabcb86d40babb4486474e085fd1 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2dc33bfbe52d41e780cdb925f3e72a45 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73054d4d30dc43e4ae9bdce524b0d375 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e058fbe3eb0d467fba3654dfaac5d87a : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4be727b06b484b5fa262472b41add3f6 : SUCCESS in 7m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/59c7a954ab984d29abbe8556da4110fc : SUCCESS in 6m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b254e941afbd42ae9ea9fae7eec3a767 : SUCCESS in 11m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0fc3333bb52e46f6bf5aee3f05e0567c : SUCCESS in 7m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/de048b7a8d894539835fb65c52c75737 : SUCCESS in 5m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2d0b9c93ed8d44ab80cc623e0250a7a2 : SUCCESS in 4m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/79187b3798bd48bcbe80f12ed71309cf : SUCCESS in 13m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1cf83c39a17240278c3e8a529709005c : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c5e0d08dd794ac49cfe17bf08f93ed5 : SUCCESS in 6m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7374de9c3c3a4dddabfa5015e89d8c0e : SUCCESS in 19m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0714ed67bc494575a213b334e7f01f2c : SUCCESS in 19m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e50ed194daf0485887d0bcf4205b1fe0 : SUCCESS in 31m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/afe90dd52d004bb397cfb527f22b1720 : SUCCESS in 29m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/11972cf0671648d09ed023c7f8780958 : SUCCESS in 13m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/02418c85b4eb4f01b3a240752b25397c : SUCCESS in 15m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/84e7e64e999c4b49b41f15fc2ff75d4f : SUCCESS in 14m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4993219f065a41d5821d4ff6016bd738 : SUCCESS in 22m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/28b6868157944c2e824a8cd6a12618f3 : SUCCESS in 5m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e3ac87a497624eaca99db1227357e46f : SUCCESS in 9m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c8ea9f8d9e73413d9673d963acef2173 : SUCCESS in 9m 26s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5c0fd5cccfda4bc19e5b36a2a7e1a06b : FAILURE in 31m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cafa71f244c84fa1abb1b8564c8c48fe : SUCCESS in 1h 23m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b78239e0d74d468e9b31f0288b0ead72 : SUCCESS in 2h 05m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e27243ec02a2461ba138a7e877c8c060 : SUCCESS in 1h 22m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9d01ecd1a94558032cad202cdb2931856c067619","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-22 04:25:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"db59583c411983d870adae54d7b8012afe576a46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 06:30:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/fcba658bd5e14e8188da2a637ecf2c9e : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e091460c8144d9872cd727795cfbfd : SUCCESS in 2m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bee82d8b7368425abbdb9e3b90ce0a04 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0acbb309daa74b808c8d733902c98302 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/753d50ed09a649a982e61ccb43d218f9 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f3d15c893c1417eafd9fc20b0d2ed50 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ea2a1079fdff416db96846a5b96c0411 : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1dd86703ae394cb19ce9508fa0585231 : SUCCESS in 5m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/28da59bdb9d143459571aa780eccaf36 : SUCCESS in 5m 47s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9844d690c4e44355a233422406abedb1 : SUCCESS in 7m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/96fbb71299a14fd6bf4edef7f15b6459 : SUCCESS in 8m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6c749a6085d54d67aa935f49092dde34 : SUCCESS in 4m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7fca5ead5b4c4c83bdbb57ac723ec061 : SUCCESS in 4m 24s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e79711e5dcfd4c0293672ef35f50be85 : SUCCESS in 10m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/456023d45d9548128357fc45e7e21d02 : SUCCESS in 8m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8054755a13714ceea834c5fb0b5dbc84 : SUCCESS in 6m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f3830f70428d4c44b6a1ee8d8b463c3a : SUCCESS in 19m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ed0f7c72fec549eda978fbf934af116f : SUCCESS in 23m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7a5fd3b95801401db7506018bf33633e : SUCCESS in 27m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/820768cc54344ade98215c7c12006dfb : SUCCESS in 22m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/37f99cb7d4cb4dcd947e1fcd404cb420 : SUCCESS in 16m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/baad24b17af24dfbadf29fcf245cd11d : SUCCESS in 11m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a16bbc895ecf4b4bafe38545ef16b194 : SUCCESS in 12m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/600909933d6741c290f791d028d372ee : SUCCESS in 18m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4415f516aa24b9591ff277afe841ce5 : SUCCESS in 4m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0b0771d03dcd4be98a665dd5186f4d3e : SUCCESS in 9m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b1db8aec84904e079fa2c5f0f8cbb7cb : SUCCESS in 9m 06s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/fefbbf6a48884b9195c06e13ed157280 : SUCCESS in 1h 04m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9a1754fbbb24bf1b515e16e442bc74d : SUCCESS in 1h 11m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/76c6d4ae78624d8387c4264665731d7d : SUCCESS in 2h 01m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/36545af66f024a10afe05dba5cc11f9a : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7d0edcfe211d01089a1d6898e3a0d04023fbc604","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 06:30:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b5e0cd6ed32b18162b7504ebdc97847c42e8fc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 08:31:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/194704641283413481666185f78014fa : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a4cfb6766be4b2fb552942601e90e2d : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0acc23d15e5f46e1a507f16c6364a394 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ebd83a2519d4ca99b23eab8cbf6b2cd : SUCCESS in 6m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c7bdef4009c4265b4ae912fddad713f : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78e7a39367614cf18e41138399104072 : SUCCESS in 4m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c4dd4b2a4a2c4048b96c1decb5b7a7cc : SUCCESS in 27m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/110dcf81c4334992a76f11eb8f3a93ea : SUCCESS in 6m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e6cff6d12e6c4fe8ba38d6487102ae5a : SUCCESS in 9m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7b86b42b02804f879ac1a962de323375 : SUCCESS in 9m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/811ec9f069a8471b966f45a3998e6407 : SUCCESS in 4m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/617350b440854c429fec50f5f6beddef : SUCCESS in 5m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7001fd1ce30f4ff08e4e0ccd6a794a9a : SUCCESS in 12m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cbcbe8fa58ee444da0473a4c67d73add : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a543852420f7457f835ed6eec5cf3e37 : SUCCESS in 14m 01s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/44e793d739904ca5839477d032eb16b0 : SUCCESS in 21m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aba5ebd980e24bb7b9c28ade0f3c9319 : SUCCESS in 30m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a06b0402ca854d95934ceecdd0d20a98 : SUCCESS in 23m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6fedf7d9653c46a196ca34e2f80c74eb : SUCCESS in 23m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/af77fc838f7b46a2900662f2e95c970e : SUCCESS in 15m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a77276be437d4f3fb241f997a93587e3 : SUCCESS in 14m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/17c3a9b3f4b540edab6c84bcac66fb13 : SUCCESS in 16m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/51adbf034217485080eda1b73da40c8c : SUCCESS in 18m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5207654caf8b4935be473f51830e0a03 : SUCCESS in 4m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1f97ed9468f94c72a2d8407513e2fb90 : SUCCESS in 16m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e1292dedd6514e3c9ae7d5acde464855 : SUCCESS in 11m 42s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/fc4a04a53f074560a4d7d1fcd5433b8e : SUCCESS in 1h 28m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0dfd78b818e476e894589c0bb930fe3 : SUCCESS in 1h 04m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f0586a85aa46407096e3b0c5db058fe8 : SUCCESS in 1h 57m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/35c25a20d0bd465f80fbe98ed08be05a : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d78be7a42b2a4fa454d6c203bb40d20a813ecec6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 08:31:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ceff2a7e2ef439ab63cec80afb059b275d05cfbc","revisions":{"ceff2a7e2ef439ab63cec80afb059b275d05cfbc":{"kind":"REWORK","_number":1,"created":"2019-10-21 14:32:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/689794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/689794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/689794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/689794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/689794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/689794/1"}}},"commit":{"parents":[{"commit":"41f6fb2ddf26179c54f6f2c21b88f0e58ab59872","subject":"Merge \"update constraint for tripleo-common to new release 11.3.0\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41f6fb2ddf26179c54f6f2c21b88f0e58ab59872"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-21 14:32:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-21 14:32:31.000000000","tz":0},"subject":"update constraint for os-refresh-config to new release 10.4.0","message":"update constraint for os-refresh-config to new release 10.4.0\n\nChange-Id: I727cdb0d87ecb2f2fbde8ad2f59453f352f1a30f\nmeta:version: 10.4.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: I240496b0032577d8dc97837dba7bc43a4f5cc2d3\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ceff2a7e2ef439ab63cec80afb059b275d05cfbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ceff2a7e2ef439ab63cec80afb059b275d05cfbc"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
