)]}'
{"id":"openstack%2Frequirements~690097","triplet_id":"openstack%2Frequirements~master~I413265a2ef080304723e1a56a031c29a60d1a707","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I413265a2ef080304723e1a56a031c29a60d1a707","subject":"update constraint for python-novaclient to new release 16.0.0","status":"MERGED","created":"2019-10-22 14:57:42.000000000","updated":"2019-11-06 11:39:52.000000000","submitted":"2019-11-05 02:06:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690097-1572919617557-752aeac9","meta_rev_id":"6216c051f6db4bd9a9db3e44f95bb33fc1a4ecab","_number":690097,"virtual_id_number":690097,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-11-04 22:18:02.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-05 02:06:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-29 09:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-10-29 09:31:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-30 14:53:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 16:57:46.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-10-29 05:52:14.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-10-29 06:27:04.000000000","updated_by":{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},"reviewer":{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},"state":"REVIEWER"},{"updated":"2019-10-30 14:53:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-04 22:18:02.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-11-05 02:06:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a2bb271248102f2aa98418bee4e1afc17450aca","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-22 14:57:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6845dc018904599ed5081584207ba6411e24154d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-22 14:58:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ab5b709a4c07878887b15c68a63b5f61b64e158d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-22 15:59:33.000000000","message":"Patch Set 1: Code-Review-1\n\nThe cinder job is failing but it looks unrelated.\n\nThe horizon job failure is definitely related though:\n\nhttps://storage.bhs1.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_2e3/690097/1/check/cross-horizon-py36/2e36786/job-output.txt\n\n2019-10-22 15:06:33.728878 | ubuntu-bionic |   File \"/home/zuul/src/opendev.org/openstack/horizon/openstack_dashboard/test/urls.py\", line 30, in \u003cmodule\u003e\n2019-10-22 15:06:33.728967 | ubuntu-bionic |     from openstack_dashboard.api import rest\n2019-10-22 15:06:33.729140 | ubuntu-bionic |   File \"/home/zuul/src/opendev.org/openstack/horizon/openstack_dashboard/api/__init__.py\", line 38, in \u003cmodule\u003e\n2019-10-22 15:06:33.729229 | ubuntu-bionic |     from openstack_dashboard.api import network\n2019-10-22 15:06:33.729399 | ubuntu-bionic |   File \"/home/zuul/src/opendev.org/openstack/horizon/openstack_dashboard/api/network.py\", line 23, in \u003cmodule\u003e\n2019-10-22 15:06:33.729488 | ubuntu-bionic |     from openstack_dashboard.api import neutron\n2019-10-22 15:06:33.729658 | ubuntu-bionic |   File \"/home/zuul/src/opendev.org/openstack/horizon/openstack_dashboard/api/neutron.py\", line 39, in \u003cmodule\u003e\n2019-10-22 15:06:33.729743 | ubuntu-bionic |     from openstack_dashboard.api import nova\n2019-10-22 15:06:33.729909 | ubuntu-bionic |   File \"/home/zuul/src/opendev.org/openstack/horizon/openstack_dashboard/api/nova.py\", line 33, in \u003cmodule\u003e\n2019-10-22 15:06:33.730026 | ubuntu-bionic |     from novaclient.v2 import list_extensions as nova_list_extensions\n2019-10-22 15:06:33.730119 | ubuntu-bionic | ImportError: cannot import name \u0027list_extensions\u0027\n\nThe nova functional job failures also look unrelated (lots of crazy sqlalchemy failures).","accounts_in_message":[],"_revision_number":1},{"id":"c72e0b84830e4f4521f6d7ca1195c2171299f99b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-22 16:01:50.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"fb40fd4f3c89328d24a67ec9bb4977f95793d21f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-22 16:02:20.000000000","message":"Patch Set 1:\n\nHorizon bug reported:\n\nhttps://bugs.launchpad.net/horizon/+bug/1849351","accounts_in_message":[],"_revision_number":1},{"id":"26f35eb20db78c9e283a98ba3d69a307dc24c229","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 16:24:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4f9a8fedc124ba099fcfaa818f74f43 : SUCCESS in 3m 12s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/99ba993327564b5d8cbc7c1ad71aebb1 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/500b0c7c4c424ef28e5cd57236a3b811 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9cb8b0965bda4b7fb42896ce0da4e33a : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e7a350a6c6f4ce6b323c21cedca5b3e : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e4231297f2d44b8bdde79e865efb1a5 : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74d29f2ee76642e9b31bf80bfc69ddf1 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7dbb87b2698146ce8761e2b208725464 : SUCCESS in 4m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/26d40685e28f4d2398d38ea0758d13ec : SUCCESS in 14m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0b75ed12940e4fbdbfc114e267840522 : SUCCESS in 6m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ce01fa7ec46c49acb01bb4acedae179e : SUCCESS in 8m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/563bd30ac18241009be2e7d8993b5417 : FAILURE in 18m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ea5665a902ad4ffa94b66834d86f616a : SUCCESS in 6m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/148130e924754d36ad7f45e293bb9613 : SUCCESS in 4m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d0282efc85ba48a0903122080f6d6f39 : FAILURE in 2m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2e36786d54f040e08a6e9d1f7a698b54 : FAILURE in 4m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c5d870fccb82444d92a85f440db8e912 : SUCCESS in 7m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2acc25973e594c98a73fb763feae5e0f : SUCCESS in 14m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/32193a98503a47f4a379f85d059a9eef : SUCCESS in 22m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3bd8dc3299c541179f0bf177de6b0c11 : SUCCESS in 25m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5a4c1ce309ae48a08adda6a27cd333c4 : SUCCESS in 22m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a229aad9d32d4ef79e26bfb131b1d1d3 : SUCCESS in 21m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f081c72262754f1f9426fb6159ed9fe8 : SUCCESS in 13m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/383bcd55956242d79b430a82620d3547 : SUCCESS in 14m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/831ac1258ef848068b856ff11b02551e : FAILURE in 26m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eb9b0c918ccd422a8fbfd630077990bf : SUCCESS in 5m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8f9adf43d354476bad4e37ccb8c3c953 : SUCCESS in 8m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8c4627cf7dd34706977c827f0c80a40d : FAILURE in 3m 18s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7fff6965c820464e8c31a194a4656c87 : SUCCESS in 59m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdeaacd1a0574198b71f9a8e8888fc69 : SUCCESS in 1h 10m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b9d384d8b3834041a88257a6c1eb35be : FAILURE in 17m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/18e8dff21bd34530a3dca9a7d1864224 : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":1},{"id":"391eae4b9cf95bc68e9d2164142eb903a561c7b2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-22 17:18:31.000000000","message":"Patch Set 1:\n\nWe\u0027ll see if this works:\n\nhttps://review.opendev.org/#/c/690134/\n\nIf so, I\u0027ll update this patch to depend on it.","accounts_in_message":[],"_revision_number":1},{"id":"5211f88a0abe12df3378ab1025ac6560e7c7fe0a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-23 01:11:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0b2db95f53d136adb0e54d4fa8b51ad5974f81fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 02:44:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa8b3f6e20b40b4b54b0916b7c19cae : SUCCESS in 3m 38s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b9fc09cc8462475cb69ef0b1a38d4774 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e992213337e4d2e9d8313d01dfc9d2e : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c47be2f2f9954f8aa9ff842d6afffad6 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb4e911033da47748df94f17b983e19e : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2affb9e592204dc59689722929f826de : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/acd20819bafc4759aa8194e00d46363b : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db50ccee49f344fabcfaedb9bc5fb1ae : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8dfd547226324cf39b3e4691242a11b4 : SUCCESS in 7m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c4d0488e285648caba0563f0c0e992f0 : SUCCESS in 6m 38s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a2a91f1900db4cbf8060adb14075be23 : SUCCESS in 10m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7ed32b73357f47238cdf37650bce1b3e : SUCCESS in 9m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/97daafacbeca4d4d85ef0ded10c9624d : SUCCESS in 5m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9b4f02fe17954b18802654b831d87574 : SUCCESS in 5m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/be02bc34b22448a2a0adf1ca6b6b41b6 : SUCCESS in 10m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bd214813174f457c8b137d667469ee25 : SUCCESS in 8m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/138afe952582445995b52acf0bfa87fd : SUCCESS in 7m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4d9a3ede3f7f46038a00056c20348a63 : SUCCESS in 14m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0ef09c7de3d54e1a868e5092666f11e3 : SUCCESS in 14m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/290ba49e6e094db8a30aaaab8f0368be : SUCCESS in 28m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f0049e53aff84988ac500b424efeb38e : SUCCESS in 21m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/fce9cd47400b4b20a631beec8477adf5 : SUCCESS in 16m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9d0a70924f3943d2be6ed3a6a0570806 : SUCCESS in 18m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f639170fbdb045ad9585e989cc068f09 : SUCCESS in 12m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c335f196ed4e4f4cbc80c0cda616c904 : SUCCESS in 21m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96f74e829a954084aae451fbb25b3d68 : SUCCESS in 6m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/27b8dc54386b4464a56be9e6bb898597 : SUCCESS in 12m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/690e71812653469886f9a47af258a355 : SUCCESS in 10m 38s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f9136429022848c79ac719bea052381d : SUCCESS in 59m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/708b58ea5bf441328899cf1afeb9589d : SUCCESS in 1h 02m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/76444a638547425db835b6e5f64785e8 : SUCCESS in 1h 31m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e9d685c10e0d4f41a524eff940d0441d : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":2},{"id":"6210590f02e7e203c6d521355be673584ccc2091","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-23 18:17:52.000000000","message":"Patch Set 2: Code-Review+1\n\n\\o/","accounts_in_message":[],"_revision_number":2},{"id":"80cf41a5909f0cfb5c11291afa0150ed0d25d25f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-23 20:31:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3a91caefb4cead0b262aefb64c86b047a4ba27b8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-24 15:31:30.000000000","message":"Patch Set 2:\n\nFYI this is likely going to sit awhile because the dependent horizon fix is going to be a bit of a mess.","accounts_in_message":[],"_revision_number":2},{"id":"6d6bae1037cdbbabe5aa3d1d4f6baab006fc003e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-24 15:48:07.000000000","message":"Patch Set 2: Code-Review-1\n\nThis is also going to break python-openstackclient:\n\nhttps://github.com/openstack/python-openstackclient/blob/45af14ca72b44c03f14189ac61a29253faf03bd7/openstackclient/common/extension.py#L94\n\nIt will make this:\n\n  openstack extension list --compute\n\nresult in a warning:\n\n  \"Extensions list not supported by Compute API\"","accounts_in_message":[],"_revision_number":2},{"id":"4f4d0dec89ada505096ec73d546376e8923c0875","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-24 15:55:11.000000000","message":"Patch Set 2:\n\nosc bug reported: https://storyboard.openstack.org/#!/story/2006769","accounts_in_message":[],"_revision_number":2},{"id":"6e6a49ef236feac7f0b855ad5660368ec65cd9c0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-24 16:57:46.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"c9007f140b900b10fcbd0fb447e108efb0443aa6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-24 18:35:36.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fe9eddb9983c82581314033c9b25c1b5598a4317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:18:57.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"ce14c42caab3755f8a5ca271ee804d8c67445149","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-24 19:25:16.000000000","message":"Patch Set 3:\n\nrecheck horizon dep updated","accounts_in_message":[],"_revision_number":3},{"id":"056f889cd34237cf7ae8da41bccf646702ca6549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 21:04:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336f6dcf166746889f40d7a3b2de05fe : SUCCESS in 6m 12s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/52fdb8e8aa1a4c12b6a058ddb84836f7 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/296008bc933548ee878460001b24d68a : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/222386483fc346079b9a4e678dd2618d : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31f2ed3a5de1470d92972bff38d8287d : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0d45c9a9f0c4de79c708962d6bacd39 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75319bc8ba4840348f9d28cb9669933f : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/710f9ebc82494492944db21c5698ae1b : SUCCESS in 5m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2f64acd4376243078659160f97ed23e3 : SUCCESS in 9m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c4f2af09c465414dad51b9b3ce45db86 : SUCCESS in 8m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c33af513e1b94b268a8cdbf8816cbe80 : SUCCESS in 11m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/16556125e0b749b580a7a4e3e99f8e3b : SUCCESS in 11m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b1182f086da34edba93f85c70981749d : SUCCESS in 6m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4a9ed169f32840749dc87684afd2622c : SUCCESS in 5m 37s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a28ecf73d08841b283e2b964627f34e6 : SUCCESS in 9m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/995326ddfd6740f6832fa1ebe18023d8 : SUCCESS in 8m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/70083d2023e046368b26255f086986be : SUCCESS in 10m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/be4cff99da2e4c718845ffea16122f30 : SUCCESS in 14m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6f3a4b3bf7ec494198074e9623262a6c : SUCCESS in 15m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/47ddcaa83fa84596bcebb68f5c135d35 : SUCCESS in 25m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4e6c2ce9f8b346409277af06d1855a9b : SUCCESS in 25m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b4989661001142f8b30b110af332c78c : SUCCESS in 16m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7b5c14b86e704940972c49d0e49b08ce : SUCCESS in 13m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8cc4ef7f389c40b19e043a2c418f411a : SUCCESS in 12m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0b0d9e9358d44f79a117bf8a775ab01f : SUCCESS in 18m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8789bc4b17994be2913df57ea628558b : SUCCESS in 4m 59s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/685fca96a84d49e5900b1824d071c1ce : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9fb12d20816144bcb538fd91ca0e027e : SUCCESS in 5m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0045ad707e304be9bad84317db61aa79 : SUCCESS in 10m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2f880870ca0e40d0a95e8017db70e6db : SUCCESS in 10m 11s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e11f1bdb40224ed9a846e7e375c25435 : SUCCESS in 57m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2ba2e69fcfc414bbc66eba5ecb106ef : SUCCESS in 1h 09m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0f72b834432f46d4988ab4ecfa5751d7 : SUCCESS in 1h 30m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b26f8ea44bf94be698bb5dcadf5d6108 : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":3},{"id":"7ab3900e981f212f11a16e1019ac6861dac624b4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-24 21:33:31.000000000","message":"Patch Set 3: Code-Review+2\n\nguess we need to wait on the reviews now","accounts_in_message":[],"_revision_number":3},{"id":"0c7b5004f7a46f7e607edb4243dbc10b01f4a5d9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-29 05:52:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d4329b8ed514ee9d5a7c585c2fed027e0aadf20c","author":{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},"date":"2019-10-29 06:27:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"635b1777460c5e968849802d93672e1388f1be0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 08:18:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fc927ef79ac49f793ebe04494edd254 : SUCCESS in 4m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c415a01fe62447bf91fe01f37faab252 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72471bd833684603a01708f2063bf25f : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51610097a7f14e7cbab1f5f9d5bc6cd9 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f47f8d15da64cc49571b5aa179c45d9 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56a5214ebc594a5e943feff88f72b94f : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6abfff4376ad44f9993b737cf6acde06 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d56da82e669401e99811725a181012d : SUCCESS in 3m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c8b4a9f3f0f4426fb81f4cda615be168 : SUCCESS in 7m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f845037c6fb1438a815a70fd027b6154 : SUCCESS in 9m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e4d24ef6be15465491194296bc7a42fc : SUCCESS in 11m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0001c23acb5f4696bab5c926952b48dc : SUCCESS in 11m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8f10425c5d314fb38d48b4ee2b2c9768 : SUCCESS in 5m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/692df7dc5c2b4a079afa50547a2229f5 : SUCCESS in 6m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/499df58e732a4210b06f98ee57189df7 : SUCCESS in 10m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7ff28cc72516442389058e4422a5ecfc : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d96da4929c4c4f709a59913cfa4fd709 : SUCCESS in 7m 21s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ea31d44d9b3645a483fcd26090af30aa : SUCCESS in 19m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8b5e33ba40ff412f8b50766d4be3e9e3 : SUCCESS in 13m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e8327a6684bc488dbd8ff967b8dea410 : SUCCESS in 22m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5f925fb9749e41ce9782c95440c1b669 : SUCCESS in 20m 47s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e3025addfca9420982cb82443a64525d : SUCCESS in 14m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/287822b4fc3140c4b7cd41e4e0e4e2ba : SUCCESS in 15m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1c7c6221bee54ad3a2431be17bad6b67 : SUCCESS in 16m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/627a1e278f054f43b98e152976378a9b : SUCCESS in 19m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/960c80728bee47b29096c990c9b3cde0 : SUCCESS in 5m 57s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4016160d0be140a1b0bfc0125a7b6fde : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68c30120e17e4b74a51a5f51a560945f : SUCCESS in 7m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a3546bd718f443f9aa0f984e330ebed4 : SUCCESS in 10m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f516835afd55498c94c49d269e12fa73 : SUCCESS in 8m 54s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/685e93edef564a218dc63dbc8b42f098 : SUCCESS in 1h 10m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ad6de4e7c254dee974ae5e439bbe0b3 : SUCCESS in 1h 08m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1cd06a1aa2014371a7feb7cdecf07f09 : SUCCESS in 1h 44m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a08dab913a2d4aa3868786938c2cfe78 : SUCCESS in 58m 43s","accounts_in_message":[],"_revision_number":3},{"id":"59a2efeb321d1c1f57b9f986d62ddeaddf5acbc0","author":{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},"date":"2019-10-29 09:31:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c119ca59a11da1861db2d535b0c4d9368eb7212f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 11:05:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/061e88b73cc042e7a212159ea0e94c3b : SUCCESS in 3m 26s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/16ab0dfde12f4b8c83a88ea8cab6f744 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e087fa68d23143d2a2be6ec3094d2d2a : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b6ca1e8a2c24cd08b37576a4168e188 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eb742cfe07142c48ae1c3f1a0aa32fe : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e8259b95c4a4c33aeb93e6144dde030 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b9c73d00b014505a7465e7645f0f897 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d106adffc4aa4f4db778977aa2f00699 : SUCCESS in 3m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c24d9f3ffd874853914d7fe6f6a6ed7e : SUCCESS in 6m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8f65d3f132514c998859fb3fccd0b6d5 : SUCCESS in 6m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a5c2dcb97586415c8a68ee137b53190f : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/16338cce0a2b46db9276c524408b737b : SUCCESS in 8m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3e3b79dfa9354dd88d26b31b3af689b6 : SUCCESS in 4m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7d7e7593aa644eb087efc041e991a56a : SUCCESS in 4m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fcd3cf737026437abdbf23928d3c1d77 : SUCCESS in 9m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cac2e7c9165b4611b7372b22c0e44b80 : SUCCESS in 8m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/248503fe21bc46268532f6542f48f939 : SUCCESS in 6m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/63aedff008a54d128fb8689dd9e2dc11 : SUCCESS in 12m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28b838df70354a4e94dd176b147fc68d : SUCCESS in 18m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f41bc06eab054d1eb1de1695db0cb553 : SUCCESS in 18m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c11eb2ad45ee4a35a8df8ca0f45df62e : SUCCESS in 18m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9208000740a346cf9a8ecbe1d728eaf6 : SUCCESS in 13m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/85f7d4d58eca4824adf25e99438951d4 : SUCCESS in 11m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/200f0e74edd74e7da2ed336592346373 : SUCCESS in 13m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95c527d0e4c540b789c2083d993e2a72 : SUCCESS in 18m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/72b2c93e5116470a98b88861752ba605 : SUCCESS in 4m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8638cdc2d71e41aaa086db91b1400494 : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/28516641e0bc4402aa238161dd613562 : SUCCESS in 5m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/30855aec1ca8496399f7f041552689ab : SUCCESS in 8m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6150c61d41b34cdc8434cdd7ffec2c71 : SUCCESS in 8m 59s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1491cb1bd0bd4be0bd129a592847c14d : SUCCESS in 1h 14m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/badfa61534044726940a531e43d06ce1 : SUCCESS in 1h 05m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b554b74c6dae450db4e7aaa25fef0a36 : SUCCESS in 1h 32m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2701f5fe8dc14bf98f20881d7a9c55d2 : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":4},{"id":"86f6f16f99a9e038123e252426728113e21e58e4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-29 14:56:52.000000000","message":"Patch Set 4:\n\nlooks like it still needs openstackclient to be merged (and maybe released?).","accounts_in_message":[],"_revision_number":4},{"id":"2b19c6a0f8e17a5c4c2ffa2d26b3ef0b7e88785d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-29 18:40:27.000000000","message":"Patch Set 4:\n\n\u003e looks like it still needs openstackclient to be merged (and maybe released?).\n\nYes on the merge, I\u0027m not sure about the release though.","accounts_in_message":[],"_revision_number":4},{"id":"0b548c7070bb8f6eb580212d3fc0a5b38a2f3602","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-30 14:53:42.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6337d34604a999fe8a4873076034313d97dbe140","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-31 23:48:36.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"01b0e0bd169413938735fdf580536986b3dbe994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 06:06:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23203a272f6f40738fa06488114ab256 : SUCCESS in 4m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6bedea96c97a4ff5bf06faeb4ed88357 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363102c4003145a5afee000c172692dd : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/949124af69e0464b94b1ab0357ab6e7f : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0dae944794c43d199669c42fb5ac232 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c860a9f04b474b638674bcadf9346fb9 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e5a8f2751014820b141c186628ca91e : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/272e96c68f3940ecaf27739365540ad4 : SUCCESS in 4m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/57f074e6350c4a20b33de67da0b0910a : SUCCESS in 7m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9ba48f239e54bdaac855cfd93b02549 : SUCCESS in 9m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2efa117d575d47979358cc781c9c788d : SUCCESS in 10m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b693cdd5028f4165b3e2cb517a613dba : SUCCESS in 11m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ec697baa377643639c4986c4b1ae065b : SUCCESS in 5m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7ea1b909dbb94fc8b388cee21d192c9c : SUCCESS in 4m 24s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/791d698b0b2442eca1024fd53b616b88 : SUCCESS in 8m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b958518570a446869ff8c303cddd206d : SUCCESS in 7m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cadf7d8da4c5473b9e46162f90ffc00b : SUCCESS in 7m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/362461f11ee94382a3c6e1c6b74f6530 : SUCCESS in 20m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28f7467df18346208620283cce92dd98 : SUCCESS in 17m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a6ea4911ad954fdca147b4367baae828 : SUCCESS in 21m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/523ec9b6b14440f6b17e1594d6538193 : SUCCESS in 20m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c063c26f2c554299897a3c76d74976dd : SUCCESS in 18m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bcad076e9c4e4d7c87cc5b0526bbe4c0 : SUCCESS in 13m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6e87666add304838bb6091ef504dd443 : SUCCESS in 12m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e0a0fd11cd5492c9785c9db81225ec3 : SUCCESS in 17m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/65b01dd6b17a4ccfa0694f1a9d88d118 : SUCCESS in 5m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f67dbea7cea24d13ba8deb282a0742c7 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0d374fd74df34ecd9925c288d6f3b446 : SUCCESS in 6m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a138f980c5c4402eb5b9fce28804d04f : SUCCESS in 12m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f41a19eedcc747a6af4c9be55b774ae3 : SUCCESS in 8m 55s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1fad447d487d49f0b5d8de5c4bb6d9e0 : SUCCESS in 58m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/14c71e57ab4f415082631413334c15e7 : SUCCESS in 1h 09m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/913d51cf08cb40959f1126f5b80fc8f5 : SUCCESS in 1h 51m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/28a519c2619a47ab8f268e1989aed8d0 : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":4},{"id":"d9ff31edab3947831c33d5bffccbc97880bacc82","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-04 22:18:02.000000000","message":"Patch Set 4:\n\nrecheck deps merged","accounts_in_message":[],"_revision_number":4},{"id":"b4aff877ce7d72808f6262cd1a5b8c17a03e9b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 00:21:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4050b01a57f14a0a90c5ccfb8dc7353f : SUCCESS in 3m 17s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6a66945794e2429b963d3897cbc4de78 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8096e35e974747bb9a713827d42f80d9 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8bb29c135b54634a8e9fe970ba7ec89 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4859c435969647a8bba2667a4cacbafd : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81b7e6e2b3bd44fd851f0f6025fdd8bc : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b90e918197f94848abbd10e518c111fd : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55296b04bb034f4fa43994159287d416 : SUCCESS in 4m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/049c66aa647048199f5ed68b08bc10f0 : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f9d21c7e1299465484101097e039b6cd : SUCCESS in 7m 35s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fb7623f5334c4f9cb5e484976523c6d6 : SUCCESS in 9m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bddf4717fece44ed959f25828bb54a22 : SUCCESS in 8m 10s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4bcb1e39e9e14deba971bbb8340c64b4 : SUCCESS in 6m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3b311abed0894967810c134f0a805878 : SUCCESS in 6m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/738897e88de3493f8476790b282b04b0 : SUCCESS in 11m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f755148477be428eb6cf0237444254dd : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5f11a7ce99e4268867ebf20fdd89eca : SUCCESS in 7m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7c672bb2996341f6b7243fc984bda7e0 : SUCCESS in 13m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/712aa5b32d3a40a58470cecdc2dafa3a : SUCCESS in 13m 38s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b4fed6ffebb7461bb434fa977642cbcd : SUCCESS in 26m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/25d58d8c514744e2bb1af115f311df7e : SUCCESS in 21m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/afba241c2cae49b8b069f79599413647 : SUCCESS in 14m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/04d8f0aab39d462d95a999e500eb6d07 : SUCCESS in 12m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cc657e6efaac4530869eb16a03455cf7 : SUCCESS in 11m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9df2f86f80fe4d7c91cf1d58f0216ace : SUCCESS in 17m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/253d88dc48224b8a99f215706ab87010 : SUCCESS in 4m 35s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7d1e0c7512044809b91cc226a05b229d : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bec1e419274349629df2ae6533a99bec : SUCCESS in 6m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/729ed869d6f3436c872b4c3b0aaa3b5a : SUCCESS in 10m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8feea67df7344b21bcc3baaf3f1a7437 : SUCCESS in 9m 32s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/85c4b569c03e45029e950a205ca4cd7a : SUCCESS in 59m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f505c2d41764932b62922a6d282990e : SUCCESS in 1h 10m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0eeb3f91739c49a7a21c2e579d49d8d2 : SUCCESS in 1h 42m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f03de39c64bf466098aaa586e550d291 : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":4},{"id":"64fb502db9e7e3de3a5fcedc380cab93a559fcfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 00:21:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"275875df3df02a7bf79e71f20ede418637f40825","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 02:06:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"dfe184fa47ce9f5f60c524a6bb2f7cc9e2ee3774","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 02:06:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3860decc3f8043e8b97780e1a2de5c31 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eddcb72273af4ab09192f08c754c856a : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87ade5c7fbcb4dd89b3690ad87d6a5d0 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27149a3a83de472c850a64cab6547e8c : SUCCESS in 7m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7fb0167a90d485ca7a01c4199ec2f0d : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d526742b585647b08cc6836ade11f97e : SUCCESS in 6m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0b8b5c16395494fa44e93936bf03ed5 : SUCCESS in 4m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/76d2d890816743c69e8f0ab2b5e9c6cd : SUCCESS in 10m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bec65d59eeed449bbe98b31c5eec5ba3 : SUCCESS in 13m 39s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0b2885efaaa445f6a37d4c5850d10014 : SUCCESS in 8m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/56468a291cab466bbc0a0da83a905005 : SUCCESS in 14m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/07824513c4ba451c9080ab5fdd733d4f : SUCCESS in 5m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/913e4999a5de46a2a8cd28eb60dca6fa : SUCCESS in 7m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8501f67806cc453e9ab414ff502802cc : SUCCESS in 10m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/55c7a9c97d3844a8b4b5e1c2310a8f62 : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2759931ec0b44ce4b25f3dff8a631e9e : SUCCESS in 9m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f2c698909b3043da98be9b8c78f3ce7a : SUCCESS in 25m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5ece40b9139d492e9badf93f9770bd1c : SUCCESS in 17m 31s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5118e34706154addb11929e5352ffda9 : SUCCESS in 25m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bf1f4bce6334416f95548fcc3c2722cc : SUCCESS in 25m 57s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b94ab4d190894e88a7fc7a7980e67b44 : SUCCESS in 21m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2c62a08ba9324951a1ce65e4350425a5 : SUCCESS in 20m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/53b28cfdb8d64fa3b9bad88c0ec178ee : SUCCESS in 18m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1c56f89de3eb4fcabd18b5cd2cf57105 : SUCCESS in 21m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5fc99a4765e148b98e38b0230cb3742b : SUCCESS in 5m 49s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8873d33713c646f489114a46a8f2ef89 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6956dd4338ba453283de51c9319f1439 : SUCCESS in 8m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/32ab49705aa04593bad4612f6e6e532e : SUCCESS in 8m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/86693f8a30c745b0ab1ae94247279831 : SUCCESS in 11m 16s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/cad27bd60d8b4943978a2d6734d0818c : SUCCESS in 1h 06m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b53f31c4661d4f3d8374c11b3ab2f278 : SUCCESS in 1h 02m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/269cf43ca1c14cdda2405a2768e58af0 : SUCCESS in 1h 39m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/58d26c0dc090470782e52877c007ebce : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":4},{"id":"48d480e759a8272ac13d027612856679174ae34f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 02:08:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5331ca29d5a4d86916cdaceb70f1b87 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4},{"id":"3f68511765a8d3dc65e36addef1984898a01b89b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-06 11:39:52.000000000","message":"Created a revert of this change as I2091519f2527bdacb60cff39acf3092072d0f419","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7ff8d5d71481bd964c12f8639b5d49695af6252c","revisions":{"4f36953354ab36910ae16a99b3c16c176229f114":{"kind":"REWORK","_number":1,"created":"2019-10-22 14:57:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/690097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/690097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/690097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/690097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/690097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/690097/1"}}},"commit":{"parents":[{"commit":"94e035e36eea0a3ad813596f9d0cc21da5493613","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94e035e36eea0a3ad813596f9d0cc21da5493613"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 14:57:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 14:57:37.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 16.0.0","message":"update constraint for python-novaclient to new release 16.0.0\n\nChange-Id: I413265a2ef080304723e1a56a031c29a60d1a707\nmeta:version: 16.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: zhangbailin \u003czhangbailin@inspur.com\u003e\nmeta:release:Commit: Brin Zhang \u003czhangbailin@inspur.com\u003e\nmeta:release:Change-Id: I5d2a039ba3f60d81a67b35592a2e4cbf2be555ca\nmeta:release:Code-Review+1: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Dan Smith \u003cdms@danplanet.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f36953354ab36910ae16a99b3c16c176229f114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f36953354ab36910ae16a99b3c16c176229f114"}]},"branch":"refs/heads/master"},"fc3f99802610d9c4b2fa8922bfbfcdb7b05d4b58":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-23 01:11:31.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/97/690097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/690097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/690097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/690097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/690097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/690097/2"}}},"commit":{"parents":[{"commit":"94e035e36eea0a3ad813596f9d0cc21da5493613","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94e035e36eea0a3ad813596f9d0cc21da5493613"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 14:57:37.000000000","tz":0},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-23 01:11:31.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 16.0.0","message":"update constraint for python-novaclient to new release 16.0.0\n\nDepends-On: https://review.opendev.org/690134\n\nChange-Id: I413265a2ef080304723e1a56a031c29a60d1a707\nmeta:version: 16.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: zhangbailin \u003czhangbailin@inspur.com\u003e\nmeta:release:Commit: Brin Zhang \u003czhangbailin@inspur.com\u003e\nmeta:release:Change-Id: I5d2a039ba3f60d81a67b35592a2e4cbf2be555ca\nmeta:release:Code-Review+1: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Dan Smith \u003cdms@danplanet.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc3f99802610d9c4b2fa8922bfbfcdb7b05d4b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc3f99802610d9c4b2fa8922bfbfcdb7b05d4b58"}]},"branch":"refs/heads/master"},"6b6564c98de7b8c6785bd50284577135f3870618":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-24 18:35:36.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/97/690097/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/690097/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/690097/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/690097/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/690097/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/690097/3"}}},"commit":{"parents":[{"commit":"94e035e36eea0a3ad813596f9d0cc21da5493613","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94e035e36eea0a3ad813596f9d0cc21da5493613"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 14:57:37.000000000","tz":0},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-24 18:35:34.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 16.0.0","message":"update constraint for python-novaclient to new release 16.0.0\n\nDepends-On: https://review.opendev.org/690134\nDepends-On: https://review.opendev.org/691039\n\nChange-Id: I413265a2ef080304723e1a56a031c29a60d1a707\nmeta:version: 16.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: zhangbailin \u003czhangbailin@inspur.com\u003e\nmeta:release:Commit: Brin Zhang \u003czhangbailin@inspur.com\u003e\nmeta:release:Change-Id: I5d2a039ba3f60d81a67b35592a2e4cbf2be555ca\nmeta:release:Code-Review+1: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Dan Smith \u003cdms@danplanet.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b6564c98de7b8c6785bd50284577135f3870618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b6564c98de7b8c6785bd50284577135f3870618"}]},"branch":"refs/heads/master"},"7ff8d5d71481bd964c12f8639b5d49695af6252c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-29 09:31:36.000000000","uploader":{"_account_id":7186,"name":"Matthieu Huin","email":"mhuin@redhat.com","username":"mhu"},"ref":"refs/changes/97/690097/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/690097/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/690097/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/690097/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/690097/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/690097/4"}}},"commit":{"parents":[{"commit":"5bf5a58293a0cbeee5dca4afa3bb69401fbd8961","subject":"Merge \"tripleo-common - blaclist 11.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bf5a58293a0cbeee5dca4afa3bb69401fbd8961"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 14:57:37.000000000","tz":0},"committer":{"name":"Matthieu Huin","email":"mhuin@redhat.com","date":"2019-10-29 09:31:36.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 16.0.0","message":"update constraint for python-novaclient to new release 16.0.0\n\nDepends-On: https://review.opendev.org/690134\nDepends-On: https://review.opendev.org/691039\n\nChange-Id: I413265a2ef080304723e1a56a031c29a60d1a707\nmeta:version: 16.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: zhangbailin \u003czhangbailin@inspur.com\u003e\nmeta:release:Commit: Brin Zhang \u003czhangbailin@inspur.com\u003e\nmeta:release:Change-Id: I5d2a039ba3f60d81a67b35592a2e4cbf2be555ca\nmeta:release:Code-Review+1: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Dan Smith \u003cdms@danplanet.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ff8d5d71481bd964c12f8639b5d49695af6252c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ff8d5d71481bd964c12f8639b5d49695af6252c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
