)]}'
{"id":"openstack%2Frequirements~690374","triplet_id":"openstack%2Frequirements~master~Ifbd842bc0eee02608c96c2f0be3a60a30c46f6f8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifbd842bc0eee02608c96c2f0be3a60a30c46f6f8","subject":"update constraint for python-keystoneclient to new release 3.22.0","status":"MERGED","created":"2019-10-22 19:22:05.000000000","updated":"2019-10-23 08:11:22.000000000","submitted":"2019-10-23 08:08:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690374-1571818136618-d71617f4","meta_rev_id":"1ab31394305a833e7963adbf48e56b3539714519","_number":690374,"virtual_id_number":690374,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 08:08:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-22 21:18:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-23 01:12:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-23 01:12:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 21:18:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-23 01:12:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-23 08:08:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cefeb4cb78ec69c0906c50f94a6e2156549bbadf","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-22 19:22:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3b0471747891d8247c293d54a0920ae021d10c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 21:10:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3360948b296b49e29dac786ba7a1e4e4 : SUCCESS in 2m 58s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0e4e60e651584520b3f4f754b1d72271 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0eefeef22db442fb45d0b8434700bc6 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a017d64f902f4b2597509c72492008e5 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb6d517591d946bcbc5b2ceaa453dcee : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6991e5ad1b7c43989c5f898206a3bef3 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fff71f30f59544d98fa76c135eebef6e : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ddb7bf83ba647b896f1e81c98eca458 : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/449da821fcf445ebb2c4bb615eca6830 : SUCCESS in 6m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8cf0dd7ee1814f0baf2feebc489d28dd : SUCCESS in 6m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a78c653dee164cb086db6ae9532ef6e0 : SUCCESS in 11m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6b293e1c7ed5421fa14603095da099f3 : SUCCESS in 8m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5caff6aa185c4081a7f13955983517a5 : SUCCESS in 5m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8b4722147bf040f08eedde8d4b137119 : SUCCESS in 4m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8659e0dd258045789352baf5e32e2e3f : SUCCESS in 11m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/88e9e95ce25e4da59863e6684ba78071 : SUCCESS in 7m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8fecd85fe0ec427e902c4f0b78fd1201 : SUCCESS in 10m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f759800c5e8642beb9e59c55f7c43818 : SUCCESS in 22m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dda725b3b27145f184e0adf85b2be050 : SUCCESS in 12m 14s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/25a5d9b989964805bceee4252678b1ff : SUCCESS in 29m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8ae486fde12d4d78bf70c6196e26fad2 : SUCCESS in 23m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a79cf854941f420ca7cb9e1746ad1497 : SUCCESS in 19m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/43c50f2a0c2d4aa18b9513f8590a95f3 : SUCCESS in 12m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a1f3008780874a519c06d92433679463 : SUCCESS in 12m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4cf0e71fe0041e88e28935af90a6da1 : SUCCESS in 19m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/93169accbcad4e60a57e8f52ba430266 : SUCCESS in 6m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ed46fd13705f4e498427094ca04dc22e : SUCCESS in 9m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c1183ca8c65a4918954302b3b1af5fea : SUCCESS in 8m 48s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/78435e527b074424936c1abd695bf928 : SUCCESS in 1h 01m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f7949590bd14f138c851a4cc9a767bb : SUCCESS in 1h 16m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b18d2a8b473b48dd8829ac6cb936b13b : SUCCESS in 1h 40m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/511fb18bdb4b4e4992bc5b31c0ec4647 : SUCCESS in 58m 43s","accounts_in_message":[],"_revision_number":1},{"id":"87a2b53c49421a74ea00d251d63329e2519cad47","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-22 21:18:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b56f8e9550eb550e60b82727fd7e88519155edff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-23 01:12:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7bddbe5c3cd4be76c9df6cbf19a0c8e89f0ffccc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 01:13:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb1e264a5b404778cf9e06ce7dbb7ed2fe2ce71a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:08:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a870f9e7977bf5e218715439a0ea5887e5473eee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:08:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0669aaa1e88d4121b50a6dac21fb89ec : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efee61f648b3459c98f76099b343f484 : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3fa1a37522e4d78aa18d7c90ace449b : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/047f818984ff44168b947207b837d376 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d9cf4b763554ccbb842ddd549fe783a : SUCCESS in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7c37cc7dbdf4656a385d2993784b8e7 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8697c33992249e9bfe900f0dbc7389a : SUCCESS in 3m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/68c2e593ad104962ab4af6211aaf9724 : SUCCESS in 6m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6f8d522537c34cc1b38ee4d6aa676ae9 : SUCCESS in 8m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d0d9e6454fbd411685939238631dd0b8 : SUCCESS in 11m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8beae7682e324a069d9a4df1a209994e : SUCCESS in 9m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a457260966a74cb9b5f9d4ce5141af68 : SUCCESS in 5m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3c77ac5925fb4f3587feec5661757213 : SUCCESS in 5m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5b6f73532fdd4b2eb2bd7ea86451e6eb : SUCCESS in 10m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/514ebac0745f4abeb796bc90b22a9aa4 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7a77d5bef324eacb6247821efa00981 : SUCCESS in 7m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7abf095bbe1a465ab365397933efa051 : SUCCESS in 21m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c13180ecf5054306bd96adfcb0c170ae : SUCCESS in 12m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4cab72990dc74524a9b0b8d17174dacc : SUCCESS in 23m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fbe5fe170bb644a5bd09814f036ded8c : SUCCESS in 22m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ebf9b1f0549341deb001ee1a96af2114 : SUCCESS in 13m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/77b80cfc387243a8bcac681e2d2ec17d : SUCCESS in 16m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/931989489e9a436793ee9811eb35b765 : SUCCESS in 13m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1350fa6aafd479f8d2d8a913369b58d : SUCCESS in 19m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cfb5dbf04ca441078f396816e3b224d6 : SUCCESS in 6m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e7b205d7422841809a9c126e7df8bce6 : SUCCESS in 9m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4eebfc1cdbe74faa86d25b423d228bf9 : SUCCESS in 10m 31s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/b2b6765fc1dd42c7b5f303ff9f52c646 : SUCCESS in 58m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/245429447d844c05b4fe26d8d4009fd7 : SUCCESS in 1h 13m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1eab41e9bfca41be8836c96961b61628 : SUCCESS in 1h 56m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b92a710c2d98465bb44afa1efcbde358 : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":1},{"id":"967cac27eaa43d222b67412af2b12e245f2dac4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:11:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82894d2d271d4aa09a091ee39ff443cf : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87af13d404b02b3ecfee8da7c01b8314e91deb9c","revisions":{"87af13d404b02b3ecfee8da7c01b8314e91deb9c":{"kind":"REWORK","_number":1,"created":"2019-10-22 19:22:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/690374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/690374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/690374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/690374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/690374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/690374/1"}}},"commit":{"parents":[{"commit":"94e035e36eea0a3ad813596f9d0cc21da5493613","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94e035e36eea0a3ad813596f9d0cc21da5493613"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 19:22:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 19:22:04.000000000","tz":0},"subject":"update constraint for python-keystoneclient to new release 3.22.0","message":"update constraint for python-keystoneclient to new release 3.22.0\n\nChange-Id: Ifbd842bc0eee02608c96c2f0be3a60a30c46f6f8\nmeta:version: 3.22.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Colleen Murphy \u003ccolleen.murphy@suse.de\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ic0f8ac4384191c33c1f3d680e5d0ea21db3f405c\nmeta:release:Code-Review+1: Vishakha Agarwal \u003cagarwalvishakha18@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87af13d404b02b3ecfee8da7c01b8314e91deb9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87af13d404b02b3ecfee8da7c01b8314e91deb9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
