)]}'
{"id":"openstack%2Frequirements~690394","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I898d0c4eaba28969201a08e05ac204628a85c9d6","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I898d0c4eaba28969201a08e05ac204628a85c9d6","subject":"update constraint for ovsdbapp to new release 0.10.5","status":"MERGED","created":"2019-10-22 20:03:19.000000000","updated":"2019-10-23 08:09:00.000000000","submitted":"2019-10-23 08:08:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690394-1571818140166-d762e54f","meta_rev_id":"2d88cc5d527e2730e2450d831e81bb697ac5da1f","_number":690394,"virtual_id_number":690394,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 08:08:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-22 21:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-23 01:12:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-23 01:12:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 21:19:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-23 01:12:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-23 08:08:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d98f26c0712dcccfdb4d455222e72dca9c842c13","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-22 20:03:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e2fd8c9eebb719812889c04df8cefbc340d3b5b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-22 21:19:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f6d5bff7d750aa0787db878c611eb2159001a814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 21:40:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/07482571f7f34f32917bb2d2a45abc8c : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/256a2bf7e276445b915242cf3ddc17fb : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/90ffedc9c5f44def87ec2a9e4f178867 : SUCCESS in 4m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c27dc4b1297c46418ede2ae40114c13b : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13c6ec02405f4d58be2e50a61a5926d5 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ace8b45d5a049818b41f1eb35fae16b : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8846281d106046888f48c99d1986fc0e : SUCCESS in 6m 55s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/568dd7d1d9934c0aa18dbadbf978a0d2 : SUCCESS in 7m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7afa3d97e7224b508f029f8a86f351ee : SUCCESS in 9m 20s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/f6c1138bbfae497aba7c0a643ec972bd : SUCCESS in 8m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9adfb68b01844318bc1d5480c82cbb38 : SUCCESS in 4m 45s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/1caea5dfb090474d969afa0e8ca70682 : SUCCESS in 4m 55s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7b21c78129ea4fa39380050f6bc24b54 : SUCCESS in 10m 41s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/5280feb89fde4e41b51a57eb4e8a4f24 : SUCCESS in 9m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/57be5af1d29b4753bec217da094fe07f : SUCCESS in 14m 38s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/49bee699225040b7b83628fd55dff63c : SUCCESS in 11m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fbd779e2b18a413b83062ff6cf088b89 : SUCCESS in 24m 47s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/9a82f618da7447a9b47449de1db8463d : SUCCESS in 27m 54s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/55e19bbebab3451a84cb1b80f900b895 : SUCCESS in 15m 43s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/d0deb4b395ca47aeb0832222a290bd44 : SUCCESS in 14m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/81f88c55fc2b46b5b8d528a12106830f : SUCCESS in 18m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cb2e992094f34d3da0191a3e010fa903 : SUCCESS in 5m 31s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/4aafb79b52dd433db20231a821ef0c8e : SUCCESS in 3m 28s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/244912742e834baeb0dbe4c6c9e01d49 : SUCCESS in 46m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/007199c2ee96453fae356635cbd78e42 : SUCCESS in 1h 00m 47s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/57018d02be4b4aa38849fcee52553988 : SUCCESS in 1h 30m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8668d64e91154fbc9e840d5b41987944 : SUCCESS in 45m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e229c3ca2063515fe741029a8ee6abc82adc37c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-23 01:12:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78a33ffac84064b94c69e0977adc041daef70ff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 01:13:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cad74156caea1653a05c256ce7be692a5c237b38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:08:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d23dfe06a341c1b1f8c257ed694f71 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2053fe77d614487eaa4042a478d101b1 : SUCCESS in 5m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/751644f660194a7083a2501d024524e9 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a669c04437a43ed9db8939f2197220f : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dbecf6b367434cbabfe2ff55dec682e7 : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0872f8e74c44424997031ed17b39aa63 : SUCCESS in 7m 07s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/325790bf2be242fd9cbe013db26e87f1 : SUCCESS in 7m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b5f5683bc39249f8a82d85cb31ae2bbb : SUCCESS in 9m 36s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/e367c45ce8c04448b77313a3e8b61629 : SUCCESS in 11m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/aa91056ee0164c5782a6ccecf64ef50e : SUCCESS in 9m 37s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/743dbe425fc549c7bf8550033ccdd1c5 : SUCCESS in 6m 30s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fac4686b75f24d4c83bd72a810fdac01 : SUCCESS in 11m 01s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/da07622a0e08468d8ad22935f69a00af : SUCCESS in 10m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/aa7510a4b1aa4aa7b284cf21ed3537e0 : SUCCESS in 14m 42s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/484da2bc752241ee9a55af207efab956 : SUCCESS in 18m 14s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5bd58f3b84914d84897b38092b7d84ce : SUCCESS in 26m 01s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/fe824de49d8748ac81b76c6aae7bf055 : SUCCESS in 30m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a87b72b668f8417c978f7bca24c3fb80 : SUCCESS in 14m 18s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/69f8dd1b232241cca17d38c160e59f88 : SUCCESS in 14m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55cfcf5fad294037a3e50df6b0c9ef0b : SUCCESS in 27m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/757c13cb204e4d258ace8f70d7c9f73e : SUCCESS in 6m 36s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/ddf047f212f44a62b0b110a2f945d829 : SUCCESS in 5m 23s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f41958a645c9427fa0fe5b5980034dd0 : SUCCESS in 57m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/389a1efb6db347bf878c8f2270b8428c : SUCCESS in 1h 13m 29s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/4de156fc80684700a5e9a069e75f785e : SUCCESS in 1h 19m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/819453cb8ebf4442a6e4056b61fa7731 : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4d2f761622a1bb65c6fafcce85659c513c940031","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:09:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"729da527fe648a33a3f8284a047ab333a1ff3d64","revisions":{"729da527fe648a33a3f8284a047ab333a1ff3d64":{"kind":"REWORK","_number":1,"created":"2019-10-22 20:03:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/690394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/690394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/690394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/690394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/690394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/690394/1"}}},"commit":{"parents":[{"commit":"bb9337cc90272f37992abcbe57ebd82d6e66a285","subject":"update constraint for oslo.messaging to new release 5.35.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb9337cc90272f37992abcbe57ebd82d6e66a285"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 20:03:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 20:03:14.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 0.10.5","message":"update constraint for ovsdbapp to new release 0.10.5\n\nChange-Id: I898d0c4eaba28969201a08e05ac204628a85c9d6\nmeta:version: 0.10.5\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta:release:Commit: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta:release:Change-Id: I757b0c9954d803a34e1db5abc21ef0a926b827f4\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/729da527fe648a33a3f8284a047ab333a1ff3d64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/729da527fe648a33a3f8284a047ab333a1ff3d64"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
