)]}'
{"id":"openstack%2Frequirements~690416","triplet_id":"openstack%2Frequirements~master~I2f22a640d14d7a1538532b14e43417d35c2cc515","project":"openstack/requirements","branch":"master","topic":"uncap-cmd2","hashtags":[],"change_id":"I2f22a640d14d7a1538532b14e43417d35c2cc515","subject":"Revert \"Apply cmd2 upper bound to python3\"","status":"ABANDONED","created":"2019-10-22 21:11:47.000000000","updated":"2020-08-10 14:00:35.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"30c5b68b8c68f2ed27bba482cef708339ccfb774","_number":690416,"virtual_id_number":690416,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-10 22:09:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2020-06-10 22:43:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 21:17:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-10 22:09:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 22:43:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a49624b481273f4f3f9b424f1d4f76896d179ad5","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-10-22 21:11:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec0838c32a28783e78af88b426ab384b49565eca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 23:09:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ad9be30b04349bf852368ec1dedbcf5 : SUCCESS in 9m 19s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cdb9809dc1ce4966973198b6c8e9cac7 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c389a642d22c4fd28cbb6e55b30f5e11 : SUCCESS in 9m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64194a61d80848d39b259063adb83277 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fea532505c142fa8cc0e94ba4fac116 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2792c9bd07994cf39441928745f2c7de : SUCCESS in 9m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be9f7655dd9246199ca8c95073f18ef3 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/918cf23e71ef44609f1b000d5be6d24d : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3d050f86b1df449fb593f03080d853a8 : SUCCESS in 9m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0b2e21c5af8743558717d0739b8d7ca5 : FAILURE in 6m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/054a464a04d646e7b5d819f9aa22829e : SUCCESS in 11m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dee2d1ad3e714ea490fc5ebfa589d3d3 : SUCCESS in 15m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/39685336c6dd4ae889f92ef4fe51a7a4 : SUCCESS in 4m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6437fafd249548819aa0452376992e5b : SUCCESS in 10m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d5f7dd96283a429dbf1138bf58fa4f10 : SUCCESS in 9m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/01efb3cb084a4c979e6bf7bebac428c7 : FAILURE in 5m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24a73527d2fb43f99483699f7a9cddf1 : SUCCESS in 5m 49s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/599a8a27921c43dca651aa0336ddfa1f : SUCCESS in 17m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1f17616568574765ad56d7e8ab2d1571 : SUCCESS in 18m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/cb250c6de87d4d98805ef0f44fab5906 : SUCCESS in 22m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/789cd4a1eb0b4b0fad3544008bac9c64 : SUCCESS in 21m 57s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1999b0730a3541039a20fddec083772d : SUCCESS in 12m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4ce53a74fa7d4d35b91cfc2dcdf83387 : FAILURE in 4m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0135ecf2dc15431f858ac8154f9600b8 : FAILURE in 5m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/00faab6f3e6e42f4a4cac4503913d3d7 : SUCCESS in 16m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a8070957a9494881b431c48cf6e2aee3 : SUCCESS in 4m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d3fa7fadc869472f8fc1832b6d3b9cd1 : SUCCESS in 8m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a21764d35ae04caf99a5ef12cb30b09b : SUCCESS in 7m 48s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/b851b4c1c8f74e99aa0a4fcda78c279e : SUCCESS in 1h 12m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdcdb3602bea459c8d5b5e4981dc80a1 : SUCCESS in 1h 16m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e54e02a0522843ef919068e96fd71d0b : SUCCESS in 1h 55m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/012b3cf9d3004a4a945c344ce77b94c0 : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":1},{"id":"052f40ff4a1925b1cb47c63e50bb4318f12c9353","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-23 01:13:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dab2d937961f452e44807ccc00199b327b972236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 03:10:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecf5e7580dc741afaaa3c27d1623dc67 : SUCCESS in 4m 33s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/dbec432bedd348de96459293ec82ec74 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f9df48590ca4b97acd81501c5593cfb : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d9d5725cc554ca58529e59f93ac266a : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a37001e6054747e185327abb4a31c78e : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99bf77f99c0946bba568f24bc2be26ea : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d134da06fae421a9e39d40367115f9f : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb7c8671685647b4afd32a00c3e0649d : SUCCESS in 2m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e98cb4084c9041c6bbf3d20aebe3ef19 : SUCCESS in 6m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f6eab7a9a0d04957a55e2487b96b71e0 : FAILURE in 7m 21s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/77c9cb5583d247798306086a67776754 : SUCCESS in 9m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/47b7ca1f1ca846eba5a7589d8f18a56d : SUCCESS in 8m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/424e432ede074964bca9a6141997882c : SUCCESS in 4m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f1300075a1484fd181066f851c4e0b96 : SUCCESS in 4m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/320ed54a9f6c41cabeb4fc3af16d9dd4 : SUCCESS in 10m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ecc6601ae4bc4ab9bc17887530e3c84c : FAILURE in 3m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46804a54f2b8465b99f49075a854719f : SUCCESS in 7m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4cb1b4a136234acb9d77bc80cedfd369 : SUCCESS in 13m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fa4f33339153416cb8a7c84c4f1f86c3 : SUCCESS in 15m 42s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3a0661b442da4365848b87eb59adbcc9 : SUCCESS in 26m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ec2062d6a4054f67a8e8330476a58f70 : SUCCESS in 22m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/54c64a2d3ad34ebca4121fc405947c3e : SUCCESS in 11m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/10c8cd77c8784355b0d7ab2ba3f32545 : FAILURE in 9m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4e7a286eeac2455fad732ca0cf5e5588 : FAILURE in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4b02da122dd24e2bb02c256c0fcb0541 : SUCCESS in 19m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f0c86dfa88d41cdac3396d3e6183294 : SUCCESS in 5m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7c245e9385564456b662c4e805750b0d : SUCCESS in 8m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/013f36c274d34d80833da8edd2cd6734 : SUCCESS in 8m 59s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ed82dc0a2556483e9b8fbfd58761c312 : SUCCESS in 56m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e581ff70571e4212a1f42a41e8d32bdf : SUCCESS in 1h 10m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/facc1053a0524ebea12bc165997ca2d6 : SUCCESS in 1h 47m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/04bab2abd9b04f51871325e0cd98ab67 : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6ddf7707fc60e65cb59c8112d02f2e1445da866e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-23 03:20:50.000000000","message":"Patch Set 1:\n\nit looks like we could be good but we are having install problems because we have to uncap, get the changes out then retest (or do the dependant config dance, but that is possible, make uncap reviews and make this depend on them).\n\nit actually looks like it\u0027s \u0027only\u0027 used by cliff so if we could remove or migrate off of it that\u0027s another option\n\n+--------------------------------------------------+----------------------------------------------------------+------+------------------------------------------------+\n| Repository                                       | Filename                                                 | Line | Text                                           |\n+--------------------------------------------------+----------------------------------------------------------+------+------------------------------------------------+\n| airship/promenade                                | requirements-frozen.txt                                  |   12 | cmd2\u003d\u003d0.8.9                                    |\n| jjb/python-jenkins                               | test-requirements.txt                                    |    1 | cmd2!\u003d0.8.3,\u003c0.9.0;python_version\u003c\u00273.0\u0027  # MIT |\n| jjb/python-jenkins                               | test-requirements.txt                                    |    2 | cmd2!\u003d0.8.3;python_version\u003e\u003d\u00273.0\u0027  # MIT       |\n| openstack/cliff                                  | requirements.txt                                         |    5 | cmd2\u003e\u003d0.8.0,!\u003d0.8.3,\u003c0.9.0 # MIT               |\n| openstack/requirements                           | global-requirements.txt                                  |   25 | cmd2!\u003d0.8.3,\u003c0.9.0  # MIT                      |\n| openstack/requirements                           | openstack_requirements/tests/files/upper-constraints.txt |  350 | cmd2\u003d\u003d\u003d0.7.0                                   |\n| openstack/sahara-specs                           | requirements.txt                                         |    7 | cmd2\u003c0.9.0                                     |\n| openstack/tripleo-quickstart                     | requirements.txt                                         |    1 | cmd2\u003d\u003d0.8.5                                    |\n| openstack/upstream-institute-virtual-environment | elements/upstream-training/static/tmp/requirements.txt   |   75 | cmd2\u003d\u003d0.8.9                                    |\n| x/group-based-policy-ui                          | test-requirements.txt                                    |   14 | cmd2\u003c0.9.0                                     |\n| x/osops-tools-contrib                            | ansible_requirements.txt                                 |    6 | cmd2\u003d\u003d0.6.9                                    |\n+--------------------------------------------------+----------------------------------------------------------+------+------------------------------------------------+","accounts_in_message":[],"_revision_number":1},{"id":"6a25e9923549622bef2d2fea087d9670e73fc02d","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-10-23 07:36:27.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e8207a435e7174057301a2a9a877d575d590bdec","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-10-29 10:54:43.000000000","message":"Topic set to uncap-cmd2","accounts_in_message":[],"_revision_number":1},{"id":"242322544c4a7c718e7ea91b3f7a55a0d6ae7a07","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-16 21:52:10.000000000","message":"Patch Set 1:\n\nping on this?","accounts_in_message":[],"_revision_number":1},{"id":"36348a93fae7ae137d13032bc99f862a8c853f4e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-14 16:18:28.000000000","message":"Patch Set 1:\n\nanother ping","accounts_in_message":[],"_revision_number":1},{"id":"5b0f8f93bba1ca3fcec6cecaab03880d04759c27","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-03-18 20:41:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f6f49b953acbd79d61582ffe6cb9da6f257abde0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 22:17:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635613512170424eb5825bc65b2b6633 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d82ecac836439da656440c62161709 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb8248ea5c374c4e86b84c8911e5ce8b : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d3a2cbf1fbd447f8f5dc6539aad5f5e : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1862a5b4c37040978c96d81b427dc6c8 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d301bdedfec24bb78a254ac30f6b95d7 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae6e0e4454c34b0d8b44d47c3db3bef2 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fc17a0480cb64eeba51af941a31e4de3 : FAILURE in 8m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/87dd8d8d5d834ef190ba8f580c23a0fb : SUCCESS in 8m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/10aee79504484dd99cea9119c9fade0c : SUCCESS in 4m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8d028115ed244c0aa14c34f407f7cf05 : FAILURE in 3m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a538aca1d41d4a09a0a2509e3ade3e6d : RETRY_LIMIT in 6m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/13894128f0604cb482637828874b26bf : SUCCESS in 14m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e8df96369aa84ca2983e1568c0c19a89 : FAILURE in 4m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1853cb87399d47448b355841ac4b74d2 : FAILURE in 5m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7be7af31085a4569aef4512b1e63c4a7 : FAILURE in 5m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4bc131fb346842cd91921b63a3d7b46b : FAILURE in 5m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/87b41af54e5c42658e5b8d79b9bcb6ba : SUCCESS in 5m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/59f0cb1763b549ec8158c7ae2bbe0e85 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9878dc78801a4da6b25562eccc16d1b9 : SUCCESS in 6m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5d69e6cb942e404496cd12610c35ab37 : SUCCESS in 9m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/235a5a1c03364bd3b8373f40b2fb3bde : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/809c89d05e2244fa885835ca01cb90b1 : SUCCESS in 53m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/deb1de81ac6144fe8c39ca2bf2dc77b1 : SUCCESS in 1h 13m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c3055e213824b388a9df636099e8906 : SUCCESS in 1h 29m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a4e571c6683f462197f9bd9625d236b0 : SUCCESS in 1h 07m 44s","accounts_in_message":[],"_revision_number":2},{"id":"03ebd7297b161b1a6cdbb2df43e0943a5e6a5537","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-06-10 20:40:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"821444003c1844af846759381a1a78fa43ed91c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 22:09:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/667007641cbf476bba8524abc75b7b97 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a21c36bdfe5471b8baa041bb7dba2ae : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e3921404b9d42698557ca6306281733 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7c98ad740c54f6b886259e21d947ad9 : SUCCESS in 8m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e2802b8b23ed4b94bab838bc5bdf1c3a : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3a7579f98c7246c9b7f4d72fc361a394 : SUCCESS in 4m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/567cd265d4ce413b8b2a96f9d5836974 : SUCCESS in 15m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fe3ae2588224409eb661a3e7b50c5dc7 : SUCCESS in 9m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c80c5abbd22a4b5496de1d43755c9158 : SUCCESS in 9m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bf33c1151f0846a4b77f9ec007980c81 : SUCCESS in 8m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a458c7429b854152ab2f78aa86be4c11 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39e9ba4dc8c4400ea2ef7f6f9459aaa0 : SUCCESS in 13m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d7f2bd5833a44a1b9301bafcb2f3b2aa : SUCCESS in 21m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4250db99fa1d4560bb859f51fc2e9cad : SUCCESS in 27m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/93b27122e95844cb8f2e7da8436c084c : SUCCESS in 16m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/29c1a119b16a4cbb918d767717a4d585 : SUCCESS in 14m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/133a2e23ce6b4c498768215843e56792 : SUCCESS in 7m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5f166a84b3442b2abe4f12ae771ba8c : SUCCESS in 6m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cb323768dbb741e6aadecb4b6961c6d5 : SUCCESS in 19m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfb0b9a648b049c2960d28e559446168 : SUCCESS in 56m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72647bb9d563470ebca2bcc76314bca3 : SUCCESS in 1h 24m 36s","accounts_in_message":[],"_revision_number":3},{"id":"11b90057802423c35311a7a4833246ed551864ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 22:35:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a3d94f88012d458198b280c250ca208e : SUCCESS in 1h 50m 40s","accounts_in_message":[],"_revision_number":3},{"id":"b0959386308777e06535099d882e36952feb3edf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-10 22:43:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"30c5b68b8c68f2ed27bba482cef708339ccfb774","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-10 14:00:35.000000000","message":"Abandoned\n\nwe are uncapped and not using py2 either","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"79626cf57380788576724c5220c7513de363cd5b","revisions":{"983e956d4eb90b5fafa2421ad04f1519eb6ac095":{"kind":"REWORK","_number":1,"created":"2019-10-22 21:11:47.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/16/690416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/690416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/690416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/690416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/690416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/690416/1"}}},"commit":{"parents":[{"commit":"9806c0bbfda9504e5447d11757ff08bac1d90658","subject":"Merge \"Switch to opensuse-15 nodeset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9806c0bbfda9504e5447d11757ff08bac1d90658"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-10-22 21:10:57.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-10-22 21:10:57.000000000","tz":120},"subject":"Revert \"Apply cmd2 upper bound to python3\"","message":"Revert \"Apply cmd2 upper bound to python3\"\n\nGenerally a good idea to not have upper caps on our dependencies.\nLets see if we can finally move forward with the newest cmd2 release.\n\nThis reverts commit af04e0c0292cbdd6db6159bbcf2723c9c9fd3b6a.\n\nChange-Id: I2f22a640d14d7a1538532b14e43417d35c2cc515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/983e956d4eb90b5fafa2421ad04f1519eb6ac095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/983e956d4eb90b5fafa2421ad04f1519eb6ac095"}]},"branch":"refs/heads/master"},"50788419648205ad44453c07accc143b697ba093":{"kind":"REWORK","_number":2,"created":"2020-03-18 20:41:56.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/16/690416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/690416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/690416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/690416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/690416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/690416/2"}}},"commit":{"parents":[{"commit":"38db6602f7862bf7a798ce5f961c42ea76c45055","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38db6602f7862bf7a798ce5f961c42ea76c45055"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-10-22 21:10:57.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-03-18 20:41:44.000000000","tz":60},"subject":"Revert \"Apply cmd2 upper bound to python3\"","message":"Revert \"Apply cmd2 upper bound to python3\"\n\nGenerally a good idea to not have upper caps on our dependencies.\nLets see if we can finally move forward with the newest cmd2 release.\n\nThis reverts commit af04e0c0292cbdd6db6159bbcf2723c9c9fd3b6a.\n\nChange-Id: I2f22a640d14d7a1538532b14e43417d35c2cc515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50788419648205ad44453c07accc143b697ba093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50788419648205ad44453c07accc143b697ba093"}]},"branch":"refs/heads/master"},"79626cf57380788576724c5220c7513de363cd5b":{"kind":"REWORK","_number":3,"created":"2020-06-10 20:40:23.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/16/690416/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/690416/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/690416/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/690416/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/690416/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/690416/3"}}},"commit":{"parents":[{"commit":"101bc83be2dd0d7870c37bdb3d8fd758c8caaf19","subject":"Merge \"update constraint for cliff to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/101bc83be2dd0d7870c37bdb3d8fd758c8caaf19"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-10-22 21:10:57.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2020-06-10 20:40:05.000000000","tz":120},"subject":"Revert \"Apply cmd2 upper bound to python3\"","message":"Revert \"Apply cmd2 upper bound to python3\"\n\nGenerally a good idea to not have upper caps on our dependencies.\nLets see if we can finally move forward with the newest cmd2 release.\n\nThis reverts commit af04e0c0292cbdd6db6159bbcf2723c9c9fd3b6a.\n\nChange-Id: I2f22a640d14d7a1538532b14e43417d35c2cc515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79626cf57380788576724c5220c7513de363cd5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79626cf57380788576724c5220c7513de363cd5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
