)]}'
{"id":"openstack%2Frequirements~690419","triplet_id":"openstack%2Frequirements~master~I989394e0d31ae69dffa3b455bdea6edf127b18e8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I989394e0d31ae69dffa3b455bdea6edf127b18e8","subject":"update constraint for sqlalchemy-migrate to new release 0.13.0","status":"MERGED","created":"2019-10-22 21:19:27.000000000","updated":"2019-10-24 02:54:43.000000000","submitted":"2019-10-24 02:53:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690419-1571885618989-89326a6a","meta_rev_id":"60c503382bdf5dc898ae099cff11eacf5f4297d6","_number":690419,"virtual_id_number":690419,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-10-22 22:09:56.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 02:53:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-22 21:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-23 03:17:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-23 03:17:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 21:20:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-22 22:09:56.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-10-23 03:17:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-24 02:53:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf3d4e0c85f3af4ddbf1ad7a57940c87539f70f6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-22 21:19:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9af710aba7953e6a0e8b4b999c465159023eb651","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-22 21:20:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8902de3845fc07e21695a53135c3d92fd69e15e7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-22 22:09:56.000000000","message":"Patch Set 1:\n\nzuul is failing on a few jobs, looks like maybe this hasn\u0027t hit all of the pypi mirrors yet.","accounts_in_message":[],"_revision_number":1},{"id":"4e8aaca4364b4b076621dc23a3d0fb0928a408ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 23:19:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40f00bf37ea74205a521a85201fabafc : SUCCESS in 3m 29s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4384850631da481d9a108278fb23e034 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d77a7c200f467883922dff2556b7f4 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a44f384cdce4257a6fa919edd6a328d : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52833815c8f44c02932eaf54722f22e0 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac08ad45170d4ebe8dc17c2350bac598 : SUCCESS in 7m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48ca4f01228f4784a64c4680e91bf021 : SUCCESS in 5m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5ba1bc204e7e42af98f1ec2907867cfe : SUCCESS in 4m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/81197511735b4b64b629f8a420b94cc1 : SUCCESS in 8m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a817889db4c646f993683810cc91ee43 : SUCCESS in 8m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f3d6f134495943bfbe31a126f08da454 : SUCCESS in 9m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bd86e9feb6fd4ce78cccc2a9c8b30b51 : FAILURE in 3m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e3eb1d72a67a46b2bc28ce98b7ffcdce : SUCCESS in 6m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ee948b688560469cb6241512abbcdb9d : FAILURE in 3m 55s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/75d116b0e4bc4686a58ae0319b493d42 : SUCCESS in 9m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c8cb70508d5e48e3bb8debc453751c55 : SUCCESS in 9m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f68bbb8116704707afcfe899b8a541ed : SUCCESS in 6m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1c59a2873a6b4cb89037912ab4168953 : SUCCESS in 15m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b4466655616b40558bc1e4e8ab54a5f2 : SUCCESS in 12m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7b8203aeca0246108a4c66b0904271bf : SUCCESS in 26m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/34335808256844e28961c6ed8e5fcf7c : SUCCESS in 21m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f4a32371ac9d43a6b8dd684fe6ef77ca : SUCCESS in 14m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/11caa1954a774c5a8dd9d2f4f3c707c1 : SUCCESS in 12m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/405285391c984d189895b6036ac39c8b : FAILURE in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3da5975eed0e40ceb404283226a0bc03 : SUCCESS in 20m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4b92544ec034218809f8bd5da9add4e : SUCCESS in 5m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1e12940664af4173bd4915c0a7d6b10a : SUCCESS in 9m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/30122491504d4e3c8a8165cd541b1b8d : SUCCESS in 10m 14s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f04123c35ccd47c0a02276ce667ac2b8 : SUCCESS in 56m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d4b5a4565db49c89a4a3604a5406792 : SUCCESS in 1h 06m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b3c995ddc86b43b084cb735f8f29e78f : SUCCESS in 1h 57m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f4affc7232f94bab9cc738ff2b28910e : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0ec71690ce05d5fcbb5cc9c30c5827c61cb6dd24","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-23 01:13:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"98e31585cd53e099c3fadcb2bffd72edcca83525","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 03:06:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/470b70a5ce0b4f71a92d821f37ae61d4 : SUCCESS in 3m 54s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/50a90fb89582408080d1033fc1dca73a : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4bd5cf382954dfa96bf7d0bf75df08a : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9d8039e4c5d4381b0de42eda064d205 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72e8b446c8d94d5c825e38810a22d6bf : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13a70e9e461a4f99bd513321661e3839 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5da6327dfca840e888cab1b821c87b84 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c5c23eec7497443a9ab7b8e321a38abb : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ff67096cdddd451595e1b70cb6978fd8 : SUCCESS in 6m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9fa80118f82c4b30aa4f58f248cde313 : SUCCESS in 6m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/249f23e4c7c1421b86f99c328c2e4729 : SUCCESS in 10m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b08797911b8e4ac7bd0cec0f3bb9e386 : SUCCESS in 7m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9cf46607b64748a198417be0bfef4908 : SUCCESS in 5m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/76bc4cb94d824b3c8044b9e22d2c0c96 : SUCCESS in 5m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a0289990274a4f48b1d29fa73b8dc270 : SUCCESS in 9m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f1b5919ee99740119072502fc40aa4b3 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/50f40420f2c14585a1871acf260edf84 : SUCCESS in 7m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/710e6b1a70b54904a6ab06c42a8a6a93 : SUCCESS in 13m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/80a65a0e1a274a8d826bf04f63a9a339 : SUCCESS in 13m 50s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/521632b641cc457aa6c0c9fcdb1f67ed : SUCCESS in 23m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/99859d517bf644c49a1225c8dfc86945 : SUCCESS in 21m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/509a649790c64fb796c73cca4e2e0a6d : SUCCESS in 14m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2cf7f669d11f43d39d83d56e98dd4441 : SUCCESS in 12m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2f82e7cc467f42a783be85afacda8880 : SUCCESS in 14m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f6bf2ccdcee9475f98c5a2327363c9ef : SUCCESS in 18m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c3b1858e6ce24e08888beb53adca23ea : SUCCESS in 6m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/17fcd3092831439da4302beee1f00fed : SUCCESS in 10m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/92b1a0cdd8474315b789bf26d8ef4aa1 : SUCCESS in 8m 54s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/619f3b88a43845df9d6f431fdeedd339 : SUCCESS in 1h 17m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac9f78f5726d40a285c144349744b44f : SUCCESS in 1h 06m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0aa9318ec2d44a959749b01efa0821be : SUCCESS in 1h 43m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a6880033180344979b89ba321869c46a : SUCCESS in 58m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6bce21db396c53cbb2b7997727f75d2af5329086","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-23 03:17:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dec7216c31d48c0c0897e2fbf6870ee4366fc0d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 03:18:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6dd4960af0d783b73b6f00907ed8ab91485a8d81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:09:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bc28fffcebb4495ad2651e8aa1e571a : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51a2e349126c4af49df52a5fe936a0be : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50c65bbd330e40d7a488d68d5fbc4e64 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/566076019caf4ae9a102a6197944d210 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d5a6975be8c423684ad8f9ee04aeedf : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef3046c506ca4178a6009caecb4ae261 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dee97f1b0d83482c842bf5402e9a72b3 : SUCCESS in 4m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d14d8045f1e341d9836c6c1516d69087 : SUCCESS in 7m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5bc27960820a49649a9cf57cbced785c : SUCCESS in 7m 13s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bf875e4cd0914a2ea81137d5bcc8ab0f : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/99cdefc5fa57469d91c735b0f5409f7e : SUCCESS in 9m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4d8ac3fb4f044c8883c3d54b1cce2305 : SUCCESS in 6m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/359ee488c4d8446cbaa6575536cb6aa5 : SUCCESS in 7m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6821d9b2d1084c2e979e54c8a1856c63 : SUCCESS in 10m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/edef82839d5344a3817e32baf1f10b0b : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/759a994b10304d2d8bbda5888b5b8f56 : SUCCESS in 10m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/65d42089ff0843bb98deca6087705d32 : SUCCESS in 26m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b2b84b94328d4b0aa8a428166ef534a8 : FAILURE in 15m 38s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/20c600320e314b0fba75972aa3472f08 : SUCCESS in 29m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bee68c916aef49dab8d45dbb8e1d96ec : SUCCESS in 28m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b2130dd3b0a54aa7a90a944beffef810 : SUCCESS in 25m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eeb76d5524564fd4ab279986754b7ae8 : SUCCESS in 13m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f84d34e2993f4dad863d49943b9ef73f : SUCCESS in 12m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/10a9e2d6a26044fb9727023e44b96929 : SUCCESS in 19m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/425db2c45e51484b99ce761328984361 : SUCCESS in 6m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/306399d86df04dfa8589f0d0e83c15bd : SUCCESS in 9m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c75bd2410fe348f88a918a25f99d17bd : SUCCESS in 10m 07s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c12c8f33e0b24c079554ba6b78a638a2 : SUCCESS in 1h 05m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f3115ba829b48398dedc010c9353175 : SUCCESS in 1h 15m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3661d4b4e2e647d19eb717a767372297 : SUCCESS in 1h 26m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/58c411a71ec0404887839f062d3dbfba : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":1},{"id":"555e7a8ada84a0ef05e21b0fbd0ca2998239ccc7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-23 15:19:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ddb0e05ec208b32f7c2639a9473b297abbd1cf8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 18:54:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66c4b22f7d241e789a1d2ee8df914da : SUCCESS in 6m 06s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/527b790899504eb5b101ca60262094fb : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0783373c4d5d4625ba7983d306dc955e : SUCCESS in 2m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3dd439bd12ac4023b5e3fdff519c9155 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/512b753e5a674d83a541f6ebbafa4e9f : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/501bf06ab71c44db92036bb1b6c4f9b9 : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cca1fa99c1e34286baaa5c5758830158 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/26d11de1eeb5420f97079b69cc9347b0 : SUCCESS in 4m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9fc92c42119247d8a4cbe97d08fb0031 : SUCCESS in 6m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c75653254cb7435b81850f144b98aefe : SUCCESS in 6m 22s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/15fcc99f2a164d1eba579a6110391e72 : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2ad922579917465c887a015133ed7b88 : SUCCESS in 8m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/74447cff9617456391912607c378ea70 : SUCCESS in 5m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b5a9025abb104fc0ae5bd9cb83e9ceee : SUCCESS in 4m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/12c99ebb108643b397f586c3b7f1923f : SUCCESS in 10m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/932e2024f8da43a4b212a44aa912a03f : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f52fb20a9794e6d825bff4281871498 : SUCCESS in 6m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6a2521448e864a2aa5ffcb72e92ae005 : SUCCESS in 15m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a15da8af1c0b4614a086815f8ea93a8f : SUCCESS in 14m 13s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/dcae6921c87e4286884d450e71054450 : SUCCESS in 23m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cd3700687b6846428ba87d7b5f2717e0 : SUCCESS in 23m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/208c1694a9834dd391d01ce7e9407fb2 : SUCCESS in 16m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3c0a47d98e8949b9be4af182916b9552 : SUCCESS in 15m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/98dc6eee8f85406f9f85d475914a119c : SUCCESS in 19m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9c7861dc0344449ea8a60d51fd8eef80 : SUCCESS in 20m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83ca9090295f4e86bebd6d20a8d38914 : SUCCESS in 6m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8a81995dbe6a46cdb31a8838c88343cd : SUCCESS in 9m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/20e159374a0849838d1c0cbd4dae970f : SUCCESS in 10m 46s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/18a6e8c208f048b392d177b3dc29bd80 : SUCCESS in 1h 35m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/86c2437181284ef398f5ad1859886701 : SUCCESS in 1h 35m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1a1e19da8aa743579a9e3b2db171a271 : SUCCESS in 1h 46m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c6f84ffab53d45d7857a0d2f8a6f4996 : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":1},{"id":"86baeea9aaa4a872b8a2f55bf67806a535182e85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 18:55:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2c6bbdc8cd1e3444d3471e45462ddc6ee5d1dc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 21:26:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aea46e206204c92a6d1a3ffb87823e8 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80bdb8173b804620b5be8458529c1f96 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5df02391ef994fffad385711504cda9e : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a48a4636d0f45129a2f3c1d8956098b : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62f4b5c7c52243c4b00ee2087288bd26 : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80959fc88a574ad896efd4651408d258 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ab175bf990246f59eb48e781647ddee : SUCCESS in 4m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6d5fc5b525c4455089e8682f7fb31576 : SUCCESS in 7m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/505f0bb63cb14ce9891f0380e5987c49 : SUCCESS in 7m 13s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/735f5723dcc347de963db2d57319f27a : SUCCESS in 11m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6c86ad75357748859d8d4cfecde8f706 : SUCCESS in 9m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8683c3f269224f3eb51378623b80c91a : SUCCESS in 4m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/caeebbbb0cab4b848d03dda431c3131f : SUCCESS in 4m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/87bfe8c437b84e47b04baa345bb6c425 : SUCCESS in 10m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/36768e3f27844ab29a3643ce5bfd4532 : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dc979c044d9b472088c7deac2bdc9d00 : SUCCESS in 7m 35s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4fca8345934745d1a25d190d83c50bef : SUCCESS in 14m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/52c4568418ae46eaaefd32feae110aec : SUCCESS in 13m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d16ff19133754c60b04fb7a68b8d99d7 : SUCCESS in 22m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a1479df8711740f49031963445354984 : SUCCESS in 22m 07s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9968e79c5f19465eadcfb5dd7d6ec010 : SUCCESS in 12m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4667715f6c6e40b0b153418e6e827db2 : SUCCESS in 12m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/babb9de8d77841cdaff607cf1834a51e : SUCCESS in 12m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2eb796dad88495eacf05302f78618a2 : SUCCESS in 17m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/536c1affd2f940e3a4a18429747e790f : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/080203ad86a646959277a7cf81dacd6c : SUCCESS in 8m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/05e55e44de6240d79a618358b4da48f5 : SUCCESS in 9m 45s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/38ee1795cf0c4757b457074df1df0120 : SUCCESS in 1h 02m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/55364c1cb22f4743965d618f250ad94e : SUCCESS in 1h 38m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bd0340498a6b4147823de733fdc7da96 : FAILURE in 41m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/30444f09c9a94c529b5ec16cdbccbef2 : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9441010d47f57bb643d31020f7225b95fd8d4983","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-23 21:32:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6b2ef6393195d4f05c6962c8e8030bbaae4a0967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 00:24:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b403c1ae42b453794d268f0ef66b309 : SUCCESS in 3m 02s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9cd111e647c443b98461a6df82f6e26a : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f36f8c63a764057bfd2f98b2ea3a4b0 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67c04e8179b84ac39efcadc2875de418 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e82d0067cf8c4ca895bd57bf81adfc53 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a3f32e71b764304b00f20fb1a37c45e : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19c56a325a2b4e43a3492a12e01245ee : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9d1e52d935c548729aca6d727fccaf11 : SUCCESS in 2m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1b38f62c52be49f4907fb07921ba777b : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3c6ceb9e5e254a74903e03ff10e0b056 : SUCCESS in 5m 50s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8d4314bf67bb426ea8d3f7f45ea7737a : SUCCESS in 10m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/caa19d4422aa4ac180d8fdae75cdc850 : SUCCESS in 10m 09s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6b0a8881e8de425d9ec02b1a6f8c0373 : SUCCESS in 4m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/94b871283a92466593aa875f7432efad : SUCCESS in 4m 55s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/52eaf26c9f134298a970b08e89293382 : SUCCESS in 11m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ce0945e1d288468cb4002714b2e28251 : SUCCESS in 7m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/38206e7fde994385a107e6371fdca4ee : SUCCESS in 7m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bf48ff178721450aab27dc5f2604df4c : SUCCESS in 13m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5c4096e6d16c4f9ab581b53d34b04d13 : SUCCESS in 14m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/62d41c86ea604ed388028df84b58c5f7 : SUCCESS in 27m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/58da16f498f44f3b949fef4dc6f94526 : SUCCESS in 22m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7b7d9601e0e945a7abde5f4fe3566708 : SUCCESS in 16m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e57dbe0c9f2647948596794f05b2bc91 : SUCCESS in 15m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7f21abc16c7847fe940f7b6ec60af941 : SUCCESS in 15m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a2e46a4c9384c1db9c55fa24be8f63c : SUCCESS in 22m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce788bd070db4ca2a8e7f39c9ee20e82 : SUCCESS in 6m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3f941a07b48645599829797edb26818c : SUCCESS in 8m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e738a28340b446f39de24f20f206c12a : SUCCESS in 11m 28s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/607cf4a6d62441f7833e1b8910348350 : SUCCESS in 53m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/db7223cb33d14aac9a99b391a3cb9179 : SUCCESS in 1h 18m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/403317b530f744d4b9bd53251259ee8b : SUCCESS in 1h 34m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f9c358b7606b4bf4b4b10a1739ef390c : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6d39dcf66b792d55f3f2189c124f1103d2697722","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 00:25:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d41909851f1eadcdeff6e9715e557ebd07bb04f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 02:53:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e932917e2347b4aacafeda4d13b634 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/093cd970233d49a085a1a802828db5d9 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de872a656e394de0b7bba2d9dbe038bc : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f769ad0f1924e19a46f96ecc5161aa7 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74b2595b97b84b39be877dd97c2b7aa7 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c23a156228042c1977890b6755044e4 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e5732cf40e72419fa2b329db80bc4096 : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d1221a8cdd9d4821ba7c15a3aeb24da6 : SUCCESS in 7m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cdb36422873749cebae9133f8c477e2c : SUCCESS in 7m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6bd1e34dfcc3442eae05c204a8936f47 : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/22a086bdbe0646f9981a74aeb5439766 : SUCCESS in 9m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/aa7a9e7ad6b6495f9f5791bf3e22485e : SUCCESS in 4m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/809c22334f8e4ffbbe635fc9676a92e4 : SUCCESS in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9fb7f755740b4cf08681d0d51ce14126 : SUCCESS in 10m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/55bab8cb40d6464b99675ce0d4a644b3 : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f64d934e18f4622a8e62698427b901b : SUCCESS in 7m 22s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0aedb7f6d5eb486a96ecfaf3e0c1cf86 : SUCCESS in 21m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/53f29c51a25f41beb21834b8e76e5ab8 : SUCCESS in 15m 21s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/bb24706a55dd480c90ae440791f65bf0 : SUCCESS in 26m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/71650ae339274999a5d3d5527f675a60 : SUCCESS in 23m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/440595a0668a4d08a608f7ca9f3bb36a : SUCCESS in 15m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aa87c5ccd83a43e38e060e84b3d0ee86 : SUCCESS in 12m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0a40cac38bec4d14b4f2b61722ff41a0 : SUCCESS in 14m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/246b1231584848b4a174889d5313cbc3 : SUCCESS in 18m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c4b2ddf0a86c4da19f68fe14b493db8d : SUCCESS in 4m 44s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7d9504f681234cd0a9a09155159985ca : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67b5e3e4ea1d4cab9aa3f7c76c45d49b : SUCCESS in 6m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c2e08ee4742a44f1a9f092b115c64962 : SUCCESS in 9m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f807456e65a4471e831bd5dfdc79055c : SUCCESS in 8m 49s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7787eed2a39c46919e0d5ddd3391484f : SUCCESS in 57m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/672f03a01cce48e2987b3df1bcf1060d : SUCCESS in 1h 13m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6a5a3b2e546846e6859474839241a47e : SUCCESS in 1h 31m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/37d7e4ffbe8a4dbfaf31b6e0e48c2467 : SUCCESS in 51m 11s","accounts_in_message":[],"_revision_number":1},{"id":"599ea6b3d617589387fe1b6319bf1712acfaff7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 02:53:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8832444bf4e9f253a863def07efbe78e6fa267a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 02:54:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c2b324c723a4f1bb5bff73cea8ce7fd : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8bf222712f0cd835fe1ee3be2bb4ec798673e78e","revisions":{"8bf222712f0cd835fe1ee3be2bb4ec798673e78e":{"kind":"REWORK","_number":1,"created":"2019-10-22 21:19:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/690419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/690419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/690419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/690419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/690419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/690419/1"}}},"commit":{"parents":[{"commit":"94e035e36eea0a3ad813596f9d0cc21da5493613","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94e035e36eea0a3ad813596f9d0cc21da5493613"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 21:19:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 21:19:25.000000000","tz":0},"subject":"update constraint for sqlalchemy-migrate to new release 0.13.0","message":"update constraint for sqlalchemy-migrate to new release 0.13.0\n\nChange-Id: I989394e0d31ae69dffa3b455bdea6edf127b18e8\nmeta:version: 0.13.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Commit: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Change-Id: I517880906019ea52f9b3731319abcad55f92915d\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8bf222712f0cd835fe1ee3be2bb4ec798673e78e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8bf222712f0cd835fe1ee3be2bb4ec798673e78e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
