)]}'
{"id":"openstack%2Frequirements~690473","triplet_id":"openstack%2Frequirements~master~Iada18ccf4a422d315fd3c54cddd7bfe219c415ae","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Iada18ccf4a422d315fd3c54cddd7bfe219c415ae","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-10-23 06:14:17.000000000","updated":"2019-10-25 13:50:53.000000000","submitted":"2019-10-25 13:49:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690473-1572011359191-cd6aae0a","meta_rev_id":"3f8ccd77eb365b2cac89d40bc95750a24e28ff2d","_number":690473,"virtual_id_number":690473,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 13:49:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-24 20:10:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-24 20:10:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 16:11:02.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-10-24 20:10:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-25 13:49:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1ed8cf1c06419f52561de3f67d10ac604c735b2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-23 06:14:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6c38354246265661c1d2fd2d62bc0ac754b1bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 09:13:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb25ccd25da4ad587d2bbee6f310fb1 : SUCCESS in 3m 20s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/809b85c047cb4a1ba87e6c0d0b4bc62d : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96616562bd0c4c39bb28a04752e39d9a : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5c7304a54b247e5ba46026679cbae5a : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91b8db4a2db3457ebc949f0fab511ad1 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2888dfe80c6c43198d0a62f274f136d7 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/53277c7d3ab6467a8c2c2b86f5fd3bc2 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2188ac02665c4f9b88229ba81725dd93 : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/aeaf787ed1934465af2aa6c63e386ac8 : FAILURE in 6m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6137468b53064b439c65d6a4c508cc72 : FAILURE in 6m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e35951acd38a4e7495b8178047ef7803 : SUCCESS in 10m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0933b98869004db39a6c2c3eac1baf04 : FAILURE in 4m 10s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/72da92cba86841bbac2caa511b5542ea : FAILURE in 3m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5c309c6f98b44ec69031315232fe90d8 : FAILURE in 5m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4f3ce27843fe4d739770355ebc87d768 : FAILURE in 3m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5da6a8abddac44d0bf9dc388bbbee2c3 : FAILURE in 5m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ca40ee18bc3a4770951f91112fed9e3f : SUCCESS in 7m 36s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5391dac4cd1c4158aaa2123d42c3aadd : FAILURE in 4m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/34156397f3824c7e8d2902cfcbb528c5 : SUCCESS in 25m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3a88d3494c594cb1be107f2592b180a4 : SUCCESS in 21m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6c911a95e0b24caab15ff9c4b668320c : SUCCESS in 21m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/74231c19a34b4d7187cda19a354f86a3 : FAILURE in 6m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/db8e092c40404e60a3fa6953d8a13d1d : FAILURE in 4m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/867b8e1893e84f1d9da7ff9bbbdc2983 : FAILURE in 6m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4566c126903a4acd822d8ca43d384641 : FAILURE in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/518897d026cd45c78cbab358a87e341a : SUCCESS in 7m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/690cb6d11ab44571aede4b06839a2d8b : SUCCESS in 10m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/16e697731a4546c3b943ed3c72ebdd63 : FAILURE in 4m 03s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9e6ea4560bab468baad20178429c9a3f : FAILURE in 20m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/436e6de8e14f41eb911c2419d6d9dd12 : FAILURE in 1h 07m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c72b26c235f842d8b5e57e64d3275c36 : SUCCESS in 1h 35m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/af155429aa954200aed6002583053550 : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7556f7386797545e8f58e9a614616f16cfee3b50","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-23 15:19:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ef4a37e898fda37397fcf9a25ca2aeff36d9e5e7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-23 15:19:58.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"2d5452cf306ff588edc0266ba12de1570c580dac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 19:19:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/526fbd616baa42d98a42b8f095c84384 : SUCCESS in 6m 35s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e8c4d9539a3e4a028f4f780411d99882 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72ad1c1c9fca414ca54fdeb2d930459d : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6bb4cc4ef5414661b618ea2c91587007 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0392c70a3f9540ec89b05b460196d409 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d49a2f248e949e4a5363ed48fe888f8 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4835a9587b00463b891b4bd308a32284 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94423c1f7b4042a9be50afc20fef144a : SUCCESS in 3m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3fc18c147f7b40c886ae6782680ff721 : SUCCESS in 7m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ca230e6704bb4b95b9ba8136d57bd9b8 : SUCCESS in 6m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c3d26059fdb64af88afd001c948cd494 : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/74b0daf9a1ff4999b2394683da6aedac : SUCCESS in 9m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b3ba4fe2b448441eaf20b9ced18f5508 : SUCCESS in 6m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/322ef327cd17497f887f5ba2d8746384 : SUCCESS in 4m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/efa69939e17649e2a14ee7475cbff2d8 : FAILURE in 3m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/92788bcc27bf4c4aab8d2e9a82040e4b : FAILURE in 4m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/743f5229feaf474db73b21215d0295bd : SUCCESS in 6m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/91b0a2656f2749bdb13bd70ae650295a : SUCCESS in 16m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b94060716ac844b1bc729add7665ae2e : SUCCESS in 17m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/48ed86d7f0d54e36886e6db33975468c : SUCCESS in 22m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/18ecc9095461415d989e56d63b24f7de : SUCCESS in 24m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9d1f5147e36747ac9e61ce43514eefc8 : SUCCESS in 15m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d90ff9e031e54ad39bc6b8172fda577e : SUCCESS in 11m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/685cafc28bc74e1ebbe3670aaa155999 : SUCCESS in 12m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3565786de2e8408daeea27e8ef304a0c : SUCCESS in 20m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/921ef642d8414b2eb50324f8bd6cddff : SUCCESS in 5m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/500964654abb4682a27b8cb275b70c30 : SUCCESS in 10m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cbe70f2353c640ecbfb2c90f37f67f8e : SUCCESS in 8m 55s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a59557d7b1934d27bd01b53fe1098e49 : SUCCESS in 1h 02m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ec08f51023c4960ab72b1a1b004d31a : SUCCESS in 1h 22m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fddd2bd4b83e49cc8846f226eed88100 : SUCCESS in 1h 59m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/89c1a17ecb274dcfa3c6c7ea29570e0f : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":2},{"id":"2bfa425f07a579764e2aadfde6f03c4555105e66","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-23 19:54:34.000000000","message":"Patch Set 2:\n\nrecheck\n\nnot my problem?\n\n2019-10-23 17:21:39.336197 | ubuntu-bionic |     from openstack_dashboard.api import rest\n2019-10-23 17:21:39.336369 | ubuntu-bionic |   File \"/home/zuul/src/opendev.org/openstack/horizon/openstack_dashboard/api/__init__.py\", line 34, in \u003cmodule\u003e\n2019-10-23 17:21:39.336455 | ubuntu-bionic |     from openstack_dashboard.api import base\n2019-10-23 17:21:39.336530 | ubuntu-bionic | ImportError: cannot import name base","accounts_in_message":[],"_revision_number":2},{"id":"50cc3a93cd93304b12d8bf3388cbb17e1761c557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 23:14:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0889817a9a42c59c0d280c76ca9301 : SUCCESS in 3m 44s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/83565e49b3e94b95b3f0233beb74ac0b : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a9156ce43b9409c99b9ef350c21cc34 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6da4e68fc0d14dd7bcdf32e62fb1f2bd : SUCCESS in 8m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41a02f0a8302454c848a8201800c64e5 : SUCCESS in 7m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b747856d3832404bbda659cb4699c7a4 : SUCCESS in 9m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9e736bcd4034d44bfb41ae4f1d6eae2 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d1c53f1b8494aeeb64f9df1a07e8f5a : SUCCESS in 4m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8fd8e2e5eb284e0ea3946f5cee30187b : SUCCESS in 13m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/33b8d6cb5b694bd68ffad4a3224195b8 : SUCCESS in 11m 03s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c2e9a6b4a0e84b11809ede1f91dda562 : SUCCESS in 11m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0f98f1125f4444339698f1fb7c5bc5e1 : SUCCESS in 11m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9763b92de97d4ab19d13ef9abdb74add : SUCCESS in 5m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/116c7d5eaf0a4866a68d40ffcca69976 : SUCCESS in 6m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c5eed4901ef24fabbb266d4574638793 : FAILURE in 3m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7da4b34fc4b048c093a3b9c7423e9e81 : FAILURE in 4m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86cf402caa3c47d1bfde0b2c425f5387 : SUCCESS in 15m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/58b9bf2b71b44faa957edd6abf2cff2b : SUCCESS in 19m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/221c185c410242a6a9d18dc8884b2656 : SUCCESS in 16m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9e613127362945cdbdc1200fc49dcd8d : SUCCESS in 29m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/50859a53a67146328a990b479468bfee : SUCCESS in 23m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/fcafb8298356483f8e517e197c01f0a2 : SUCCESS in 15m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7a97c3794ba94fb6b029154997b4a0ba : SUCCESS in 20m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/21973789b86b4c4e9c43be563f067e92 : SUCCESS in 17m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f4b148a1b644739b16013b80c7f704f : SUCCESS in 19m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1532249c25b248018e973521081668ad : SUCCESS in 5m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/39165bf4d6b04cfe93cf325de3a95eed : SUCCESS in 10m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ef84c9e19cc24bb59b05b66e66884fc5 : SUCCESS in 9m 57s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3ea08c5c111d46a69b2bc8c2a8864132 : SUCCESS in 1h 18m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/65fd367c95924ef7b5c4ecf4a27001cb : SUCCESS in 1h 13m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5dc04a18e5ab4931b19da0a9928d2753 : SUCCESS in 1h 56m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/984ccccbdb9748e3ae2e0316be8a5b1a : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":2},{"id":"5aa7d9cfae39a75ee6b2b84b3c88d72b6714746a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-24 15:39:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cee689238b27391f43f78bde5ea8767207e78a97","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-24 16:54:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f9056db510965236863afabc0acf86e66377ff2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 20:09:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643ba1c7455841c590aebae2b0c06243 : SUCCESS in 3m 04s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/88fb48794eb1441b8f6d128531d650c4 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/296ab973e98e415d91150f2df5f32105 : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c04343fd3d646e790165ac228228a11 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53bc3d3a429243c282895fc6b7f2626b : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a222db9065b470e969cbff53a9fbdc1 : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ba22e64c92d4149aafc7c2cbcc03f4b : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/332fd5fd4cd34dcbb433f34d3e817cc6 : SUCCESS in 4m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/843e9bada38640d286a4d8bb8bcbb491 : SUCCESS in 9m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4377ed791daa45a6a0f6b00b97072b94 : SUCCESS in 9m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9366651a2d194037a48ea45b5eb9f050 : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d444b47938a9467c9c83ea4ef97119e7 : SUCCESS in 9m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c1107f54428a4da789bdb8360605ecb4 : SUCCESS in 5m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3916c33746ac4a88b535e7b08e08466a : SUCCESS in 4m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/03b42caab8ff440c976efc9ab47cbb1a : SUCCESS in 11m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2384b5d8d9344e3ba12c1d3a7d24b5e4 : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5beacb3551aa4003b3b9f87e94b925d3 : SUCCESS in 7m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4aa3e99d45924bf3882c5769d5f4d493 : SUCCESS in 13m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/63deb17b32704c4497364ec27b97fa8d : SUCCESS in 14m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3c01ecd4a6aa47d4a2f37c975127758a : SUCCESS in 29m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d61e0e85bdbe4aeba2fc3866bb5cbf98 : SUCCESS in 19m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0378ee95924f45cc8b4435a8d9722323 : SUCCESS in 14m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cb7f391ceadc415f828d5ef5dc594bcc : SUCCESS in 13m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/eadb50e2939743d4943475792b4b0626 : SUCCESS in 17m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b0a6a8824d2242039442b36b8ccd2c0b : SUCCESS in 19m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cffa0fc3b8af449096fa64b3e63f79b4 : SUCCESS in 4m 38s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b16b2f7ff97f4cd387bb60d802960660 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b6cfa15563f469a863b5b3c3347c5f0 : SUCCESS in 6m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7cd7baf580c84eadb168dcca067d731b : SUCCESS in 10m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8d37507105ce4dd0bccdcbf03df7fdb8 : SUCCESS in 9m 44s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/dcbb2d4437ee40d09c8d4a38844c4b03 : SUCCESS in 1h 08m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba1bd95add97434f888a4ae0f73b1159 : SUCCESS in 1h 21m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/98bcb5660ba14762afd682a877119eac : SUCCESS in 1h 36m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d452029fcee748d3b36eeede8f33ac41 : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":3},{"id":"35ed52b1afd78fe4fe16fc8918565abc3141b7f5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-24 20:10:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7960b1e0d4058e329e638fe891eac21e8b05d92c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 20:11:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"96e5b5bf20205a38f36e2fdfc8f481af9f992936","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 22:39:51.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ac8b336b344fc7bffa10181d725735 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f832f2ad6c2479191e6337466d83f10 : SUCCESS in 9m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17514323bfe2441a8c873c6892c6df28 : SUCCESS in 11m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bf577fe6b8c480d81a3d0abd12a4d08 : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0409c5dd6f44412280f742b78ab00bdd : SUCCESS in 10m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c73039036614b88b3f694ab549a9c10 : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/419441fb60194a688615bbd5f00fc63f : SUCCESS in 6m 05s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8b19d8f1fb2848228652f09346b68e9a : SUCCESS in 10m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9aa04a9edfce472aae159eaabb882fbf : SUCCESS in 11m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7ea2bc696fef40868146150ef8e5e294 : SUCCESS in 13m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ef5933bd03b1469b97a52aa764262817 : SUCCESS in 11m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ce9fc332cd8d472e877ccb107a5520b1 : SUCCESS in 6m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/67bbe3ad61b54487a9c06669c9527db1 : SUCCESS in 9m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/980128b2364d461aba23cc3e8dfd9570 : SUCCESS in 13m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fc4d3a9be54c4f4a956a11c06f5f0982 : SUCCESS in 12m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/115d16c2aef5401b92e84e3518d53973 : SUCCESS in 11m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e6275054f73648a5a161933ffcfdfff7 : SUCCESS in 18m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6c1e5310278e47c5ab32f2667af095d7 : SUCCESS in 21m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7b89b9c4396443c08b4b000aa254acd5 : SUCCESS in 29m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d56b33c1fb8c4effbf283d8a9adad2ab : SUCCESS in 20m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dc3ac6a93965425dab66517e786bf2f4 : SUCCESS in 20m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/66bd36f4b9d14a908366b3512cf07205 : SUCCESS in 14m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/57065acb51b14e57b8148f9a61af7434 : SUCCESS in 14m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1fec7a370b004fc9bcc876c5c8e433ac : SUCCESS in 18m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/23b00600185a42cc8e5fe3dabd8e3312 : SUCCESS in 9m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e448d5da71cf4f5e8638816e8a51a1eb : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/688d63d2e3da485db1bcb56e0608889e : SUCCESS in 8m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8b58b2b44c1847cabfcb651bd7f17c45 : SUCCESS in 11m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cdd230409c774087bca278c55ce8e2af : SUCCESS in 12m 17s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ef9543d187a74606832aab42db99f060 : SUCCESS in 1h 01m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/39869e7e87b142a191c3192d0ae819b1 : SUCCESS in 1h 10m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3d3db93edef145a892de6827191bd202 : SUCCESS in 1h 54m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1a27144a5f1a4bcba39168854644c8a4 : FAILURE in 39m 02s","accounts_in_message":[],"_revision_number":3},{"id":"963f9dc63ca61d3bb2558271f4d2d122204196a9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-24 22:43:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1f05f44cb474cbc6a37adccc163fc347c1aa4ba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 00:19:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a427608f9434f15ab7d1bc63f728cc5 : SUCCESS in 4m 33s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/54d383f6e3164e32a46ca8b61bf096bb : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be1b1f824eb4d8d976294f30cdf7156 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b0ca0dc1ced43eda3b0bc38cbd540cd : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/440dc6ac958e4a19a71b56b87dd40ef8 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2f9b9a0cda840608995b929742cc03f : SUCCESS in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c7ddb5901ca945f88b76fa01993118fe : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/afd076350b6743aaa517d7e4ff9c9104 : SUCCESS in 4m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1ac4b502f426458bb0f00adce36e7f54 : SUCCESS in 9m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f99584dd39154895b9eb06272d72b54c : SUCCESS in 6m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/57ff0804d4b540e3af92ee435563aa70 : SUCCESS in 10m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c087fded680646faa774e5937d8e9e9e : SUCCESS in 8m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/eb4616767abb4d90892764ce0a2deecb : SUCCESS in 5m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fb695b5be1424448a26d747053c53aa0 : SUCCESS in 6m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0189dd85a05d4909bcfcb7a269fd3171 : SUCCESS in 9m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c09b44e51d8e4c6ea1ca388966dbe7ba : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83db71ca62484ea5873bd3efc0e8d4f2 : SUCCESS in 7m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9ec424dcc2cf4217bf8c73ce19d237a7 : SUCCESS in 15m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aad206de80eb4e839d038f8a710fedf0 : SUCCESS in 22m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/95235c03e77b4350b8f2ebe0330dfb62 : SUCCESS in 20m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/aad436cfeb8b4f5e8487e168730cc042 : SUCCESS in 27m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/aed13fa2fd254c08abf78bc5c2bc1d9b : SUCCESS in 13m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/07506fa174b44501bd20ed78d4556f5d : SUCCESS in 14m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9a05b880d34545669d95190e446d8626 : SUCCESS in 16m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/30dfae8f3ff040ea822a0a971cf008af : SUCCESS in 17m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8b4fdb9e0bda4f5aaf463a273046c2c6 : SUCCESS in 4m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d39061bbe6ab4c7a9e4f437a12542d19 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2434b83639234f118d2eafdbd15e0f18 : SUCCESS in 6m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/18b32b241d6b4a9cbbca0725c5a1dcee : SUCCESS in 9m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f28a5d849bce45739793d8a7dcebee0a : SUCCESS in 12m 00s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3e13993aa9464c5898d6322d326c88a1 : SUCCESS in 57m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2a06564baa04c08921ac417e61e552b : FAILURE in 1h 22m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/372491e1855f4d428c973b7377a47644 : SUCCESS in 1h 30m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/47303841cf444d1fba6d2317f063b260 : SUCCESS in 54m 45s","accounts_in_message":[],"_revision_number":3},{"id":"ca4f2fbd6338db16e7d75167894deec9202fd340","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-25 02:53:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b3fbed8d55ef7729bc8e23120bb503552a9e494e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 04:31:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9336948a2f4e31b696ab9cf042e58c : SUCCESS in 3m 17s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d0188c9aa3c74c33a12ddbb7dcffeaf4 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88be5bff73744bbb854baaab5ab31ce6 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ecdcb615a5b423eb273b4286d38f323 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5208994d906640629a00cd533b329489 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa9ba8298f584927aa862df9cf7549ff : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/057154277e8f400da9fa6b1f7295e25e : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/865da53f2e05471ba03516f00b619418 : SUCCESS in 3m 55s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/14d404ef07514de3b3209ff002e8d7eb : SUCCESS in 8m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/841cbb3c3df943c58ac3bc69488f9ebf : SUCCESS in 10m 31s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b4eb0d3ebb54467da7734d052859cabe : SUCCESS in 10m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cd0fccff012a4853a6e7f24660cd26ee : SUCCESS in 9m 10s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7bc50c06a9dd4ee9982ae28890a27ae6 : SUCCESS in 6m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/94c6a7369c8e4ae0ae962af634665f86 : SUCCESS in 4m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/83e1a4c2ee224e3290bea3ba3f02fd38 : SUCCESS in 11m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c75971aa19334f159e21e497f48c8af4 : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f63e5936e4fc4791bc8535729f20204d : SUCCESS in 7m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9e1867fe938d4868a2c3b6c705c42f4d : SUCCESS in 15m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/09a1e661ff3a4016927fe1d46481b14b : SUCCESS in 25m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/68460617765e4dadae7abf953c12c149 : SUCCESS in 22m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eafed6b252b94876a84173e11162e221 : SUCCESS in 23m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8923d73e992948e589c69084b64b72f4 : SUCCESS in 14m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/55ce1007bc7243f39a8b824aee5181e6 : SUCCESS in 17m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d98dcee6a8124b2db0758e7b41144748 : SUCCESS in 14m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa93200312c74765a6df0216da432ec9 : SUCCESS in 20m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f71fdd0b2df0422383b89f2ee8c1cb0a : SUCCESS in 4m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6e780ce12dc0467d9f2c7dfe714c71af : SUCCESS in 3m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4df2befc54134f14a1da317643dc1f82 : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a02329e2406043d387be47bfe5fa4414 : SUCCESS in 11m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/232b2e3cad04429dbb2e923130d4296f : SUCCESS in 10m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f790e7d09a194dd0b3fef03c26da0df6 : SUCCESS in 48m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b981455357a4e9bae554431720b8483 : SUCCESS in 1h 06m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b4e1798b6f4747b1afdb9e3f38b77e73 : POST_FAILURE in 1h 35m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2555e6b33e0b494f8acf86a69c18bb78 : SUCCESS in 1h 04m 10s","accounts_in_message":[],"_revision_number":3},{"id":"3f27b5d5acadd5b8728fe3363fa9c54d715304a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-25 04:40:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8e8ef8e11e90578f20715c482d3aaa95e2fc04b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 06:14:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/329a465e1f414b72945d320b4312108c : SUCCESS in 3m 35s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/34d02303e9024631addf0a64133f00ed : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a88c8573f44b749827c822f8484d22 : SUCCESS in 2m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d49001edc3714192935e376d9332b973 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c3e8721961040ecbf43daecdbff7b9d : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6fdb84d7ed946e4bbacf1ac4b8e4ab7 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5af85502f48a4981a53628baacbf7282 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17545221a1b745cdaf1ecbe9e138a744 : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b38b223513634e4a9bd1268aa9efa628 : SUCCESS in 7m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6c2ebd952f6740ab96b7d5a74c164625 : SUCCESS in 7m 09s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f778414f587644cfb23ad2567267ad27 : SUCCESS in 9m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5ff727d9a5af4dcba2bcc4a67cc39b1c : SUCCESS in 8m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/14944893a8b04527aa9cd4bf3c59bfcb : SUCCESS in 4m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/244572528a1047c0b1b77111f31a7e7f : SUCCESS in 4m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5651217f2e204e09bebb346434c79814 : SUCCESS in 10m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4a73dee5c20c4b5886950f0e76e3422d : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d38739ecca34187a927ae45cfe229ff : SUCCESS in 7m 02s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cd91458b95514e80a8115a15b46f189f : SUCCESS in 13m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a1df1ca6bafc48f593dc3e2d4de0836d : SUCCESS in 19m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1c361398809d4fa8a21ad1be31e1ee00 : SUCCESS in 24m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6c168e9f7e5f4c7095cc422dc659499f : SUCCESS in 19m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c5b3dda12738469da69a0583599bcddd : SUCCESS in 13m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1ae5c886807e4ffca331f4852cae6b93 : SUCCESS in 12m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5c813a92f4874f2798ae6fef4ae5f528 : SUCCESS in 12m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f182f5f15e3492db2aa943ed857e9d6 : SUCCESS in 18m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/88b7830f463042fcbd525a3b43719e78 : SUCCESS in 4m 01s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d59ea13e7dc04d68bcd7bbd99bbdb2f8 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5f05714548346daa002a4712b7e705f : SUCCESS in 5m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e2286b5126354e92b18c31fc4bf96778 : SUCCESS in 10m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7131ef5845af44e496644d0ae3550bdd : SUCCESS in 9m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/243583d54d634972bf13e86b226775dc : SUCCESS in 55m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/930fea87e2e4405c90de3cd963d9d39a : SUCCESS in 1h 12m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0d89374729df4887b5e5173201851f86 : SUCCESS in 1h 32m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d466748f81fe4de1882c607fd8b6ca55 : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":3},{"id":"c97e4127cdc898a51e86cd0ed6a25d4315d44051","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 06:14:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"86f99658765b0e64ce448927b4f2db22bd24e946","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:49:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/636c08f5781b4e37b1a1853115d05789 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19eabf6dbd7a41d5b2bc3917e4b472cc : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce8e141216fc4236af19d868f09538f1 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/556fa445006547bcbe069d3ca210258a : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3813fb26e32f4b5a99bcd445a5227c36 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09657b397130467498e09c4fb02e5d08 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/74454b24112e4502bfe29984f5d6bb94 : SUCCESS in 5m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/003c65bf74384c44abd9deb819087856 : SUCCESS in 14m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5b15a92dd3ad4fb5a8eefd4314b04013 : SUCCESS in 6m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b0dae1ac8a3a411a8a65e542aae8c871 : SUCCESS in 11m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9e2667a9bf9642c4b0827ca98a95042f : SUCCESS in 11m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/32d5b816f869491998903184bb75211a : SUCCESS in 5m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/75db5b67801b4c6db292e59917590e37 : SUCCESS in 5m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/46f7fed74c49410b8015bd78463ef28a : SUCCESS in 10m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6ac39f55e9e44e9c8adde7bd8ada6411 : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99c2d459cf9e41488ca69b039ddcd5e8 : SUCCESS in 7m 43s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2ebf3c7e96274ef69e2d613504e51a48 : SUCCESS in 16m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/762d273f37114fb78f50d995897e09ea : SUCCESS in 16m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/06c95b085ba8494e9e75066d871c8d00 : SUCCESS in 28m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/57fb8f2c3a74495daf4e5cae558b9c64 : SUCCESS in 27m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/273efa0c22524e6b89384b5efc07f718 : SUCCESS in 14m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f795027c1bfd4cafb86ba9c5a384fb18 : SUCCESS in 13m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c77e8db45421475a9ffd6b189bdd29bc : SUCCESS in 15m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8d815e3450674e0aa05738c6ed1f3be3 : SUCCESS in 17m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2bb9e1745afd4ed8b95703ca67e56b29 : SUCCESS in 6m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/eafd9333a8d6469f82337d37b26d529b : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e1f393aa38bf426ba911e673cc62c606 : SUCCESS in 6m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7d49896c897a4ba0b036715d539e2ece : SUCCESS in 9m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5569b37f3d8e4c86a2d96ace6b87aa62 : SUCCESS in 9m 59s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3bfd8e7db2aa48d199d458afd2d21ff1 : SUCCESS in 1h 04m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ad42b2a8ea34ee98f5fbcfe8f023485 : SUCCESS in 1h 22m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/10844147fe6d4e5598d8bb4c5c5515f0 : SUCCESS in 1h 44m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/287b16f508464d138d2e95428eff4662 : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":3},{"id":"aceac13ded0401fb9cd293dcd88a2af4835c6eb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:49:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3f8ccd77eb365b2cac89d40bc95750a24e28ff2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:50:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aeb652ead8249a9986857574a714fa5 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2bc848d92fe57a072923132146af0f2453077aea","revisions":{"26bf24931b80c67d1d244566831559833a28dab9":{"kind":"REWORK","_number":1,"created":"2019-10-23 06:14:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/690473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/690473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/690473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/690473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/690473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/690473/1"}}},"commit":{"parents":[{"commit":"94e035e36eea0a3ad813596f9d0cc21da5493613","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94e035e36eea0a3ad813596f9d0cc21da5493613"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-23 06:14:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-23 06:14:15.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iada18ccf4a422d315fd3c54cddd7bfe219c415ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26bf24931b80c67d1d244566831559833a28dab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26bf24931b80c67d1d244566831559833a28dab9"}]},"branch":"refs/heads/master"},"e672341dcfa1bc31653371500bf01f844bb7aaa3":{"kind":"REWORK","_number":2,"created":"2019-10-23 15:19:52.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/73/690473/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/690473/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/690473/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/690473/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/690473/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/690473/2"}}},"commit":{"parents":[{"commit":"2bd26072fe8f0b2aa6f0651c7e0702d053128441","subject":"Merge \"update constraint for keystoneauth1 to new release 3.18.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bd26072fe8f0b2aa6f0651c7e0702d053128441"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-23 06:14:15.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-23 15:19:43.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iada18ccf4a422d315fd3c54cddd7bfe219c415ae\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e672341dcfa1bc31653371500bf01f844bb7aaa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e672341dcfa1bc31653371500bf01f844bb7aaa3"}]},"branch":"refs/heads/master"},"2bc848d92fe57a072923132146af0f2453077aea":{"kind":"REWORK","_number":3,"created":"2019-10-24 16:54:39.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/73/690473/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/690473/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/690473/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/690473/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/690473/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/690473/3"}}},"commit":{"parents":[{"commit":"2bd26072fe8f0b2aa6f0651c7e0702d053128441","subject":"Merge \"update constraint for keystoneauth1 to new release 3.18.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bd26072fe8f0b2aa6f0651c7e0702d053128441"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-23 06:14:15.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-24 16:53:03.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iada18ccf4a422d315fd3c54cddd7bfe219c415ae\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bc848d92fe57a072923132146af0f2453077aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bc848d92fe57a072923132146af0f2453077aea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
