)]}'
{"id":"openstack%2Frequirements~690986","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I5585bb3851a0c3ac25e9e6ac8df115335215364a","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I5585bb3851a0c3ac25e9e6ac8df115335215364a","subject":"update constraint for os-brick to new release 2.3.9","status":"MERGED","created":"2019-10-24 14:43:51.000000000","updated":"2019-10-24 22:39:53.000000000","submitted":"2019-10-24 22:39:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690986-1571956793922-b8c1e99e","meta_rev_id":"fd9f2ab1d0d2f7a2e2adf05186d8eac8f602d7cd","_number":690986,"virtual_id_number":690986,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 22:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-24 19:06:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-24 20:11:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-24 20:11:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 19:06:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-24 20:11:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-24 22:39:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b20d3db520c71ad826e51c3acb90f6e42f66d36","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-24 14:43:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b184e8e231a2a59aaca1fccd56eeb3229ef1c82","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-24 19:06:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e5007b1e658ff0ffa57479e605e5e9218b538533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:34:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9307ed2042b54b9a87bc5d62ef1266ff : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef4697dcfabd4b078e4c83286d69fb3d : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04a661579830463199e5349920875820 : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6341dbe809ae411596eebe3c2e89a77a : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9cef3417e2bd438ab3bcf047d0027521 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75be154025cd4c838808fbec5dfdcc3d : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4fb3c042986e4b988b9ef8285e75dd83 : SUCCESS in 7m 02s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/c1fd19cdefb5498a874d43fc17ba5b38 : SUCCESS in 9m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0d60df92628e493fa46ef026b203b38c : SUCCESS in 9m 30s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/75964fa5bda3423090272502b3458d84 : SUCCESS in 8m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8f69ec4fcceb4d8d8e40c241b7c22734 : SUCCESS in 6m 40s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/04bf360b81894ae59b848c2901efa2b6 : SUCCESS in 6m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b683c46b3fc545228873da58383807c6 : SUCCESS in 11m 13s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/7d88c55971b3487da405e6fafe9360cb : SUCCESS in 10m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bdb65f34f50748d4939663be00990ed6 : SUCCESS in 20m 23s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/ebd3a601242342d69fbde4ef812af5a3 : SUCCESS in 17m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1244c27e2ace4df6bd32de685a00ee5c : SUCCESS in 23m 52s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/d14fb95365ce4ff3a84b8e1ab5eba778 : SUCCESS in 26m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c8e32fcb70ca45b495f2717f22a6d28f : SUCCESS in 13m 28s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/68c5299b6a4040d2bd03ef3ee9a09ebc : SUCCESS in 13m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7485d13b57304552bcb1933b81111e38 : SUCCESS in 23m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cccbbcb7f01f4808bccfaff5a2d392ff : SUCCESS in 6m 38s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/a0940b0694cb4ace8c797fa4c27a13b6 : SUCCESS in 6m 43s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9fc369d65d464dd4a8b6cbeb39671555 : SUCCESS in 1h 03m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c10766cc2ef411fa088d449b561e1d4 : SUCCESS in 1h 03m 26s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/4f79344ec0f3459fb1551afb353e18b7 : SUCCESS in 1h 17m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d027e5121fc84329bba30d13a0f38d18 : SUCCESS in 53m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d0c9ba5e5339892a9447dc6d0d25494265ae4eae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-24 20:11:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"739baa90de3a08fdb42686e8a7bba91e682e02cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 20:12:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"440c3351b0564f0895f705ed4d23f34dd90b3054","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 22:39:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de2506dff2b4c8eb7faf5ffa90829a0 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c129636065ea4ffe8c4a3514a8eacb0d : SUCCESS in 4m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4d4034387fbf4843ae7aea8d516849ec : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a7d41f21584467a8782e4f9cfe27c7e : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/364fe3e90c3b485387524c408cc2ad05 : SUCCESS in 5m 17s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b5cb5c01db7b453cab33a8c9777c53df : SUCCESS in 8m 04s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/6b043952fc2f42de9a2e268390044219 : SUCCESS in 6m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7f8fe12497f1419284c2311e62f410bf : SUCCESS in 10m 43s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/7b325cd1e03049fc84b043f4387cc50b : SUCCESS in 8m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/36927adb4ad24ef3a5561a746c82a59d : SUCCESS in 5m 28s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/a4eddaf363804d0185092467b1e69d3e : SUCCESS in 6m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8e829c7928904319b3715b541e5a28b1 : SUCCESS in 11m 38s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a07a758d2e0244918f8f80dfeb1f8a32 : SUCCESS in 11m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b8d2213e2ba740af855267ef5d0bff32 : SUCCESS in 13m 09s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/8aca8979aa494701a4fc12ef081f3eec : SUCCESS in 13m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5f3d8e0cd2fe4c328c905555a608f307 : SUCCESS in 25m 04s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/29bb3a9a38274d7ca6fae155f44948ec : SUCCESS in 27m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/69bbf603210f4c30beeb0a2786b84ea4 : SUCCESS in 14m 21s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/ec42b9f4a9d345ef8a20c3561af255f8 : SUCCESS in 15m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf3f37a1f48e47adac820255c1d0d9f3 : SUCCESS in 20m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a06caaed5fa34646998b50e87b6dd510 : SUCCESS in 6m 34s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/1babfd7fe4ca4816a9c7368cfb04312f : SUCCESS in 5m 30s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/242230c23a1f4312971a5450d8673c18 : SUCCESS in 58m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8950817acdfc4589b11f8068dec65741 : SUCCESS in 1h 13m 28s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/7ad0c06a128c41358a10edb3a47c4ab7 : SUCCESS in 1h 32m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ba9cff6bc9cf4984aefe012d09f53964 : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":1},{"id":"6f5c5e706d3ecc4ce11af0948b4e6aaeb530e252","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 22:39:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"464849fd217fb490abbbc85b25de405f094c3649","revisions":{"464849fd217fb490abbbc85b25de405f094c3649":{"kind":"REWORK","_number":1,"created":"2019-10-24 14:43:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/690986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/690986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/690986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/690986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/690986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/690986/1"}}},"commit":{"parents":[{"commit":"1ac97d8cecc96f4a98c53328a2b389db9b673e61","subject":"Merge \"update constraint for python-monascaclient to new release 1.10.1\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ac97d8cecc96f4a98c53328a2b389db9b673e61"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-24 14:43:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-24 14:43:49.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.3.9","message":"update constraint for os-brick to new release 2.3.9\n\nChange-Id: I5585bb3851a0c3ac25e9e6ac8df115335215364a\nmeta:version: 2.3.9\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ib830c037549cad0075f5645f9311755ac93d3452\nmeta:release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/464849fd217fb490abbbc85b25de405f094c3649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/464849fd217fb490abbbc85b25de405f094c3649"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
