)]}'
{"id":"openstack%2Frequirements~690994","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I76b76bfd4219b435eadf1d1663f57e07cdd97800","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I76b76bfd4219b435eadf1d1663f57e07cdd97800","subject":"update constraint for ironic-lib to new release 2.12.4","status":"MERGED","created":"2019-10-24 14:50:42.000000000","updated":"2019-10-25 13:49:17.000000000","submitted":"2019-10-25 13:49:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690994-1572011357886-b8c1e99e","meta_rev_id":"32b1e060d01b2fcac546af980051d11ee08103ac","_number":690994,"virtual_id_number":690994,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 13:49:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-24 19:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-24 20:11:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-24 20:11:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 19:06:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-24 20:11:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-25 13:49:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4056b39982c1e2dc6d39c12bc3227131919593cf","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-24 14:50:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9cc1a9176f580c2f9c75190b557d73fd652dad4e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-24 19:06:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3cd1c32d177a414991693107f1160d2a2dc329bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:47:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e382b13daa184e358b10a0dfa0095d66 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ba136bdfe74dcab6f5e5e6ea4089ee : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e6c5e36cf8244e2a083d1e710705376 : SUCCESS in 3m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fcce9965c47b40a5b64e023d56ec31cb : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ac1f0abd89a4068a0a3cefa99d6cd6f : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/84533f4cfeb24fcba35842474629c6c6 : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/843fc045a7ec4218ada9f00b557d7890 : SUCCESS in 8m 35s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/b5546ebded8e4877a5ac89b98ad0d5a7 : SUCCESS in 7m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/adaa19c2228b4223bbd782ed52986848 : SUCCESS in 10m 30s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/681e8de19fba4696801e6a67da4bdf75 : SUCCESS in 8m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/80be276288ea42b291c873a7858b4dd3 : SUCCESS in 5m 32s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/ebbe0191893949e8b82f5613add663f0 : SUCCESS in 6m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/aef2216834e5460e8588cd5f8c2f13b4 : SUCCESS in 12m 08s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/190c2740f4cd4d3d978db10f5b255248 : SUCCESS in 9m 58s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a143319b5195444694a9dd79b7811ec9 : SUCCESS in 14m 47s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/981e008fcb504b8eb185a784a0499dd0 : SUCCESS in 13m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6b924ac2649c4929ac0103279293b04f : SUCCESS in 22m 40s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/0588dcd2723d4fa8822c46fcc65c200b : SUCCESS in 23m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3942a7dd20ef4bf89099246da7b5a74a : SUCCESS in 14m 07s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/bfef246beb0c46d49ab0ba539aa2ff4b : SUCCESS in 15m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac7be140c2c64db0974e672496d07d31 : SUCCESS in 18m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/195b3a41b0704dffb78260eba88b7693 : SUCCESS in 5m 33s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/eadc41f5192344c09e9ba0e8c894621f : SUCCESS in 4m 17s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/91a34d5049fb405ebe068ac5b398a73b : SUCCESS in 1h 06m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2feb0a2e3d1740219b3e11044c04de96 : SUCCESS in 1h 02m 01s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/b1577f78587341a0a8a63373b68c5403 : SUCCESS in 1h 31m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e568ddb28be3412694c4255e344c2069 : SUCCESS in 57m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d3e9278d76c6a1ac357478bdbe845cc4a3c94167","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-24 20:11:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7264cdd89227d85090eb5a0aca89d7e1eabf628a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 20:12:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9a70754c2d3df113b1b71402d98296a8b970659","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 23:56:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfeb2bb4dcb14fe9b40d56fda0ac8ae6 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ab32bc98b414151baca4bc9cf2da5fc : SUCCESS in 4m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/88375e132c794639a5e302670367ea51 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8615a882e1142ffad35802f6c696193 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c5d421721a0549aa84fb5bb31bfc5343 : SUCCESS in 3m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8898032cbe0c443d8eec946f6241477c : SUCCESS in 7m 21s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d7c1f8605c464ec58cf707bb468f4001 : SUCCESS in 7m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ee61f1cbe70444f0a65b6f4df78cccd7 : SUCCESS in 8m 35s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/d5d1fab4d79849b680d9eadfe915ef5f : SUCCESS in 9m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/66dccb9c940f4fcab3a67628e8a27211 : SUCCESS in 6m 17s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/085652b32a86485e8499cea0eda8eade : SUCCESS in 6m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7f7de8d0b5bd42759d9434f0aabeefb0 : SUCCESS in 10m 33s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/9799beab6089483da7c500b9b75781f1 : SUCCESS in 9m 58s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1617d1924b764fdcabe299fca707936c : SUCCESS in 15m 25s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/22ad9b63f42b42e18b2f4684f6aff05e : SUCCESS in 16m 55s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/913926a86b8b4e9b8482d2b22b0dbe5e : SUCCESS in 22m 07s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5065875967c64f6883503bd21a518d0f : SUCCESS in 27m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/089730431b4e4db28f04f6b930e0f1fd : SUCCESS in 14m 20s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/5cebf81c29204ce2b570a699f53b5db0 : SUCCESS in 13m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ecd8759f09f7458fa9580b3e31b00bc5 : SUCCESS in 21m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0cb306e9921b46be9d27a881e2f26ebe : SUCCESS in 7m 05s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/a1403fa55ed54c228618b8d4a939dd4e : SUCCESS in 5m 06s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c9c1a039863d47faa1a5693e7854d921 : SUCCESS in 1h 03m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/67b330d64d3c4b56ad5e9c0cff55cec9 : FAILURE in 2h 55m 31s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/cfc86c5ffbc246d18adf0240bfd63743 : SUCCESS in 1h 23m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e849d96919f743e5a8db8eeee1c130a0 : SUCCESS in 1h 03m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9dc032f6abf53e20c7d824cf745833c1df436b98","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-25 02:53:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2d24ce67923b1f13ffc33d29c8de0afafd4ec456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 04:20:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f42bcc13ad5242feb7f08a3d16f42216 : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b547bab896a54a78b482c52928cadc1c : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/421673378af6490ba4a3626326adb8a5 : SUCCESS in 3m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/016f5fc7f414441785017120435b74a2 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81989e15df8c47fc9a42b84bc2ee6b37 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e81a3b56354c4ac792c7e4a257cedcb6 : SUCCESS in 3m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/202afc7019d145c08d32f4fbe452b0a9 : SUCCESS in 6m 11s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/87db158631d14523a055fbf4083d54b3 : SUCCESS in 6m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/96380355f3bd43799d78a8f64af20382 : SUCCESS in 10m 10s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/1b5b58c60b8a4fc19355b6fba910915d : SUCCESS in 9m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/842a33d4e67a468f99c43888ef147163 : SUCCESS in 4m 38s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2aa43d77959840f28b3d711d5a6805e8 : SUCCESS in 5m 30s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d7bc8165bb02457bbf0fb3bab786b4ea : SUCCESS in 8m 59s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/56f8c7319f91406b82616c77ed63e905 : SUCCESS in 9m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4f2e49cc0e2e4944bef3e0e589ab3293 : SUCCESS in 18m 12s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/158765e3211546e0bb3ff450abe429ae : SUCCESS in 18m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4e860756b62947969798153cf008883a : SUCCESS in 26m 47s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/a1d89b3816bb46b68e2f75fed3e91dbb : SUCCESS in 28m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/39bd6acbbedd4480b5e0a4ef491a0cba : SUCCESS in 14m 25s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/8253c36ba70a4fbc81d0d0efd5c7dfa2 : SUCCESS in 15m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86ebf93744dc49289c885373d13a7c94 : SUCCESS in 18m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b5f1945c606e444687c36a054a433d87 : SUCCESS in 6m 48s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/9f3ad4cafeed4f518eef25a60301f0e7 : SUCCESS in 3m 56s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a4f75157d5bf406aa215d50f731c0f38 : SUCCESS in 51m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e023092ff47402fa71fe7b98a6bed28 : SUCCESS in 1h 01m 33s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/93336ef5311c4c3fabbd91cae63b819e : SUCCESS in 1h 24m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f1c15a833f574b73b725def1bb1e09d7 : SUCCESS in 53m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1ff861bdded6a32184a9c015d544ddc499910e97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 04:21:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d44d1eef05976f851f3d3fd33d0d5ebf2fb5ab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:49:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cfd9f681b5f4cc5ab218a926d8d0f3e : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1e738f4cefd4813bd5af1bba26b76de : SUCCESS in 3m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8378b717f892461bb06c39651e0dac22 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c8d044d8e484ac28f15126cfe54e630 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b594ce8e14aa4db4bb7eb0cace8df1a5 : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9e82a03915aa44389f64f281c400e272 : SUCCESS in 6m 03s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/5535eb82dd554618b47ae15eddc63a80 : SUCCESS in 7m 15s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/33b65209ca414fa3bade0f0f3a8ce977 : SUCCESS in 8m 40s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/4840f0876ae74a968a4277b6da8d9581 : SUCCESS in 8m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2b495817b48e46338802154ead81f3df : SUCCESS in 5m 46s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/72a0782fc050482886e5c05704bf6028 : SUCCESS in 5m 24s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b644b24b9f3d4f23aa0ab89fdfe05dbf : SUCCESS in 11m 14s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/2dc183f80528471db6f7657220aa3bed : SUCCESS in 10m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/516442a1dc204748b662e67f78f94e28 : SUCCESS in 13m 44s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/f3dbcef9469b45978dc6357b565a598a : SUCCESS in 11m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5a212b34167047709caee6e1a2cc3f43 : SUCCESS in 31m 02s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/ab127e3b5a034b5b898edc43582cf226 : SUCCESS in 28m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1d4e3b663f164e188f15f09d19463862 : SUCCESS in 13m 37s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/498726d6e3994bb2bc748d3154fcba76 : SUCCESS in 13m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7baaab9f803c4694b138c322d8bde0aa : SUCCESS in 19m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7cf622ca3f1a43bba1fbdb62114dbcbf : SUCCESS in 5m 39s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/4818ea8f433849a8be2beba8ed291d11 : SUCCESS in 3m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/857e8bef991641f0bf43305c91903116 : SUCCESS in 58m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2fde021bb8ce442e9c82cac95199fe98 : SUCCESS in 1h 10m 52s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/1d514546cec14bef8f6a72cc3b318ecd : SUCCESS in 1h 34m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f531d3b4b6e142178aa02d9da1ef6ebb : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1097537498bbc6ed483ec66107b78c93e4467714","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:49:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54074166a517c7a2546e1c2cc77db96477a1d4b8","revisions":{"54074166a517c7a2546e1c2cc77db96477a1d4b8":{"kind":"REWORK","_number":1,"created":"2019-10-24 14:50:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/690994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/690994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/690994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/690994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/690994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/690994/1"}}},"commit":{"parents":[{"commit":"1ac97d8cecc96f4a98c53328a2b389db9b673e61","subject":"Merge \"update constraint for python-monascaclient to new release 1.10.1\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ac97d8cecc96f4a98c53328a2b389db9b673e61"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-24 14:50:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-24 14:50:41.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 2.12.4","message":"update constraint for ironic-lib to new release 2.12.4\n\nChange-Id: I76b76bfd4219b435eadf1d1663f57e07cdd97800\nmeta:version: 2.12.4\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Commit: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Change-Id: Id02ec0a24ca34523fd9b177cf9381336d469fd29\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/54074166a517c7a2546e1c2cc77db96477a1d4b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/54074166a517c7a2546e1c2cc77db96477a1d4b8"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
