)]}'
{"id":"openstack%2Frequirements~691141","triplet_id":"openstack%2Frequirements~master~I7fc8469bee078effe7dade50b8fb89356c880a78","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7fc8469bee078effe7dade50b8fb89356c880a78","subject":"update constraint for openstacksdk to new release 0.37.0","status":"MERGED","created":"2019-10-25 01:53:46.000000000","updated":"2019-10-25 18:23:42.000000000","submitted":"2019-10-25 18:21:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691141-1572027700836-edb63827","meta_rev_id":"6de9b21a69080c8d03b6f19ed14d049fcf5d0488","_number":691141,"virtual_id_number":691141,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 18:21:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-25 01:57:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-25 15:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-25 15:00:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 01:57:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-25 15:00:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-25 18:21:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aff614d548ecfdbd0372d34ae0828f8afdee22a9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-25 01:53:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"132ef715460b701773a81c3780216dd7b832e22e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 01:57:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6d1e65c7b48b54c04bf7db70c96de3cc7b806b7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 03:39:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bbefd5ed4564da0b3f70e731294df1b : SUCCESS in 3m 56s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a978673f87ca4aad9458bca4ac4d2a29 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/502f7ec91eba4f8ba8a0bf6e3226a9a8 : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/855e5f9cc3d447b59edb8416f91d595c : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d5837e78a904f5d88c3bb7a9d46b654 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db901d74c8dc44fa935aeee5b02cec06 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffd052d20d8d41c38d5402b7ecd79985 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c44ad030033d4e5695279e6079475236 : SUCCESS in 2m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b3abd547ef6f4f44adb29b95f16e1267 : FAILURE in 3m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6208ad33a43b4595946b9c8b7c7ee737 : FAILURE in 3m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/39db2aaaa19841169e77bca3bf56ccd4 : FAILURE in 3m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/35f2a02f23aa4a38bd1d566ab251b799 : SUCCESS in 9m 21s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ecfd75397ae34977b58ad2bfe15225de : SUCCESS in 4m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2141dc24479c476c9094c731d0e1a78a : SUCCESS in 4m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/365f086b62884b13a08ab4820e51cf96 : SUCCESS in 10m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3f6333b8f88d44b7bfccb6555f19945f : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2116399e11849ecb4a998f4830386a1 : SUCCESS in 7m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/eaaa61b25a984942b09dda0cef01d4fd : SUCCESS in 14m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/418548b54c4044d3904bbb3b27c43202 : SUCCESS in 12m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f0e88a8600a5400fac7d7197fa57c11b : SUCCESS in 28m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a8f5f0cb6f134480bfe1d4cacdfec723 : SUCCESS in 22m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/33ececec1e914f88935e2d59604b4794 : SUCCESS in 13m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d8b2987e1c644f1ab858bfe34be91547 : SUCCESS in 11m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3c134887b01d45448b5007d5b9dd4a80 : SUCCESS in 12m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4b8878ba02eb45fd9e3fa00f074bcf2e : SUCCESS in 20m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e0049ccfbb7f48eea0e9dba620cb1bb3 : SUCCESS in 4m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/08a7bbf4d4844eac8b9561a3b0502b91 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad2ce7aa103546f890838f2b32ae63a7 : SUCCESS in 5m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c3861252941044f198140f9760a18ba2 : SUCCESS in 10m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cadafd27d78c4d1ca61c50ca6d6baf51 : SUCCESS in 9m 05s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d2a5852fce2448d89b6e46ca5b4dbee3 : SUCCESS in 1h 35m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/76e4d98c722945ceb7bbd92341ba64e2 : SUCCESS in 1h 16m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/de2e8d4c31ca48d4a9073b8089f3e565 : SUCCESS in 1h 40m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3685be81071549efaa8a51b6ab8ae9f6 : SUCCESS in 1h 10m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c3a109a0a93ec23806db4927c1ba8d920e531323","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-25 15:00:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a068b4c13bdc86bb81cb49c04f1f3f8d2530169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 16:49:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/437ed2a721614c75971af5f6c9f62413 : SUCCESS in 5m 06s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f0729fece44545108d0d8318b8384255 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d17635c21c8c428cb28c97ee952b66d1 : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/559db91c8e714e6fb8adda4abc00c713 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1326b9f51a4d4fa3843503cb97b87fe0 : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34a15a11a51142198933ad4b9c023139 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac269404b1a74458a65f7c53283b43d5 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/faafca1be7484d48b2deba4d9a7d5728 : SUCCESS in 4m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/97d5d03a856549acab58d4caad3ea477 : SUCCESS in 7m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f01982b23ae740aaa897221423cb95ef : SUCCESS in 8m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9cf7ca41d4eb49a59540b5bc4507d51b : SUCCESS in 13m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0ada469317cc47fc8961caa55e0f90db : SUCCESS in 10m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3f9eaf8af01c45e09ab3719139e44a34 : SUCCESS in 6m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7f7bf2293f3048fe9fcc21aa04967b20 : SUCCESS in 6m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8bc0d7d09f8445c5b0dee4cddcd5d65a : SUCCESS in 10m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/61828f9e60464c8481b91cca727c8a10 : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d011fb6150b9422c9068781f461d1e4d : SUCCESS in 10m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/517ccab260fb44c98beca7273b5a89b5 : SUCCESS in 14m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/66b91c8bbab84b44b041e2ac103ab2c2 : SUCCESS in 22m 49s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9d25812807434a348d498b5462321b55 : SUCCESS in 26m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5a512a79330440279aa82b0886223087 : SUCCESS in 22m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/059c26428a9e48578c63b7368fe0e3e1 : SUCCESS in 19m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e7b1b24542694128832d9ff9fb4546bf : SUCCESS in 18m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0e1aa1e8f2404ff5a644aef2f5be1946 : SUCCESS in 19m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9105ee192e114624840e5342ded1bc34 : SUCCESS in 21m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/10a2adb120e04f8e9e713cfb4c9081bd : SUCCESS in 8m 10s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f36577b2d5f24b30a4c7f0dcdbaf2f94 : SUCCESS in 7m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/704c3748359c44ee83450b518dd9d9a1 : SUCCESS in 9m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/848a803da45c4f419c22162f973357cc : SUCCESS in 13m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9178546c0769475c978d77a0534f6ea4 : SUCCESS in 13m 52s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ad8e19f6b2d744a3afa65ef6a45ba5d6 : SUCCESS in 1h 04m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/10e4548326a64606b8499d2724aba613 : SUCCESS in 1h 24m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b195a2ae603d47ecbff7e051eafb267d : SUCCESS in 1h 35m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/aaca451d22714015ac9d10d8525e3e5b : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9b77d7342a0444bc0337d883bcf5bfaf40dd552b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 16:50:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5b0d6408ee0894b9bec4f581d561ba3e5b29aee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 18:21:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87cacd1df25d4ebcaa1dc103d9414a33 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e59ce58322457eb612004f4cb87e4d : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67b20272bdbe4d5f89f2ec4d95f8412c : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/646cb49a7ff44281a791ac6a7593468e : SUCCESS in 8m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ea0b4c04045451ea0033b1154a43f04 : SUCCESS in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a2e296786f4486daf1b8905f525cbe1 : SUCCESS in 5m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/084a34b3feef47eaa6e5630ff7bc04a2 : SUCCESS in 6m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/07c5802b9fe24aadaee6e130317c24f7 : SUCCESS in 9m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2fcf4afbc6a0424eb23558f4c073ff3c : SUCCESS in 8m 07s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4683c1ae51cc476ea94fac1cb3232229 : SUCCESS in 10m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4e5e50c78f38401ca351bb768def3092 : SUCCESS in 10m 33s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f74ab8787c6f4624aeb18d7f2bc4541d : SUCCESS in 4m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ed46c91efe1e4210b59a6c661099c142 : SUCCESS in 5m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/affaa9f2dd954e0c815c0cc6c4fc7b56 : SUCCESS in 13m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6106090bf38348eda4e2042f1725ea48 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bdf5b994fa7749968d833cac441ca47f : SUCCESS in 8m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8e996ca49468419cbeb0ffd9e6df7fc1 : SUCCESS in 13m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ee267eb056fe43aebfdbc1046e547270 : SUCCESS in 17m 50s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1f0a5e4a9308437dba0e803e45f49f95 : SUCCESS in 27m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e93c71aa1ba74868ab5183ce9471440d : SUCCESS in 26m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9a64da6834c740daa125927cb54f3ee3 : SUCCESS in 14m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/90184df57a3645b8a5df04428c5ce849 : SUCCESS in 17m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/18738c196f55492a838966ca00f09731 : SUCCESS in 12m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c34106c1c67a43f38f729bc108e580f3 : SUCCESS in 23m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8c4cfeb7aa6241ed87e9161c47158a58 : SUCCESS in 5m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dfb51bd65760424e898ec2ba429feca7 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae901864e9a646539eea7b1cfd9d58a6 : SUCCESS in 6m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0e10ad130de548c5acd52ed7b23c2265 : SUCCESS in 8m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/29b7efdfa5684255b7efc01d65dd61dc : SUCCESS in 10m 25s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c6c1915382a4413781a45a60710ebc0a : SUCCESS in 59m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc44430afb6b4f08ae6a390d27e60b41 : SUCCESS in 1h 03m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ffab92c1849941239cac80703527232f : SUCCESS in 1h 29m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6b3323a7782d4d13a31980a28bff5129 : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":1},{"id":"18882d8a6849b77fda20a901704b68e92fb83222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 18:21:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f2b20f118eb8f832a6ca8831bef213b935c1804a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 18:23:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f450108b4a47ec9b15d4f491cf03a3 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19aa7c83a28b54c602c34c53d6f7bd0bf4a4f6d7","revisions":{"19aa7c83a28b54c602c34c53d6f7bd0bf4a4f6d7":{"kind":"REWORK","_number":1,"created":"2019-10-25 01:53:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/691141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/691141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/691141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/691141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/691141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/691141/1"}}},"commit":{"parents":[{"commit":"888ce8564abe71be3aec2401d36f917af6404c1a","subject":"Merge \"update constraint for python-octaviaclient to new release 1.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/888ce8564abe71be3aec2401d36f917af6404c1a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-25 01:53:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-25 01:53:45.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.37.0","message":"update constraint for openstacksdk to new release 0.37.0\n\nChange-Id: I7fc8469bee078effe7dade50b8fb89356c880a78\nmeta:version: 0.37.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Commit: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Change-Id: Ic08d2500eb66a966186d355c69f296503bb81294\nmeta:release:Code-Review+1: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19aa7c83a28b54c602c34c53d6f7bd0bf4a4f6d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19aa7c83a28b54c602c34c53d6f7bd0bf4a4f6d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
