)]}'
{"id":"openstack%2Frequirements~691208","triplet_id":"openstack%2Frequirements~master~Ifcd920159a1f5ea8a654d410231540970bdfe6ca","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ifcd920159a1f5ea8a654d410231540970bdfe6ca","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-10-25 06:32:44.000000000","updated":"2019-10-26 01:47:11.000000000","submitted":"2019-10-26 01:45:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691208-1572054355266-f546210b","meta_rev_id":"92618fc4b7d27cce4a1b7c30922f4c5df880b55d","_number":691208,"virtual_id_number":691208,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-26 01:45:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-25 20:45:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-10-25 19:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-25 20:45:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 19:30:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-25 20:45:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-26 01:45:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82b70b4c99b81c94782b75f42efdebd55bb61f02","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-25 06:32:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73aa418a562e0d943706262c99968aadfdf86054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 09:46:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81a34770bd94405a38703c01e1a77e6 : SUCCESS in 4m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/847a60fc709e4294a360df96085079d8 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43db13f7a53e41f2b96e4b5c44a612b6 : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3c434dedc724295aaa4e52aabfc2d4a : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e59fddc32e364a22acfbdedd11ebf16e : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f713290faaa14d79b0e560f2434f36f7 : SUCCESS in 9m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a605297725414227bca45750f6e92d4f : SUCCESS in 6m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44136b9bc3cc44b68ef8ddb670b8cf33 : SUCCESS in 5m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1523a836d7244399b7b915c32c8ef6d2 : FAILURE in 8m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cebe7cfce2c440d089ef613d30483d1c : FAILURE in 8m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0789e6f8b5e34d88a11f6aa95b8f22f0 : SUCCESS in 12m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4b4d0b8ac0bb440fa1d0dfdca88974a6 : FAILURE in 10m 47s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fa4d383783c543bfafa673af20136c44 : SUCCESS in 7m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f294958aa7d648f18fca7a02af14c410 : FAILURE in 7m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e679ebf76b854fae8297d9e237ecb7f6 : FAILURE in 5m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/35a4a6e98ce643858ade3f9949f2d400 : FAILURE in 6m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/acf6bcf311354dcdab252d13153abab4 : SUCCESS in 9m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f316096702cd4c6eb9ed5eae1e668583 : SUCCESS in 22m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4f7fb35870004010b1cd484bb406d962 : SUCCESS in 23m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9e07440da60f4dba8afbec8fb2ba2687 : SUCCESS in 21m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/53cff64be9c740b6bbab011b9da3a9f2 : SUCCESS in 27m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4fb9f7792cf340d8a8bd15acb1338aef : FAILURE in 7m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3f8946b7085c4a3fb8050e078549fe69 : FAILURE in 6m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/45f3021401424776862758d855cf9c17 : FAILURE in 6m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7d37554662764fd782e1fb2ef0dfaa57 : FAILURE in 6m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/405329a95cc7421fb1014d0dfc114bdc : FAILURE in 7m 24s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/41a9a8f0fc40471c97ee9c9f85931ebc : FAILURE in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f1a341a671349efa633208a02619167 : SUCCESS in 8m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1d019b80d2d64170baf76782786c1cc6 : SUCCESS in 10m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/40ab6c3ee3b242b6aab5477d9e0d4a71 : SUCCESS in 10m 18s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ecd4836785fb456598300d4e9e8d2164 : SUCCESS in 1h 12m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2277ffe7f6f54c089062c991c37068a9 : FAILURE in 1h 23m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6587696c8a25474cbb6a18be395fd753 : SUCCESS in 1h 59m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bc129d6d3e154437b40486e21e840756 : FAILURE in 52m 06s","accounts_in_message":[],"_revision_number":1},{"id":"aa1ff88cd3bbd0066a3d43431087557631309763","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-25 15:11:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"371d1cbed1d8da7fae1ef54d8ee3fd179c28c578","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 16:08:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b476860eaed45e2e0ac4beb2d586ec0da555a0b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:08:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6685e2adae52499d9e43e5bd286cc9a3 : SUCCESS in 4m 10s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/242d86ba12494c00ac53b8decc13d981 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e504bc22594c8b957b8039f6d5a057 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/906077199a094dbca09691755637a5b3 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29fe8ed1bf17444fa60e1389fa2b9750 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d1e0f9e9fe84db1bbf6097b308ad0a0 : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23e9d964309d4c67821e62d052e8e60b : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4272a28d70d140dc874c9e34ebd7bd04 : SUCCESS in 4m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0a1bea99c8354210a7b3f23cfb8a4c92 : FAILURE in 8m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/384065f58eb745879ee90ee155114bfc : FAILURE in 9m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ceee2bfa248f4cf8aefc61503bd4f2c5 : SUCCESS in 11m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dec88901d42f40bfbff20430b4be883b : SUCCESS in 11m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1df5a27c06074a02b9f66d4f7fdfbe1c : SUCCESS in 8m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/70280116c9824eed92c7df4925c1289a : SUCCESS in 6m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bb8ad6c2eedf44c6b4415eb277d90100 : SUCCESS in 10m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/41f486877cfc4c729823a7755e8172be : SUCCESS in 9m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/994157c2bf9a4d07bc78f1fa7637680b : SUCCESS in 7m 28s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/08cbd023d9fe4f62952c8b7f72e444b1 : SUCCESS in 22m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b3a49a14bce74557b4a7543cef6db6a0 : SUCCESS in 21m 29s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f92e372f950f40e9a59012e056dd1cf2 : SUCCESS in 24m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5803c11b76af48978fb7ad219a100ae7 : SUCCESS in 26m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1095944a0fed4d93b16f31bd0870b13e : SUCCESS in 13m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/706ac79d8e464f7da25d1ce5014e747d : SUCCESS in 17m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a7230743a8c740dd999692220be9d6f7 : SUCCESS in 14m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7d45b486e95476faae9007cf07bc520 : SUCCESS in 17m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2bb00370db0848e88422d4b5a4fc5598 : SUCCESS in 5m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b2ae79d0ab224c99b1aba955c9cf1fc3 : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af5081d40c7d490b93b07682c15d132c : SUCCESS in 8m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ff5f148944c04ab1ad2218a98e0f37ee : SUCCESS in 10m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/650c2db0a9b44428836cd6b336a5a977 : SUCCESS in 8m 56s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0e2b0cfdffa243a2ae16dfc109a9d2fc : SUCCESS in 53m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8d038065aed42e494d61a4f5a88715e : SUCCESS in 1h 28m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c3d6c96592104e5882b0b49b4144b87f : SUCCESS in 1h 49m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d94d3275d9f5444f915f96234a8a755f : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":2},{"id":"c13dcc801c0bef771a692350c9687dd5c1bb77c4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-25 17:56:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c4c0cb03a2fb5b36ace7f111b3eac9ab336eb736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 19:28:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d482236b5e04f56b40d2076bf04d07d : SUCCESS in 3m 35s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d09c33a253fb432c8ffc53d5a498f6b1 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e324465c618f430ba78b991eea84f6c1 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49c2a94a910249fa81acec8d84ef4ccf : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5888f503e821473e96e8331f94a91dac : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b29732fd08a4044a6944688aa035809 : SUCCESS in 5m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/695c4ded64bc4e439e8d1d9d50b19371 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81fc36d181be459e883fd6182a9d9b08 : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a002d60695c847b48c80d134cc6a15b8 : SUCCESS in 7m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e4890da6c14449d882651f9c122de291 : SUCCESS in 8m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/31c832bcef9e4329b7694b146dca6960 : SUCCESS in 10m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e715153f4f87469c97b154661dd1aa09 : SUCCESS in 8m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c516900401ee48cf941693e509eec834 : SUCCESS in 6m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/54ead39c86724420871a63d1d8ee7989 : SUCCESS in 5m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/95893fb460914fe1ab114cb10be605bc : SUCCESS in 9m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/57019be385f74dbd83430f10d746d5b7 : SUCCESS in 8m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/445f1e70206d497589dc4ea2b9de95e6 : SUCCESS in 7m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f3052af7c5134da290b9b091890da08e : SUCCESS in 22m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/751b070043104eb8a13a67f4f1ea8c54 : SUCCESS in 23m 42s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c2e047141350429f94f989b36d7cd34c : SUCCESS in 19m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fefc94e128a04c6cb0d7903110a7a437 : SUCCESS in 23m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a752fe3d269d487baca4dce2fc478925 : SUCCESS in 13m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7585456ff46e40f09e486b1e83e46a45 : SUCCESS in 14m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4f68207e50e8426dac75c16d2919aae6 : SUCCESS in 14m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d5333eb80b23450dbb83db4acf4e6b32 : SUCCESS in 16m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/305b845c86b5428da82203328295435c : SUCCESS in 4m 00s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/32afb5b8d04744babc614501b06b953a : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f55c2417671b4e1595316cf7b8773cb1 : SUCCESS in 6m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d4cdbc9129c44a28827da4d818bd6bb5 : SUCCESS in 14m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d58af0fafeb5465084abf32a5602f34e : SUCCESS in 9m 10s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/07021782ae3e4539b766b7e5204f41a3 : SUCCESS in 1h 12m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f36e45550138444d8d899058c4aa8af3 : SUCCESS in 1h 04m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b4d69a1dd637449f8e428a96fca5ebda : SUCCESS in 1h 30m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d63fe3adb61d4fa99dccdc22e1b961a5 : SUCCESS in 54m 10s","accounts_in_message":[],"_revision_number":3},{"id":"9270978741dc3b61953731d4647fb56fa5cacc94","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 19:30:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b8d9d7764ff81f752150dfcb1f9db15da4e3f8a5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-25 20:45:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"70282caf25958e796b98eaddc1cd7fe53a13a50c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 20:46:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f9fefed546c207931bb5946c6e50c268b7e046c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 01:45:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd94a6c82f7c4051a8e1cef7c5ec07dd : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d00dd8a21e4935864e3c89dc2d3f69 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74dbe18ed8e641af9aa4f1a1e50c3eaf : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea9d0ad51e30441ca375f62fd30e865e : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24df61492107479aa6ec253fc3963d5c : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54835c7ba731441f90490cf52d184dbe : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c76465a8973944849b2c79641d8d371e : SUCCESS in 5m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/35b1b19a0e85419186196cb7d1acc70d : SUCCESS in 8m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/329ead1286304e61aff4a5609a888816 : SUCCESS in 7m 58s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/87ad3b8cc8b2485ea67b6c6248390132 : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9cc851b831af43489cd76dbbb1e2bd8b : SUCCESS in 10m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/372c3743c3df4dafbaa48f722539ac09 : SUCCESS in 6m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/227169866f604d5ca640c1709fb8a327 : SUCCESS in 6m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/939083062356433f8750dcaa1fdb459d : SUCCESS in 11m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/76bac740487d4e0db4b8b7ce767c61b7 : SUCCESS in 18m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/861971b9d95349a186844529d7a01777 : SUCCESS in 6m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d478d64094a340cea4d0b5dd8a174641 : SUCCESS in 26m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2a4dfb06ad424960a6325bc9ac088fb6 : SUCCESS in 26m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3e49c98980aa4d8f9e25db49debbe2fe : SUCCESS in 37m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a06eeed472e7478c9e3a9382cdaf1217 : SUCCESS in 35m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2f1e850b435a4cd3b007a6f5963cf14f : SUCCESS in 26m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eb3c38288fb944f48aac6c87ae93da91 : SUCCESS in 23m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a5629cf110964f5f970400364bbd834c : SUCCESS in 12m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/db232d9bd4e04f6bac3c34d38bf2d3e8 : SUCCESS in 31m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d3ff958745304ee9b9b3e0351eabfc57 : SUCCESS in 4m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e89bcec4c11447fbb1f5fecbfe479c6c : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddc0a5851bcf4eb78acff98ba6a30e54 : SUCCESS in 7m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0c3cf872cc054d57a72271e280ee8f25 : SUCCESS in 9m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1b8c0895f25d42149c8686a86e730151 : SUCCESS in 11m 29s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0e18d85492dc40afa8f46d3acb8e34f1 : SUCCESS in 1h 05m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/105b486683a9400198908d8f54597b4b : SUCCESS in 1h 09m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/99b4c0b9debe435d8e425ccc41c7ec2e : SUCCESS in 1h 35m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b0749f8c6b98478eb059ff166368ec86 : SUCCESS in 1h 10m 10s","accounts_in_message":[],"_revision_number":3},{"id":"8f938ced7184cae03bff5e4f72bef8dc6cc84fdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 01:45:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d089a8014c9a59e621746111616576254f15ea3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 01:47:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/900324ca89ab4030b53d24fc727c5cdc : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"46bd15bfe6e62c223eb9b2e6ebce42cf7a11306a","revisions":{"e14d6afdc4cdb2ff84db7dbb9c390d7b0c3cce46":{"kind":"REWORK","_number":1,"created":"2019-10-25 06:32:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/691208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/691208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/691208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/691208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/691208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/691208/1"}}},"commit":{"parents":[{"commit":"888ce8564abe71be3aec2401d36f917af6404c1a","subject":"Merge \"update constraint for python-octaviaclient to new release 1.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/888ce8564abe71be3aec2401d36f917af6404c1a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-25 06:32:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-25 06:32:43.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ifcd920159a1f5ea8a654d410231540970bdfe6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e14d6afdc4cdb2ff84db7dbb9c390d7b0c3cce46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e14d6afdc4cdb2ff84db7dbb9c390d7b0c3cce46"}]},"branch":"refs/heads/master"},"c406169b65d62017ad136290897c2e0a2b079a0f":{"kind":"REWORK","_number":2,"created":"2019-10-25 15:11:26.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/08/691208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/691208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/691208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/691208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/691208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/691208/2"}}},"commit":{"parents":[{"commit":"87f4f2313f438cc7d42e8a8a83978438f5e0341e","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87f4f2313f438cc7d42e8a8a83978438f5e0341e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-25 06:32:43.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-25 15:06:30.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back the following\n\nsalt\u003d\u003d\u003d2019.2.2  # going to mask as it holds back \u0027PyYAML\u003c5.1\u0027\namqp\u003d\u003d\u003d2.5.2     # kombu requires 2.5.1, not holding back as kombu could update\nnetworkx\u003d\u003d\u003d2.4   # need to mail the list and have a test review for updates\n\nChange-Id: Ifcd920159a1f5ea8a654d410231540970bdfe6ca\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c406169b65d62017ad136290897c2e0a2b079a0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c406169b65d62017ad136290897c2e0a2b079a0f"}]},"branch":"refs/heads/master"},"46bd15bfe6e62c223eb9b2e6ebce42cf7a11306a":{"kind":"REWORK","_number":3,"created":"2019-10-25 17:56:17.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/08/691208/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/691208/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/691208/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/691208/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/691208/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/691208/3"}}},"commit":{"parents":[{"commit":"87f4f2313f438cc7d42e8a8a83978438f5e0341e","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87f4f2313f438cc7d42e8a8a83978438f5e0341e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-25 06:32:43.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-25 17:56:08.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back the following\n\nsalt\u003d\u003d\u003d2019.2.2  # going to mask as it holds back \u0027PyYAML\u003c5.1\u0027\namqp\u003d\u003d\u003d2.5.2     # kombu requires 2.5.1, not holding back as kombu could update\nnetworkx\u003d\u003d\u003d2.4   # need to mail the list and have a test review for updates\n\nChange-Id: Ifcd920159a1f5ea8a654d410231540970bdfe6ca\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46bd15bfe6e62c223eb9b2e6ebce42cf7a11306a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46bd15bfe6e62c223eb9b2e6ebce42cf7a11306a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
