)]}'
{"id":"openstack%2Frequirements~691303","triplet_id":"openstack%2Frequirements~stable%2Fstein~I86826c07ca316bd739a468cb3c7e6331b3546004","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I86826c07ca316bd739a468cb3c7e6331b3546004","subject":"update constraint for python-swiftclient to new release 3.7.1","status":"MERGED","created":"2019-10-25 13:05:50.000000000","updated":"2019-10-26 00:19:36.000000000","submitted":"2019-10-26 00:19:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691303-1572049176136-34f7ffe6","meta_rev_id":"99cad0c1b87766a5fbb2239552bc4741517e7369","_number":691303,"virtual_id_number":691303,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-26 00:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-25 20:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-25 20:32:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 20:32:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-26 00:19:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67e7dec3203ce2f4516f4d9853374e2852c60dc9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-25 13:05:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48345ed6c990852da42a8d24cf0d342477b55716","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-25 14:57:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b8b5af0a03f34b4f493ca346c4150aa7ccdc6d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:58:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4dbb2a97d7a1475fa600e1f8d7e929bd : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849f796c0cf141a7b7e354ba52884018 : SUCCESS in 2m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51987f3d06074bfd88b02bb8e7324422 : SUCCESS in 4m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ebd2439fe3614c669432b9edc9467e9d : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5614f21162e640bcba7b7a49577a5340 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1d9c080bb0b4586b709cd2aed089b16 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7c3c57b27744248818a579f6d0fe8fb : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3b51c917066045ca830e7f7669498b4a : SUCCESS in 9m 07s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/4bd9d45ee7484db5875dfd68375b9f7f : FAILURE in 3m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7b05487bee174af8a0c750727fcb17e7 : SUCCESS in 10m 23s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/be0bb912b13d4190b7f36194f78932ce : SUCCESS in 9m 21s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/4daaae2e0f1249f1b89c3a875726a3d3 : FAILURE in 3m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3b5146f57eda41679f9e82affeb460fb : SUCCESS in 4m 19s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/79b25b50b63048c69579191c132ab8dd : FAILURE in 2m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fd4708f0d0324bea84d47334a53d66ee : SUCCESS in 11m 54s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/c5c695dc2d2b49a89c702f08ba43648a : FAILURE in 2m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/768cd988322e4d099031247885f445a4 : SUCCESS in 6m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/43aad75f7f2248a3bba2284a36f9da06 : SUCCESS in 25m 53s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/4f7a9a3f3ec04553badffe415e858883 : SUCCESS in 29m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/64a0c60098814e619eced6744ecf29aa : SUCCESS in 29m 55s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/88557abbff2d400188ecf8f345734dc9 : SUCCESS in 26m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f7a4588858914227b5fde84f3e4a1e82 : SUCCESS in 16m 16s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/25e6ba4388764bd6980d083ead1d1695 : SUCCESS in 13m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5962dee72bd64afaa24a97e7b90554c1 : SUCCESS in 24m 26s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/162fda647fd7431fb11ca9918de0cb57 : SUCCESS in 5m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f40c55dd802449edb711e90d7387647d : SUCCESS in 10m 16s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/1ec7d20e2ee848aa8d2fea7cb9a39815 : SUCCESS in 11m 48s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f6aa465f643340c7849c4549ffc788ac : SUCCESS in 1h 14m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/99808b2e515648ac9ad6772059b1320d : SUCCESS in 59m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/735af6ca032b4de4b8f3f5ff322e4ed5 : SUCCESS in 1h 43m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/69ecb39f84314411982fe12898243bef : SUCCESS in 46m 30s","accounts_in_message":[],"_revision_number":1},{"id":"79d7c28728f900adf334d8433ec042f871b08508","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-25 19:03:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fba03a1f5204b54430998768f9482177f3fa62ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 20:29:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/89735ca8cdfd4283a6b57eddaf62b655 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66b30f21842d4592b781b3716a6b1307 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb97fdff281c42259cc6d8ddb23c3ba4 : SUCCESS in 3m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bb71c721aad34150869a36fed451a7e1 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f12c771b9ad41378e56b51a27665bbe : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b52fd43cdc0948359db85ea79347fcfe : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cba6679fe14e44ac8be5e05f89eaed08 : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3151bf8b781f4132bad4fc34c168398c : SUCCESS in 6m 56s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3e60ae76184245798377985e9a35213b : SUCCESS in 6m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7045ad8ed0e34933abe2f5ebab8d3b34 : SUCCESS in 7m 50s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9a954026ca3745f78512bd99d82267db : SUCCESS in 10m 06s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/7e2aba0eec074bd4bd778ac363be9c93 : SUCCESS in 9m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0c1cb2e5b8604e7ca406335e55bef8f4 : SUCCESS in 4m 44s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d2e671868fd447c8b9c5586acca8088e : SUCCESS in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d7c2802c11944e49865880a78ce037ad : SUCCESS in 10m 05s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/d7f74f77b4a748c3ae210429d0b3c80e : SUCCESS in 8m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c76c0b3c822c4e75b75e72baa31f1bc1 : SUCCESS in 7m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fe491416ead0403381fca9a07bb7ea3d : SUCCESS in 25m 45s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/43dfc858a5d74f68bb1ac861ac337d8d : SUCCESS in 21m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d9edb52e0c1547f4a6e3196da7375e8a : SUCCESS in 21m 15s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/00176f0f056c41e79a1ccc545351f8e1 : SUCCESS in 20m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e3f292154dab4c2b89217292361c2e2b : SUCCESS in 10m 07s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/2af60d8147bc4e83a5837a43f09c992c : SUCCESS in 12m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5dea4ccd293b4604bd70f79b8bc31546 : SUCCESS in 18m 10s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/bfef868aa34542c5bec6354de2744eea : SUCCESS in 4m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bc9caca13f684c3c88368ef19bf95b39 : SUCCESS in 9m 45s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/099716657fbe4ade8d7520bdab0c6fd7 : SUCCESS in 8m 11s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3589a78cf54740f4a23557cf3132cdc3 : SUCCESS in 52m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3101ede050c417599a73189afbc0262 : SUCCESS in 1h 07m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/514c2f2ccf9c41ba98b0a98b19f90859 : SUCCESS in 1h 18m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9ec901b399dc4cfe8d21a933b46f29a9 : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":1},{"id":"335fb7d20c21e811257ce23d701095884e3ab1ac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-25 20:32:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd3be8f1913b99cb19d2adfb31728786d1091e11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 20:33:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3bfb38450f5d4c72eae82688a6f4af192a0b55bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 00:19:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfd954f5859e4ab1ac0a07f8316e105d : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b40883291a04d13af4401818805541a : SUCCESS in 4m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eb590796ef164bc6955d2469c79bdbfc : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e599d602174940c4b9e276b6b35d4215 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df8206fe6454499097046103b860f83a : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e05b982a350d475e8e762aa6b2cd6190 : SUCCESS in 5m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/375a0494628b4d08bf8c6b54fd5ebdef : SUCCESS in 8m 41s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/4097591b0e2b453c9e60c613e0e4a361 : SUCCESS in 8m 04s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3138961e89594e6b8b1aee8580873009 : SUCCESS in 15m 39s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/28244539b191447bbba41defaa712f06 : SUCCESS in 9m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/806ce4a905fb4ec1810491ccf48e9173 : SUCCESS in 5m 34s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/7944aabd20ad4951b1cea5cfe16ad860 : SUCCESS in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0850f029643548cc989e7bb1c3357c53 : SUCCESS in 9m 50s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/bd37e18e66d54e858f563a6388624985 : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a4d28f34cc3465b96c4c8c67b80d609 : SUCCESS in 8m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/74e68a569114494ea44189f91dede018 : SUCCESS in 19m 18s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/51372aba03884a03a12d59945682d636 : SUCCESS in 25m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c7ff1aba48c4486d9418ca375715ba0c : SUCCESS in 21m 30s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/66e0315f15d04cf2b60b05a57407503f : SUCCESS in 29m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dae1db0db8324c62bc323b9fa5344e40 : SUCCESS in 21m 43s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/18c7232488e54a31a5e1910acc8ee464 : SUCCESS in 16m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3b4cfb80a4f40f6ab60d203f3ef7f13 : SUCCESS in 18m 38s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/f7f7c5d06ea24e458abaab50694222b6 : SUCCESS in 5m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c220629782f044afbb781059c6ebeb79 : SUCCESS in 9m 09s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/6c785b575b104e0783c81a0023077897 : SUCCESS in 9m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/b2a8067a19e24d898c8426dc9cbf10f9 : SUCCESS in 45m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb4bb957b3674c79baa58fd0e39dadf8 : SUCCESS in 1h 03m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2a8f58c684a04c50bac4a5ffd5801ffa : SUCCESS in 1h 47m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0f5c9c5818dc4b79a978ebf7c13b2da3 : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d557b0a2360fc116cfa1318bba591bdb15d526bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 00:19:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e28c429e77f597091f011448e9a2d47be9d69a52","revisions":{"e28c429e77f597091f011448e9a2d47be9d69a52":{"kind":"REWORK","_number":1,"created":"2019-10-25 13:05:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/691303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/691303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/691303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/691303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/691303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/691303/1"}}},"commit":{"parents":[{"commit":"d4f1977da2dc3c1e8c22bc835fa597ef09a7b8bc","subject":"Switch to opensuse-15 nodeset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4f1977da2dc3c1e8c22bc835fa597ef09a7b8bc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-25 13:05:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-25 13:05:48.000000000","tz":0},"subject":"update constraint for python-swiftclient to new release 3.7.1","message":"update constraint for python-swiftclient to new release 3.7.1\n\nChange-Id: I86826c07ca316bd739a468cb3c7e6331b3546004\nmeta:version: 3.7.1\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tim Burke \u003ctim.burke@gmail.com\u003e\nmeta:release:Commit: Tim Burke \u003ctim.burke@gmail.com\u003e\nmeta:release:Change-Id: I6bf53b012f285f1eb5978628a7b8ebee7995f8aa\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e28c429e77f597091f011448e9a2d47be9d69a52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e28c429e77f597091f011448e9a2d47be9d69a52"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
