)]}'
{"id":"openstack%2Frequirements~691304","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I9de68784fc5d7703a9f6d179558234db581ed4ee","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I9de68784fc5d7703a9f6d179558234db581ed4ee","subject":"update constraint for tripleo-common to new release 11.3.1","status":"MERGED","created":"2019-10-25 13:06:13.000000000","updated":"2019-10-25 17:43:29.000000000","submitted":"2019-10-25 17:43:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691304-1572025409236-e84120e5","meta_rev_id":"18b058683be70ce998fcf48e818c6e01b319f895","_number":691304,"virtual_id_number":691304,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 17:43:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-25 14:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-25 14:56:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 14:56:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-25 17:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b67a9e747880d83fc7788be3dd41cae014e219b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-25 13:06:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0115197b34554c85fc3d0d0f000657ab0603aab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:50:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9be16d44502348e68f70b93508e5c41b : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf17a06c7ca04a6d9109143990202825 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee7f61a1eddc4ed1b71d62eef6d41b81 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71fdebe077704c59bffc87f0b6ce47fa : SUCCESS in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e5778d5fb00458c832db10fdeded4dc : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d203cbb669074faeab6f9156762e1a55 : SUCCESS in 6m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a431e2e29181405e946ec11d2cf95550 : SUCCESS in 3m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e9b17bb99abd4ef6a548d5e566f1823b : SUCCESS in 15m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3f97cc9ae8ce4b379ed5728b565add23 : SUCCESS in 10m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/38ad40109b6e48baa3ca6a4f60851968 : SUCCESS in 11m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6c223346254847d79a706acea91d832a : SUCCESS in 9m 47s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/955979c7662140eaa1af96be651f6628 : SUCCESS in 7m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ecb68b97a7534bc98ea260e73f66f615 : SUCCESS in 5m 49s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1eb05a73630e4ce0a916e8c3137e491f : SUCCESS in 11m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d4be28a561364f04b6279313aacd6c8e : SUCCESS in 8m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ba292801df74bc1812d1d875c975cb4 : SUCCESS in 12m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6eedcc8baa6e4b02be413545da977ffd : SUCCESS in 19m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/09d1b78bd7224f0abb10eb2b69714d5e : SUCCESS in 22m 40s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3abf67f14ebe451abec3215ce8d8620c : SUCCESS in 25m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0074ec3a9c3e4f0cae3268151ffbcd3e : SUCCESS in 22m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b2e57c9ba2104d4582c59c3a3d1bbe2f : SUCCESS in 14m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0f664b408d36410f9df2ffdef09fb95e : SUCCESS in 16m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9323f1ffd9a14157a272348b2b2c52cd : SUCCESS in 11m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a490ba8fc07e4205a03e54706e4d2de6 : SUCCESS in 19m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2826c4dbdbd9421e8ad5b676e038782e : SUCCESS in 5m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ad52660cd8a0462d9b126b0839566405 : SUCCESS in 9m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/42d1bead9e8a445880d21e587b1e3d9c : SUCCESS in 12m 38s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/347ed82fd79e448db1a6a015813eaaca : SUCCESS in 1h 15m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/abac98f820b54e76985315cb0f7c9529 : SUCCESS in 1h 34m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bbee6a009c0d4748a43404a17226ab70 : SUCCESS in 1h 39m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/145cb4e601ba4aa1a8359448684a7086 : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"47cdc4c065029e2d6de99933fee8996939dd0012","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-25 14:56:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d6be02c22f8a1ddcf688a5bddbc6aa51a3e097a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:56:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7149b775c954b99694ff821d4b89bf85abef29b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:43:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35ac8b972ed49f49eb6035f2f7e72b6 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba6f628819da44269c533c737e981e7f : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b3579ffa16d4e8185a5248961c294a3 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0262d93f3f4b4dc1b43b592a8bf844ef : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/acdad3b384e647eea143c79470bd43f7 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a3a980c7a324d1b9b12de1dd5803cfc : SUCCESS in 4m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/81b1cc6f1bf24f2ca750dd594b502029 : SUCCESS in 8m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8306fddcda03496185e3fcfa884ad216 : SUCCESS in 7m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c05e1239cfdd458fbf527b5e88242a63 : SUCCESS in 12m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1d1bcd5df5c441a3a7299b21b4e884e2 : SUCCESS in 11m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d86831e214c14633bdcc74f800ff154b : SUCCESS in 5m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ec24468b5d3043e88c0a6757678d22bb : SUCCESS in 7m 36s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f0f99465ede446999b20f55e5fdd0b44 : SUCCESS in 8m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/937f49a69e19416d832edb7f273cfd2c : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba5fcbe0096940d9884ee294cf3b5fee : SUCCESS in 7m 21s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8e3dd87cd33c48d2b396baee52057b33 : SUCCESS in 17m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/08cfb63a344a4a779615e6c443d5a4e4 : SUCCESS in 28m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/83f7bd642960482aabf3e46abfaf785a : SUCCESS in 25m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/245f09b3b02f42209701f900ca5fe957 : SUCCESS in 20m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/458c54414cdd4e2898ee8b821b264278 : SUCCESS in 15m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c7d63b633b8a4d60a82c27eb4d94bada : SUCCESS in 14m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5af339495b8b447eac9d7e274aee87c7 : SUCCESS in 16m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c498fd53adee405dada1af8f537a1638 : SUCCESS in 23m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a79a5d0720be4ac3852c75eb38d1577e : SUCCESS in 7m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/58992294eaf64014912a3e797f6a91d6 : SUCCESS in 17m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/575a65a73c5146deb3e5be57d15ce6a0 : SUCCESS in 11m 02s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d48fc0810acb426d84e685cae2a9764a : SUCCESS in 1h 02m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bae7efe65ea94fcf8743d7ab5c428b94 : SUCCESS in 1h 01m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9c36058be3d84b31893ab3848d70a1de : SUCCESS in 1h 29m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9af03e752f8947ca87c800f6e3cd6b9c : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":1},{"id":"4926ba7493c062da93a06f53c29eec4aab0cd1f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:43:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6451931b35e993df6f5dc36a147d59aa3f07c44a","revisions":{"6451931b35e993df6f5dc36a147d59aa3f07c44a":{"kind":"REWORK","_number":1,"created":"2019-10-25 13:06:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/691304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/691304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/691304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/691304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/691304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/691304/1"}}},"commit":{"parents":[{"commit":"491efcdbd12cc4dc1bc1ed736ae491f2e107afba","subject":"update constraint for tosca-parser to new release 1.6.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/491efcdbd12cc4dc1bc1ed736ae491f2e107afba"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-25 13:06:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-25 13:06:08.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 11.3.1","message":"update constraint for tripleo-common to new release 11.3.1\n\nChange-Id: I9de68784fc5d7703a9f6d179558234db581ed4ee\nmeta:version: 11.3.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Commit: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Change-Id: I4bd98f655455464b40d522dbb1b732b0f8c6add5\nmeta:release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta:release:Code-Review+1: Harald Jensås \u003chjensas@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6451931b35e993df6f5dc36a147d59aa3f07c44a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6451931b35e993df6f5dc36a147d59aa3f07c44a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
