)]}'
{"id":"openstack%2Frequirements~691307","triplet_id":"openstack%2Frequirements~stable%2Frocky~I8d10aeaeb75d594b8588778d73bfd7a17ee29868","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"I8d10aeaeb75d594b8588778d73bfd7a17ee29868","subject":"update constraint for ovsdbapp to new release 0.12.5","status":"MERGED","created":"2019-10-25 13:16:12.000000000","updated":"2019-10-25 17:43:30.000000000","submitted":"2019-10-25 17:43:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691307-1572025410472-b61b701d","meta_rev_id":"131b61c8728c99c6188ba7444567f8dfbd2480a7","_number":691307,"virtual_id_number":691307,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 17:43:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-25 14:56:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-25 14:56:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 14:56:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-25 17:43:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d6a694178cf47247aa7549349d28e6b474208c1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-25 13:16:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f0c0db631364a61f8e3328d165279f040b496bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:54:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a68537ae2dde441388165a088080e7c5 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95e7820f03ba461b8a31509bf3c6d353 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2054599a40ee465daa9364ff3e0abf9f : SUCCESS in 5m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9f8cea324c6e444ab7e008255f315128 : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84222320ee284c7bb5683415779e3cbd : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e7609c837b1643e1afdfdae807c25732 : SUCCESS in 3m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/17806cc5218b4917a7fe917f0e51a61f : SUCCESS in 7m 20s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/0666379bf97d49aa9c04355c5de40251 : SUCCESS in 7m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e2409b0b07ee4d6f9c64d4735c567e15 : SUCCESS in 6m 22s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bdfd60993e7d4f51b253f4a74886c6a1 : SUCCESS in 9m 57s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/76a5fe2e9b9b44fd98b9e3d3c2bed0ac : SUCCESS in 9m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/be825f3afa0c42d2b204eb80b70506e6 : SUCCESS in 7m 08s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/71905332e7844b3e90ef71002f9e00cb : SUCCESS in 6m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/cc538771ed7b43f58eca010fe4a43842 : SUCCESS in 9m 49s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/66fd00d28d7a4d7aba2229f5b2673b8c : SUCCESS in 10m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/341977f9e2114165b01d8eea5dac3a3a : SUCCESS in 17m 28s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/021fb80adda34c64a37b432fbc330cfa : SUCCESS in 16m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/496ff5cdd5744dd1b32ca020b69c1b87 : SUCCESS in 25m 49s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/59a2e22504b8404d8e7374c922d4a266 : SUCCESS in 25m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/41ebdc85a7d449c28b07c9c4ece6d266 : SUCCESS in 13m 30s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/df266885de48408f9b864344eb666574 : SUCCESS in 15m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/626d1d02e98e4117ae41655f5247771a : SUCCESS in 17m 31s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/e1350db0ffc84a109fc8262b306952fc : SUCCESS in 5m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6c0841dc3c5a4849a2b1ae08473b7252 : SUCCESS in 14m 31s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/5926f2cd16274f8897a94c6d354bfc00 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/042f5556771449eb9ff37f6a28cfd320 : SUCCESS in 4m 06s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ba636219092a442ab7f097acdd8191c0 : SUCCESS in 1h 06m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2bca0aa49e8f435da55529b36b685c7b : SUCCESS in 1h 14m 24s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/6fe8d98ca80e48be8074cb3f191fdf4b : SUCCESS in 1h 34m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b4e39e7870af44f6972a0dab749f1687 : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4339296210710559b5c9887bd1a48faa9c40b80d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-25 14:56:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"acfe775133d95a127cfec8f4a399f22e01194e82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:57:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e1234cc9b23f479f59dfb24ca06d0baa0b23299","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:43:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"34e57388ddd231d5e2a62206a48bfb0c53265815","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:43:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b04c79fb0b44e0babcb95c347f7c4d : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f77bddc846e48c6ac25330774dd17ad : SUCCESS in 5m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ff8e2b30ca7c4fe7ad9b6721ff372675 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81385df53d804684a22a1e00aee25a94 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/18e7db7563334851a8d37d3936036253 : SUCCESS in 4m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/01d22684d66a4537afbdb2b1fe2e1060 : SUCCESS in 7m 08s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/ca3c97d9c5a44624b7beaa376bdd4e84 : SUCCESS in 6m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c162b49c451446588d8d5fac1a26fc55 : SUCCESS in 9m 33s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/86c0f1ac33294c36a4443191726141aa : SUCCESS in 9m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a7b843fa84384485a731437b95203bac : SUCCESS in 5m 55s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c3f43ccf1e174a67b0ff796214fe5c67 : SUCCESS in 5m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0ae10ea6a4994333b5456f9666637fc6 : SUCCESS in 9m 27s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/c4d8d3e1cce84dfdb8af0c706af020e9 : SUCCESS in 8m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bf59b4963c4d4eb0b15ab695f1a80056 : SUCCESS in 14m 15s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/ca565262df5e447fa1fc3afb9c97166b : SUCCESS in 16m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/cdb4a19a53cb4998bb2fb6434179f6e1 : SUCCESS in 27m 23s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5983e2edfb74402284e28a564520af96 : SUCCESS in 26m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6394ce30674348ce84994514ae14fead : SUCCESS in 13m 36s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/17f91b951c5246a19c79fa02e6af1eef : SUCCESS in 13m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef8c14ad3df643b9b0ca40314d48aae3 : SUCCESS in 18m 45s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/ae354f3867e143b2bdf5acc96e3472a5 : SUCCESS in 5m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9cf6f5509227412b827faefd598c6442 : SUCCESS in 7m 41s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0af0da604041437591ee5b99bb5a110d : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34c2e89db3344728ac4560613b23d6aa : SUCCESS in 4m 25s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/849ec0e706934c2f81130bfc2870acfc : SUCCESS in 1h 02m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a1b506f45fe4acf8fcc5c2d6f9e36b2 : SUCCESS in 1h 03m 20s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/27d401f4370d4327aca72f9ec5a94b95 : SUCCESS in 1h 29m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2312f6ac97cc4de88f0abb8cf98b5e8b : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f7c1ed170d03d3efa454bee88f161157ec6302f","revisions":{"9f7c1ed170d03d3efa454bee88f161157ec6302f":{"kind":"REWORK","_number":1,"created":"2019-10-25 13:16:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/691307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/691307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/691307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/691307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/691307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/691307/1"}}},"commit":{"parents":[{"commit":"006192485b2969cd71654f12eadde86187b55f2b","subject":"update constraint for oslo.messaging to new release 8.1.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/006192485b2969cd71654f12eadde86187b55f2b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-25 13:16:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-25 13:16:10.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 0.12.5","message":"update constraint for ovsdbapp to new release 0.12.5\n\nChange-Id: I8d10aeaeb75d594b8588778d73bfd7a17ee29868\nmeta:version: 0.12.5\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta:release:Commit: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta:release:Change-Id: I44fbbe14db422ef54e591dc866e05ddbadc9a5b5\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f7c1ed170d03d3efa454bee88f161157ec6302f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f7c1ed170d03d3efa454bee88f161157ec6302f"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
