)]}'
{"id":"openstack%2Frequirements~691682","triplet_id":"openstack%2Frequirements~master~I85a222b9c1ca59e7ba9669155838cfc34d081a7a","project":"openstack/requirements","branch":"master","topic":"networkx-2.4","hashtags":[],"change_id":"I85a222b9c1ca59e7ba9669155838cfc34d081a7a","subject":"Bump networkx to 2.4 for py3","status":"ABANDONED","created":"2019-10-28 14:30:54.000000000","updated":"2020-01-09 01:26:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5810086ff07f381af3a25e0fec2e06c5457ab03f","_number":691682,"virtual_id_number":691682,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-09 01:08:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-09 01:26:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 01:08:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"638f47460822308def540b0930689a329b93e602","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-28 14:30:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9b215ea95c51778c59cd8eff72594e97639485a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 16:17:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a53873b17104a6fae44d0870133b5f1 : SUCCESS in 4m 11s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f1b99473b47045fea37848e9c703c1f6 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2cc2e465eee4a9f8fa87ced3fb78c40 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7589b6e9f2834a8abc4e5debcece887f : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85fd9e8a3bf1454fb24f7650165d9613 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7405734fd3f44d04aed30a321e4020ac : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c375d88e615459f910f8cb21f43a016 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c1fec0f4e887452bafb3ced1c9351014 : SUCCESS in 4m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4052ede2876449d1bb0ba2d6c955d12d : SUCCESS in 8m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fbfca494305449e8ad365e1014616fc7 : SUCCESS in 8m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/50fbc5f6c26c45f282617ffe39936557 : SUCCESS in 9m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/90302547366645e2a98926560a9ef0f2 : FAILURE in 9m 33s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/947d6ddfb08a4956b0fd00d6e066a493 : SUCCESS in 4m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c3bc3c0b94654d37aa71c06d1ec1947d : FAILURE in 6m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fdf8238dc01e476791e4785ba5c62678 : SUCCESS in 10m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/efe4127a63b04d7ca744c76b97d8b4ec : SUCCESS in 12m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d4dbbb24b1e64615b907bacffbe15dac : SUCCESS in 7m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d37c260fbf8d441d9f447513b3a59453 : SUCCESS in 16m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f6fea2125c8c44449d8bbc820cc636fc : SUCCESS in 13m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/bc13bed9f13c4336aa3cbda659480b51 : SUCCESS in 25m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5128b17f5124419b9db1a6f7ffbb1bcc : SUCCESS in 23m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5da743f67c0a42af81438d8991379381 : SUCCESS in 15m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b9f4d2dcd3a5465e875249d62be88e44 : FAILURE in 13m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2ef62397271e4f2e81b5e154759f654d : FAILURE in 14m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e6db1319396455992e5de4faee21cbc : SUCCESS in 17m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/198babed9dc647b1b3b8c3a66c7bbe60 : FAILURE in 5m 59s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/280575d5e52147108239516e190b4cff : FAILURE in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c8fb6644b66e4b69a92c1b261c5e297e : SUCCESS in 6m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ad2cbd8dd686459a87f38ec31f118f09 : SUCCESS in 9m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c33e1c29660e493c9bbbadaa318de954 : SUCCESS in 11m 13s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5b63494140e54a65ac6fccfd94d8091f : SUCCESS in 1h 05m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/339b8cf76d834ebb9c57fc463a9e2c0d : FAILURE in 1h 16m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1137b5d5cf6247e1b9c9198288e267bd : SUCCESS in 1h 40m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c776eebf1d2c465bba8427886bbc4a54 : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a7634d73b10ef911fe69af251a926f5b4ccdc7be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-28 18:29:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"16af1245ca02a8c116ac38e85d874a4b6a1663c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 19:59:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3848d10c3c4059b60639181e9921c0 : SUCCESS in 4m 11s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/130d28ce8e564b77ace16161d35550ee : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6fb8a68568e480d9a5c2a3ef5a8a4a8 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2bffba3b70043a49f9ab19114723bd6 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1bf9b0b6aed42468792a5f86a0a4943 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af6d1dac7721479899028d69ec67c66a : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebae703e7e9141169cee133b1475bfe9 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44de3eaa49e2489fbc2b44967ffad708 : SUCCESS in 3m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e902865b23e74461b3477b2d1b245626 : SUCCESS in 7m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4af74b5e85cc4f04a73bf4e51f1d7f30 : SUCCESS in 7m 43s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/eb6d721979444ddfa0c78236afe7c7c2 : SUCCESS in 8m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6b29f145d426476f88d57aa8227a79bb : FAILURE in 8m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/894449fbb19b49ba952df39d3bb0b218 : SUCCESS in 5m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2bed55a76d5c4af5b05e2d05e8bd0e5f : FAILURE in 7m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/66063476f3d04a4da3930e8244648096 : SUCCESS in 10m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b60868da2a9b45d7a50955d441df3a9e : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f67cb46ac2ea4167a7c844a60a91846e : SUCCESS in 7m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b4de6baf69b849db8932b69d9c2cacd5 : SUCCESS in 18m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c76c7355ee4e4e1199b72fee478518a2 : SUCCESS in 14m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9c592fd49e0444c491bb23461195cddb : SUCCESS in 23m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6eb7de3677a84c529e7e71624ca4c45f : SUCCESS in 20m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/aa1a97664f0a43eb8e2035e1cb2934ce : SUCCESS in 15m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4631259a59c64611a47659ad122b4fc3 : FAILURE in 11m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8d6b4cba20ae41d5a659073a8c9c393f : FAILURE in 13m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3565656ecfb242229951fcf0cfbc52fe : SUCCESS in 16m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/71cbb0227a2b487f8d8005fce9f71a75 : FAILURE in 4m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0ff3ad1c2ac34de79c8eef16f06eeffc : FAILURE in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/122e940cb21f4536b5611660918c9c21 : SUCCESS in 6m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9a988b77f179492c8918f1ac58c8ce49 : SUCCESS in 8m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fca082122bbb49838e81eb496e318c21 : SUCCESS in 8m 31s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1b8c30b3fe7f4a5da9e962fe708cae96 : SUCCESS in 1h 09m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ac7fde91083409b90124858b1427825 : FAILURE in 1h 06m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7180c4e071124ef290d2b043ae025940 : SUCCESS in 1h 27m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0d97c4c014434262aa8f879004a35fb4 : SUCCESS in 53m 36s","accounts_in_message":[],"_revision_number":1},{"id":"63d4db5fef3e19ab483b7f41b4d1c99c037741da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-28 20:06:58.000000000","message":"Patch Set 1:\n\nlooks like the source of the issue is taskflow?\n\nAttributeError: \u0027DiGraph\u0027 object has no attribute \u0027node\u0027","accounts_in_message":[],"_revision_number":1},{"id":"9dc74186924371484c851a86462de13dc39d4c3a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-28 21:17:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f9b256c09f55e58069c4184aa8fe32451e2d3c24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 22:49:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5489058ea94309ab6157acf738b5be : SUCCESS in 4m 34s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3a159da7088545cab8b5096db170584c : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efc47f7e0f82463983cdd9937fe0cd8d : SUCCESS in 9m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3429aa64ccd041ac9d258d8e180b9525 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/568f9dcefaeb41c2a98b98f19df062aa : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aac01ac7241649299aa942ed984f9288 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5554c10ecea486cb22a04e55fd86705 : SUCCESS in 8m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/43e191f90b8c4e8697e2a0fc297182b4 : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ecf13fcc27c34b4c8c3f6347502a5e61 : SUCCESS in 7m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2aec723abcb643c1a1fd43a92f17c36f : SUCCESS in 7m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4a87aff3052f4b9295aea19f1238e019 : SUCCESS in 9m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2f01f838314a4346b642362af20c6ec4 : FAILURE in 12m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4dd891cfc29c48e19de7c9ba35fb0026 : SUCCESS in 5m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/70392038941f4e82b827831da743afaa : FAILURE in 4m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ab30a094c96c470e86e4d4558a901113 : SUCCESS in 9m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8ebc057ac6f740b2b7fbad664c924e58 : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6715d3af2c2040d291d828139d6c2974 : SUCCESS in 8m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fb7e365a9301447a9f2faf62d4b064a1 : SUCCESS in 19m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4660fb2852da4abf968e86e7e60f20d6 : SUCCESS in 12m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ae849ee5912d4615b29e30a208c3663d : SUCCESS in 20m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2eca5650868b4772b251e4a347474ab8 : SUCCESS in 21m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/99dc36c35faa423b9306b827785f6a21 : SUCCESS in 13m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b73558609b62472db429fbd0e0fefa22 : FAILURE in 13m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/480fce9e02ef4c09ab8f0e376ce84838 : FAILURE in 13m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb3f7da3b7c141e4911f96b5032c4f5e : SUCCESS in 17m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/06250060341345528435c794628778a7 : FAILURE in 4m 59s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6a38851adde44e4db38ef9f809a96663 : FAILURE in 8m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8f7bdd2335b848d1a1b290146685b71f : SUCCESS in 5m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/90ed94ef91d04fbd953a65ac0b6e5216 : SUCCESS in 8m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9c327b114840482ebcdac10a51e481e7 : SUCCESS in 8m 39s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ce7f19ed78864d9daae61fa1ecccf9bd : SUCCESS in 1h 04m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/815da152a16848c2b5502d9bf5cc6886 : FAILURE in 1h 01m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6e5d2ab1323d47a9a10515750f6bd40d : SUCCESS in 1h 29m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ea68c36e234e47b882d3dcd5f39f58ec : SUCCESS in 1h 04m 13s","accounts_in_message":[],"_revision_number":2},{"id":"7c16d15bd0e568ef890e26428bff56bb6d19e6be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-28 22:52:02.000000000","message":"Patch Set 2:\n\nI don\u0027t think this will succeed til the taskflow version is updated in upper-constraints.txt","accounts_in_message":[],"_revision_number":2},{"id":"0ad785794017f75bf3acf5808b03f8065de1ad2f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-08 23:30:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8c8cb9f5b940d1ce30fbf370d6774109a1737462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 01:08:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6180f7c67e854a00b4465705f51e39eb : SUCCESS in 3m 56s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8b895d7561b64de68ab2225f894e0bee : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e933ce5898cf45578fc73e4bf4727b20 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0bfdd8cb94b04226b99a8235645bba4f : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d18bb59aa4ae4a0eb37067abf8cce803 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b6d6deadb9b49f28c43f910e91c31ba : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dab4970c99834f2a989ecf412bb0eabe : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1425fa6c146849e68f546cb2dcc022ac : SUCCESS in 3m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/997535f47a1346ccbbb90d06ab8aa922 : SUCCESS in 7m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/adbd1b049b8048169394eb1b4d026485 : SUCCESS in 7m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e44b382dbb3747819efbc7dd9ec85343 : SUCCESS in 10m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2cc5a94b55684fc1ba17f67d664c42f6 : SUCCESS in 4m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2de2b7c96aa5432da740175f28faf384 : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3cb991a0a5224381a189316b0217c3f9 : SUCCESS in 6m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a177019cc1454c3e852eafc31f116b9f : SUCCESS in 16m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b12fcfde6b5c46139b61ca4f50d9074a : SUCCESS in 26m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/df523a7347534d66b1450e797aad5702 : SUCCESS in 12m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7637a9f8093d45728e107f8bd9f24b9e : SUCCESS in 12m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fed87be5c3f44e9da75ba5ea024ce9ad : SUCCESS in 18m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cda4fb3159b14ccc8baa01c965042fb1 : SUCCESS in 4m 47s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dcb932e32fe94985963c582d71d842ef : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c18ef011938b4b69a0eda3c071d10d0b : SUCCESS in 6m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/42074ccd6c3f49c088c78eb403e32a70 : SUCCESS in 8m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fc4c297a394b4ff49597f7a57473062c : SUCCESS in 9m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbf90cc112d04b1a93091954d16f7ba0 : SUCCESS in 53m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0efbb4b0e2364dd687a4d2418bab3d2b : SUCCESS in 1h 14m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dff0be4bbf4e498b8103d990c236e0df : SUCCESS in 1h 30m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6bd8bae58cef437ba044f1362c3d9b9a : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":2},{"id":"56bbb9287035437955f05ea7754cffdf894064f0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-09 01:26:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5ee60c1005c42fcbe240c84ec5e8c511a739db2a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-09 01:26:52.000000000","message":"Abandoned\n\nalready done","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f579d96a2d31bf3e94ad2d5c549c2a6347fb206a","revisions":{"db4c78b129b4bddee91ee56b622ed476474d781a":{"kind":"REWORK","_number":1,"created":"2019-10-28 14:30:54.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/82/691682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/691682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/691682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/691682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/691682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/691682/1"}}},"commit":{"parents":[{"commit":"c02add3c9eb49d222b757619c8950a35303c5cde","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c02add3c9eb49d222b757619c8950a35303c5cde"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-28 14:30:16.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-28 14:30:16.000000000","tz":-300},"subject":"Bump networkx to 2.4 for py3","message":"Bump networkx to 2.4 for py3\n\nChange-Id: I85a222b9c1ca59e7ba9669155838cfc34d081a7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/db4c78b129b4bddee91ee56b622ed476474d781a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/db4c78b129b4bddee91ee56b622ed476474d781a"}]},"branch":"refs/heads/master"},"f579d96a2d31bf3e94ad2d5c549c2a6347fb206a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-28 21:17:31.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/691682/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/691682/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/691682/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/691682/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/691682/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/691682/2"}}},"commit":{"parents":[{"commit":"c02add3c9eb49d222b757619c8950a35303c5cde","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c02add3c9eb49d222b757619c8950a35303c5cde"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-28 14:30:16.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-28 21:17:30.000000000","tz":0},"subject":"Bump networkx to 2.4 for py3","message":"Bump networkx to 2.4 for py3\n\nDepends-On: https://review.opendev.org/689611/\n\nChange-Id: I85a222b9c1ca59e7ba9669155838cfc34d081a7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f579d96a2d31bf3e94ad2d5c549c2a6347fb206a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f579d96a2d31bf3e94ad2d5c549c2a6347fb206a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
