)]}'
{"id":"openstack%2Frequirements~691690","triplet_id":"openstack%2Frequirements~master~Ie196d7c2430d4e7f01fbcc25ecaf3869e2243f0f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie196d7c2430d4e7f01fbcc25ecaf3869e2243f0f","subject":"update constraint for python-watcherclient to new release 2.5.0","status":"MERGED","created":"2019-10-28 15:22:47.000000000","updated":"2019-10-28 20:37:05.000000000","submitted":"2019-10-28 20:34:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691690-1572294894452-c4b06128","meta_rev_id":"a63a2a77023e3f81db0df82a2117fc7f6f09f4ac","_number":691690,"virtual_id_number":691690,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 20:34:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-28 15:27:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-28 18:29:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-10-28 15:31:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-28 18:29:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 15:27:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-28 15:31:42.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-10-28 18:29:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-28 20:34:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d3c61407de832b7d300c1d1ba6e45145197620a0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-28 15:22:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20e5d8f6cf487cd1688090b57ce8dc5f7d470811","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-28 15:27:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"954149834562b42e79e3108cb6e44ee53be65a4e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-28 15:31:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"80f29db09b73257b8d8c4f765702ad6c69dc5a0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 17:13:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93805454b3944b4bbd555f5bb6ab01b5 : SUCCESS in 4m 38s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9d3e6c26fb7f49ce84ba78c3f8e22587 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d72b8afb97045d98fdeb8feb2400c48 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1ec595d60284d0692a96c08ef4fbf93 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ad8df06649749b1b6b89606a9d53cba : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40b737e102fa46fdb490793dcf5f8631 : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/292cccd2304748f0b9f888fe7c953018 : SUCCESS in 6m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b3aa273ceaa1434a877d044146c58aa7 : SUCCESS in 3m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f4badddf655443b2acbb28f92a9f2925 : SUCCESS in 7m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2615311f2f54b309434b44f02d0c57a : SUCCESS in 6m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1d3de924fc5644fe8c7796a484dfa709 : SUCCESS in 9m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dfb7666d151a43bb90ab0808e0b55226 : SUCCESS in 9m 16s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/28f29d42dae84e4192b3dd480a261c3c : SUCCESS in 4m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6de33672ab64484ab80e7f65d012cbbe : SUCCESS in 4m 22s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/218015966f11445486bcc1e624c98964 : SUCCESS in 10m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/90a289eead7d4a3ab2183aef2d34d519 : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/805905efbb7c4516aeb32e37ba10f171 : SUCCESS in 8m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ded0b6e43b8144cb946e3fbe9aab8675 : SUCCESS in 14m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9302fd565b88416fa4fefd998dbaf568 : SUCCESS in 15m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9d269c918a4244c092c98c258fdb5967 : SUCCESS in 21m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a90bf19d2c8048b2be4ca69feca193db : SUCCESS in 36m 00s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2c8c0fcd8900410f937ddd81b4e128a3 : SUCCESS in 14m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/09fb441d86774c0086f574b85b17eae4 : SUCCESS in 13m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/093f25d80f4049fe8cd8e88b0c857c98 : SUCCESS in 13m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7cddf8d89404511be5a8333d17c8dc9 : SUCCESS in 17m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/187ef5bd87d14423b5f2907b25caa9bc : SUCCESS in 5m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3123b42950534121bca58f22e6c8820f : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/11e38db2adf04f85b6bc961992576a56 : SUCCESS in 8m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6756257dfcaf4055b65c28025e46b574 : SUCCESS in 11m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8f667398aae045ff93853e7c7127ff26 : SUCCESS in 10m 11s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8e5b2de82f9e4efaa0ffc5eeacd77b6d : SUCCESS in 1h 18m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3810ff3025014a4c9e37cd5fa36cddd4 : SUCCESS in 1h 30m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6d8ff826d0ef40e9b1195808af6fa7bc : SUCCESS in 1h 44m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3f1080fd2d674ffbacfd39a358c5ea4d : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":1},{"id":"702b230529b0c08465012f26f229567ef931a125","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-28 18:29:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2809391f88194d36085df821ffe82aa6fba1c785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:29:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47f65337004f0f8edba90eb4d7999a4a0a43f218","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 20:34:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925d28a6de014ceb95a1cbf88eb662f0 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49ec639a0b44460493aa7420e69cf4dc : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5768a6d23dd2427ebd2a637ddfb55ba2 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9adb2bceae3a44b7b4bc6ba2f39861e1 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3dde184d520b4aadaa38167ecf0780a3 : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/952c6ade6cbd43b481170f7d74717206 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/13d075d9957546c085b3596ec557d92d : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/62e391b9e63049cd89a31a76d937bb4c : SUCCESS in 7m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a2089a6fc1eb4e51ab0d9e26c681de4b : SUCCESS in 6m 47s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/745418447ddc4c69a0e304abd6eca525 : SUCCESS in 13m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e8994ac4f14e4449a913069b5421b945 : SUCCESS in 8m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3d5062361ac549638769bd1ce16aa34a : SUCCESS in 5m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4dab34d29e554879a62042436af5fcf0 : SUCCESS in 4m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/20e587d3853c48f09a31a3cbc12781be : SUCCESS in 10m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9fd29bdf2db74dcf9a5388c2bee569c0 : SUCCESS in 9m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c5a8ae131e44c79bd20de84d1e0c137 : SUCCESS in 7m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/864c3ac385964b5ca91dd7a812b489c9 : SUCCESS in 13m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e1c46898c00f40338f106ec593a7ea9d : SUCCESS in 13m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2201db09871f4439911929385b1d6502 : SUCCESS in 26m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/076c49058f814968a2bf1145185b910c : SUCCESS in 22m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/036a59ea1fc6471ea284fad16f821a46 : SUCCESS in 13m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7178418c4c4a47b889ea244c98ed2acf : SUCCESS in 12m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/40563214bba346c5bfda34579a2216fe : SUCCESS in 12m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/94f2e27a1bc34123a8b4a84fa2e5627f : SUCCESS in 18m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1b0ff50d121744c3a773f2991f6cc74f : SUCCESS in 6m 10s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2dba55f858854550a2d441fcc80079d0 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2313217762fa45a2b7c4845d3f2ead86 : SUCCESS in 6m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ed690eec46b041f48508c52c7138985b : SUCCESS in 8m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/510c6be28e464e93a75a9a9486b44405 : SUCCESS in 8m 32s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/66e4739128474db987cbc42eb1902cd4 : SUCCESS in 1h 10m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c2e8334353f45e6828531762461693b : SUCCESS in 1h 10m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bb8cb339d83d4a02bdca15d686eb687a : SUCCESS in 1h 39m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/df14638de0d54e208120257809bbad46 : SUCCESS in 1h 02m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8b5997f4f471295857d73a5d5e8caac5a776a5f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 20:34:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"28a303f2ba50e2562106ace79247595fdfb67c9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 20:37:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7dc5a63e2fc4668ad6645b555b9b780 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7bf643976d03865000037b443fde8a95254fb15","revisions":{"b7bf643976d03865000037b443fde8a95254fb15":{"kind":"REWORK","_number":1,"created":"2019-10-28 15:22:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/691690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/691690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/691690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/691690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/691690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/691690/1"}}},"commit":{"parents":[{"commit":"c02add3c9eb49d222b757619c8950a35303c5cde","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c02add3c9eb49d222b757619c8950a35303c5cde"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-28 15:22:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-28 15:22:45.000000000","tz":0},"subject":"update constraint for python-watcherclient to new release 2.5.0","message":"update constraint for python-watcherclient to new release 2.5.0\n\nChange-Id: Ie196d7c2430d4e7f01fbcc25ecaf3869e2243f0f\nmeta:version: 2.5.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Commit: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Change-Id: Ib121130f198d3f4cbd76b73e3202562470fca1dc\nmeta:release:Code-Review+1: chenker \u003cchen.ke14@zte.com.cn\u003e\nmeta:release:Code-Review+1: Hongbin Lu \u003chongbin034@gmail.com\u003e\nmeta:release:Code-Review+1: licanwei \u003cli.canwei2@zte.com.cn\u003e\nmeta:release:Code-Review+1: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7bf643976d03865000037b443fde8a95254fb15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7bf643976d03865000037b443fde8a95254fb15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
