)]}'
{"id":"openstack%2Frequirements~691691","triplet_id":"openstack%2Frequirements~master~I0000fb5d6783890d71db8cbb497f584f7573b9dd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0000fb5d6783890d71db8cbb497f584f7573b9dd","subject":"update constraint for python-zunclient to new release 3.6.0","status":"MERGED","created":"2019-10-28 15:24:38.000000000","updated":"2019-10-28 20:34:55.000000000","submitted":"2019-10-28 20:34:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691691-1572294895641-c91a571c","meta_rev_id":"6d436bcb401ad1b3c772b82c0f0e360e164c1857","_number":691691,"virtual_id_number":691691,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 20:34:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-28 15:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-28 18:29:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-10-28 15:31:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-28 18:29:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 15:27:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-28 15:31:50.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-10-28 18:29:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-28 20:34:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"948b6e97a68ccbfbb8907e5040b373b29122ee22","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-28 15:24:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2424617e28b96354a9591c401da9e18cca09727c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-28 15:27:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5e2c07bc8182f7cc9d716a26f671265823860624","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-28 15:31:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"717d1b69741c383b6538dcb1def22a673b565d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 17:40:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb34e1d46324795801b9ecd61d93702 : SUCCESS in 5m 15s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3e358d10ee17417c9d7b46ea46022015 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ecd5087d35f4100ac201b24d2d943a9 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d97e2357fb34daba8ba4690c68b8e4f : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbd67afe9f2148f392b86f2a1edbc251 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1c80943fdc24365bf80babd5cf478a0 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8bfabf542d114332bc7f4af136bb059d : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f169bb4c2294cb5b59e3a013efb2162 : SUCCESS in 4m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/09cc65e8a4854986a74d29d6c4b23508 : SUCCESS in 7m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7b9bebf17caf41aca60aaa7fb726145c : SUCCESS in 7m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a27b270d5b7348668089dd6f76dc7937 : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/341fb3743463472bb5f29540a47d26c7 : SUCCESS in 8m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b7a57dd1e51d44a2bd2f65e89d1ed08a : SUCCESS in 4m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6f9f36a58b26473cb2951922071ccb08 : SUCCESS in 6m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/885150de2456473b82effae0fb17a0d5 : SUCCESS in 10m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/81f6235b6f1c4351829c76d535989380 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/279c9caadc1746a0a5de3abaca62c84c : SUCCESS in 7m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/63157f46df204291a5d3d6a23a51828c : SUCCESS in 15m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4a142fcb6dba445bbeed007108d7a13d : SUCCESS in 27m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8c852680005b4975a8ebb0f937224b19 : SUCCESS in 19m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/18aa1cbcb34848eb9c9c78ae668cb351 : SUCCESS in 20m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7e36c79943e94b93914b7c37141561b8 : SUCCESS in 16m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ef55d1830fb04f3f8b1168a132190696 : SUCCESS in 11m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c9067ba92e7241b8b708f18e99b9a851 : SUCCESS in 14m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d456fdbdb2374589b13901dc69e45adb : SUCCESS in 20m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/06eac206bc594ba184dab4a2f04a8110 : SUCCESS in 5m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b13c0bd3d7034e94beea354656b42019 : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/25cdf196b5e84170beace4582ec4bf32 : SUCCESS in 6m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4db95dac44fb412ca59be9a58a545b9c : SUCCESS in 9m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dcec458a1b164b0aadc70a3c90f49ac9 : SUCCESS in 10m 13s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/b5fb155252e04508a8edc5fa1b3e8bad : SUCCESS in 1h 11m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b17e60e2376e44f8b38cc0e2c7ad4852 : SUCCESS in 1h 09m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7b12193e3b674c7ba717c90b89b5c55a : SUCCESS in 2h 08m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5b6c8a5a4c7446fb9e44e9a8cafbfbd5 : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ff18021b705afad02d81f07a2eb25dcc6bdd2d1e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-28 18:29:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"329281fd0bb330fd2270c45605684586bdda366b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:31:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba1cf692302620f13dd77aff4483d443c5ecf355","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 20:34:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"207c97c83055254c1def6694b8a8ed0039bf769e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 20:34:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6864bfb5590e4b10b3c83229a13f737e : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d50f590bf2948869854aa992bb8a803 : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85a584a357c94e9cb73de47a2a8c7a7f : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cba82d0fedc4fa193d7d6693f3a23f4 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d4ea822c3d44c798e00315d44364a69 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12d14260e86b4ee59433cdc1d42c06f1 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a1676b7480f49938e2defb0363ab02a : SUCCESS in 3m 05s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/80957a87dcc94157b6303c2d4bbc4a43 : SUCCESS in 6m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/961141f40f2a4f57b746cbd7b3b5674f : SUCCESS in 6m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ea319315083e470e9003e32b640837c0 : SUCCESS in 9m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/be2a1d2edd144bf79debaf56b38486c4 : SUCCESS in 8m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d5d1f99f24494c3bbbd4b1d5b283f170 : SUCCESS in 4m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/56a278ddfd14445781b3dd68a47adfd8 : SUCCESS in 4m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/888057b38efc48ac92e254d81ce9eb5a : SUCCESS in 10m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7709cbb3f3404b27b4e0217fdaa44859 : SUCCESS in 8m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/842ad3a37d604b9ca9d29dc31a61127b : SUCCESS in 7m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d43d5db81a6246c3b1d58512cc2bc8d1 : SUCCESS in 20m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4dca5518064f472a92d626d7a8bce754 : SUCCESS in 33m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b94c29dbc50f413893e7b78d87e6e7d1 : SUCCESS in 31m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a401064422aa485e84652c09e146800e : SUCCESS in 21m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c6056b14ec604cd58e75f4611fada9e5 : SUCCESS in 15m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3e09d35783ab4b76ac6048a2468ca390 : SUCCESS in 22m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/714be2e4f6cf446fb3627193b0ba497d : SUCCESS in 15m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dbae748f163445f09fdd5e4a2933c374 : SUCCESS in 27m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3ff22f38615d47f08a9f9511db8fc6bd : SUCCESS in 6m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ceef7b2bbaf84bd8a7d53b25ea8ab851 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55b413d7a3604368bf50f0c76ee0d9cb : SUCCESS in 5m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2a9344d5a93740639b5b82e2a79bc59b : SUCCESS in 11m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e41cf33d29cb4fd88d89b2f60d720b0d : SUCCESS in 17m 09s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0ec853469afe401bbda4a4af4f2fa9ce : SUCCESS in 1h 28m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0064d79dbfbf4a57afdabbfae8aa543b : SUCCESS in 1h 14m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6fea00be25b143beb8a1df1cfa75131d : SUCCESS in 1h 39m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e3a48a8b576145858e4eee07c569b05d : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2f40d11e9e63a6f0ce4de49c8b05f22574e64fa","revisions":{"d2f40d11e9e63a6f0ce4de49c8b05f22574e64fa":{"kind":"REWORK","_number":1,"created":"2019-10-28 15:24:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/691691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/691691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/691691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/691691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/691691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/691691/1"}}},"commit":{"parents":[{"commit":"c02add3c9eb49d222b757619c8950a35303c5cde","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c02add3c9eb49d222b757619c8950a35303c5cde"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-28 15:24:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-28 15:24:36.000000000","tz":0},"subject":"update constraint for python-zunclient to new release 3.6.0","message":"update constraint for python-zunclient to new release 3.6.0\n\nChange-Id: I0000fb5d6783890d71db8cbb497f584f7573b9dd\nmeta:version: 3.6.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Commit: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Change-Id: Ib121130f198d3f4cbd76b73e3202562470fca1dc\nmeta:release:Code-Review+1: chenker \u003cchen.ke14@zte.com.cn\u003e\nmeta:release:Code-Review+1: Hongbin Lu \u003chongbin034@gmail.com\u003e\nmeta:release:Code-Review+1: licanwei \u003cli.canwei2@zte.com.cn\u003e\nmeta:release:Code-Review+1: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2f40d11e9e63a6f0ce4de49c8b05f22574e64fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2f40d11e9e63a6f0ce4de49c8b05f22574e64fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
