)]}'
{"id":"openstack%2Frequirements~692153","triplet_id":"openstack%2Frequirements~master~I67e50a959007000eaa79f24b7b0535efe2279289","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I67e50a959007000eaa79f24b7b0535efe2279289","subject":"update constraint for aodhclient to new release 1.4.0","status":"MERGED","created":"2019-10-30 14:30:00.000000000","updated":"2019-11-01 08:25:08.000000000","submitted":"2019-11-01 08:23:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692153-1572596626296-e7a1f455","meta_rev_id":"e1106ba348ba5201db1771a10f331cae7d0b7de5","_number":692153,"virtual_id_number":692153,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-01 08:23:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-30 14:32:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-30 21:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-30 21:48:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 14:32:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-30 21:48:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-11-01 08:23:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"926356fd69dec0cbadbf5a8a64775cdbf53b10b2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-30 14:30:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"345162cfe3584d8bd887429c87fdf29a137d1c69","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-30 14:32:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a76d9c1e9fb042e468b850ca808ce2b10586846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 21:41:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047153fa61774f9e92e93f4d82dd80f0 : SUCCESS in 3m 33s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c07cb2df8f16409ab0c46c5dac778c56 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1980b9c2f7f4178ac40f417cf479518 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7b02c38626f41e5a6a401b5069bb927 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df18effa53754c5e8125a6c388eb305e : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a07fe71694cd4b439c8b6b657395991d : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e02ddbfa86647c599f22843d3a78e9c : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c99f8d27f2e44d11abc79d72bdb6316b : SUCCESS in 4m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d60814c8470f49f29974f15609a40d01 : SUCCESS in 7m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f2b84a9bfb4848abba4c01d1d2e77470 : SUCCESS in 8m 01s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f8d60c73937c4f40881dc0bcf91c2d51 : SUCCESS in 11m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/973298cfce7d4ae7815a98530c064ad4 : SUCCESS in 7m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/083be3e0abd9478b94d188b5a4d70ea1 : SUCCESS in 5m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/09076e8c59304218be282708adaa2804 : SUCCESS in 6m 05s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d511039da2c942adab83f853653d4c22 : SUCCESS in 9m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8e06d55c9ede40b8b688ff6cb1d8e5e0 : SUCCESS in 8m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3de20a707a3c401c9316b09939cc2614 : SUCCESS in 6m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/81f892a0b13d422b85e13766aae1836a : SUCCESS in 17m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fc71b80b2b39468ea7037cb0184a5e36 : SUCCESS in 21m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f2f495057c3c40edb6a27dcd2dad10d1 : SUCCESS in 24m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dcba6ba343404fec9271a0c9f21f64f8 : SUCCESS in 32m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b0b2f8278d4e44ef9cf8043d393d3b1d : SUCCESS in 12m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/330fb50c767740ecb95080a87aeb8ce2 : SUCCESS in 14m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/65addaa905d344d288f88a425e0e9ad6 : SUCCESS in 15m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1613ed8e740b4fe9b5af94748f0d52b9 : SUCCESS in 21m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2102ad079b564ad89d0dde8b33280d18 : SUCCESS in 4m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4f6e3fc4bd3f461c804fcc6f808195f7 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76a9457ed6fc48eb9f93eea7bb4e5a0a : SUCCESS in 6m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f033673c198643fc95790df6cf7e6979 : SUCCESS in 9m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b001659e1497431687b1669b636827b5 : SUCCESS in 11m 07s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/31ca1f2a75d94431b5c596eabcdcf031 : SUCCESS in 1h 08m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4df210ca702f47588d3b0de6a6e0c679 : SUCCESS in 1h 17m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e3d647cf624d4032bc7594499b3c02b7 : SUCCESS in 1h 50m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5208e2ae74484235b8cca9a63f375681 : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":1},{"id":"35cff01bd734fe33575b7c24749c30f9eb17e21b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-30 21:48:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"61008588b995bee6a6a3bd9f0e93c2c07c32fa60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 21:48:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b785940ee11216c9dd9a89a63457bc4fa40200b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 05:02:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a7009cb144e49a1a4dd3152865a9849 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4c61d5943a5422297dab7a9a56b82df : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a62e047e0790464092bc4c88b19c3a17 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b08dbff4835c44998e18b1b812e0f00c : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50d3b9a7890647a98aba7774c92a7980 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/066ab3af857b40cf877dcf56810870db : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ca4ca8edddb24dec94635607d9054c0e : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/947970e370cb487081099e4bbdd07cd0 : SUCCESS in 6m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/18b9fd1d3ac345479625e1dc04bdc2cd : SUCCESS in 6m 39s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/24c64aa353e34230b27402cef15f7fa2 : SUCCESS in 9m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4c74ae2ff5a14c00b0fd770aa94556f7 : SUCCESS in 8m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/aa5f4798c9d249f19e90311ef2d4e4b3 : SUCCESS in 5m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2231457223aa46a98c13d74cc406b3c5 : SUCCESS in 4m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b2785b29e7694655877db95975bafa73 : SUCCESS in 9m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/862e710222804bad9e355744ad9bad2e : SUCCESS in 8m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8151f1af8d5d44bc9c4be3e0cba04030 : SUCCESS in 9m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8ca62ec754bb4ea999ab8201cec1a386 : SUCCESS in 17m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2677fe7db0ee4dc49ddc2e7497de41c7 : SUCCESS in 18m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d81059cb06fd47689c6adfc0e00ca6d5 : FAILURE in 23m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d0354684ec9d4928ac0aedbe52c6bfb1 : SUCCESS in 25m 08s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c2d16c4f054142549f4e2c965d6b07e2 : SUCCESS in 18m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/025ad0dc6f3e42ff8d4b91f572c92144 : SUCCESS in 17m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/61b3b4fcd04a462bb1d3cd6a59de15a6 : SUCCESS in 12m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2c904e288484a6d9415355f683ca740 : SUCCESS in 18m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e2d6e777ce224144bc5728880d758725 : SUCCESS in 5m 56s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e2924aa3142c4a13ab3f97cbeada8e6c : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1a58f309881e42398a0126af2e1a72de : SUCCESS in 6m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0d12f60152fd4abb8e5071573a202d5e : SUCCESS in 13m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f8bab2c3e4ef4b9c87993d0127265a72 : SUCCESS in 13m 03s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1ea88653bf344e04b88a20d15c5caa40 : SUCCESS in 1h 00m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d31daf15387d4b0ab65916f82345e202 : SUCCESS in 1h 03m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cd0285c226324ff3a93f7c4cd1d65e1c : SUCCESS in 1h 46m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3d9d9ba3effa48ceb8e27b0be92a4a0a : SUCCESS in 53m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6a9646536fa485b9ac47fa165e01580f75d7d0a4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-31 05:04:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"891be70666c18e4368e5d6d2c378fac3793aac38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 06:54:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b83ca8bada7940b2a6bfc440201f2815 : SUCCESS in 3m 15s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/34f8871b33884f369e639ae6f3156143 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3953a32f011542f0a5f098603d5246f4 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9bcddbd4a4c4720a970a88cabe010aa : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a5749f29bd04c22a565e1d6f6449dc4 : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1f3a0693afd4d2ab9a5b9bfbf45221a : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce6d74ceef5747feb40a04bec007f49f : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f90c5a154da247adb4595c75dac4744a : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7d6b221dc5aa4b28a7849b00b8530e20 : SUCCESS in 7m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9cb5795c9c814b9e9bb8aa01d38c37aa : SUCCESS in 13m 15s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1b66481ce26b4e8e9d19e012b95fda1b : SUCCESS in 9m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/634c0ad10e7b4b64a7eb60ed6c172369 : SUCCESS in 14m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/58c6bc2716c5478faa5b1d0f55a7f316 : SUCCESS in 4m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/375fa8d7a6414736afd479af6347a267 : SUCCESS in 4m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5dcd0e59a54147dfb8c37a592e827e88 : SUCCESS in 10m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/40af7e53849d4d0190b0f1c24a04f1d7 : SUCCESS in 13m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fdd2025bd6f5404ab4969ed8e23e01f1 : SUCCESS in 7m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f89e1aa7cc69415187c43600e34233c8 : SUCCESS in 26m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8cfe74b257494999a62525b8d11fbe12 : SUCCESS in 18m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7cab5dbed154456dbd0305010e1c52b3 : SUCCESS in 18m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/433b69c0bf214b71bfb50998a95a5bad : SUCCESS in 27m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ac274c65e8a04e50b33bdcea2e681fd4 : SUCCESS in 13m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cccea3b9a52f4750a6918cd5704e70e2 : SUCCESS in 15m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3045028e5a9945a1b9caf87ee6077c86 : SUCCESS in 15m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3bde695ccd74275b4e5dbf067f5e850 : SUCCESS in 21m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0c984070f30047388fcb1bff5ffc203b : SUCCESS in 4m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b290a6e6498d4e3699aaff2965e9ac5b : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f0fed9cfed2411ca9c3e3ef392b9fa8 : SUCCESS in 7m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/651b909043a747e5895c62194cf3be5f : SUCCESS in 8m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8ead3f5d433c4def815f73289f2786fb : SUCCESS in 9m 38s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f580b25d2ee241c8b8d7e6a41b9d7704 : SUCCESS in 51m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d06f93f5d0a4b09a826ab2623586986 : SUCCESS in 1h 02m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e3a49c7a1cdf4f6b874b0177ae4706f8 : SUCCESS in 1h 44m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5e4e03e8d7d9401192c70de935fd23ba : SUCCESS in 1h 09m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a2efec66bc38c74dcaffb9cd016a283d655d5d1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 06:54:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e4ce17ebb93d4c2a75229b35df0436f5a87e13b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 10:31:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a14350e53994106beba335369ecc6a9 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeaa99249ac04e4aa63877ab5a28230f : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08b32601b6954ce580e04b8a926c6933 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6ff9f16901d4f40b7579f237b3efe5f : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c83a2a88d7e141ed9e860f9ae233b542 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8029009bafb7464499440c1b794a50d9 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ff5492099af4ce0ba1b1d9434c19138 : SUCCESS in 4m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ab73c98e4fd24e9aaddb1783dfc18d80 : SUCCESS in 12m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3a6bc6d413c14dabbb0aac5d1fc099b5 : SUCCESS in 9m 47s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/94a25924c9324c89b5e3f7e5819bdfc0 : SUCCESS in 10m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6d943059e98347beb971fb52f6f03206 : SUCCESS in 9m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/67583b3ba2c544bdbd727166aad7ec68 : SUCCESS in 4m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f0b664313a5b4eebba2b60016fc4ab41 : SUCCESS in 4m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/eb098203e0104b4da697a3d0b6bd86c2 : SUCCESS in 10m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/119b03c7bc02495094ee86c55c06f616 : SUCCESS in 8m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c703d5d004a140fbbee4e373998f2350 : SUCCESS in 7m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/27682130ec4e46e5a84e0fe44ab27c2b : SUCCESS in 18m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/caaa6e5edd6a4c9cb645936842398f27 : SUCCESS in 18m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c0ae0e0372844f04b8e3872a2a3aaac2 : SUCCESS in 19m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/94ee282fcf1043519a9f85769c08c641 : SUCCESS in 19m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/168d5adc81784309babd2226b9599f95 : SUCCESS in 14m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c824878a4e3e4118bcdc31861e5cc625 : SUCCESS in 12m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fe27017ba3e9468abf98dd986088d2f3 : SUCCESS in 12m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce6fad0eeb3a46af8a66175e80299b3e : SUCCESS in 20m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/381002a2c4a6404bb03a65951d3bf45c : SUCCESS in 5m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/53bad266abb4434685d87a7b05d99538 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/77a970d266e4479caa05bd04efdaedf5 : SUCCESS in 5m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9bf755c5b46a43008295efadb020e47b : SUCCESS in 8m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0a9c61d5e92a4bb0b70e51627b302bfc : SUCCESS in 9m 57s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5f417c68595243dcbb3f386c036482d2 : FAILURE in 1h 14m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/47fae3ecdccc4978af5f95c4ca6a93f3 : SUCCESS in 2h 08m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7c10c880c8cc4d83a1038e5a69b8f67c : FAILURE in 1h 23m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8f9d017857354390a64875e77ee3d025 : FAILURE in 1h 17m 54s","accounts_in_message":[],"_revision_number":1},{"id":"705113d3e5502cd2fd101e570fdbf03a0bb5d285","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-31 15:00:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4da5a3665e9bb5f47225b447fba0771383f9f73d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 02:57:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba3d534d9784858aa11080b98914c2a : SUCCESS in 3m 43s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0419aa2745ad42ea93461932cf30f39a : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca353105d85647cb9d771cc200c0f6ba : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3008f8c519049b1a465694606ece13c : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec54c4de6abc4a36a2eb94eec1cb7c89 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac05192a20fc4ed894275e248a2de293 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1de03640b40441a932f0e7970ac2f52 : SUCCESS in 5m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e85c578568f54b84b8655d71fa847aa1 : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ff18736fb7544b17a1445b9c142e3d4c : SUCCESS in 8m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6c8d01d75ac24e92a8ac67933cf33a46 : SUCCESS in 7m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bac1b0816e3546b88e47c591e13a4030 : SUCCESS in 8m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c583c3d9ed484a67894d75d0a526f654 : SUCCESS in 9m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/771dc490764048f0ab4313cd0f83975a : SUCCESS in 5m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/96aa72659d5d443e95f5ff892e81150e : SUCCESS in 4m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bfffee2298c9409ab6201629a826fafb : SUCCESS in 13m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7f2402ab543d49f19fb5667335b23317 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f843d574811e42288a313806d22529eb : SUCCESS in 6m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/845cd1a3528241d3908773880eae2c48 : SUCCESS in 17m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c3b66617620f4178b22362724eca79d7 : SUCCESS in 17m 58s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4524af75286345ecbf3f0007a5242536 : SUCCESS in 22m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fa8855f3d3284463ad3d2184a18f5aec : SUCCESS in 20m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3f34003cbb2a4085878dc98f6fa41da3 : SUCCESS in 19m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2a324855355948cb823b5919b2d821f7 : SUCCESS in 11m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7436c7c094484e0499e72d17e385e616 : SUCCESS in 11m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2902e24061d45998646d09ecd5ca9b7 : SUCCESS in 15m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2a89c13031ad460a972373560d6f29d4 : SUCCESS in 4m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6d946f8a4f62424787521230c834550d : SUCCESS in 7m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/093cfc02428a4262860552ff514c7c78 : SUCCESS in 8m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/18edb7c00a0947d3be37a5f44df11c7b : SUCCESS in 12m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e5c28b9f9f734fb6b2c9bf1349d95e6f : SUCCESS in 9m 45s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/fd9c6d2a441f4fafbac4f7bbabbc3472 : SUCCESS in 54m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbe279bba38149129661e5cf4518139c : SUCCESS in 1h 02m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/330d312b469c4e4abdadb1e28205addb : SUCCESS in 1h 32m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d46f85272b6d4d6698064c6e4cc32086 : FAILURE in 1h 01m 59s","accounts_in_message":[],"_revision_number":1},{"id":"fa0d0dfba9a8981e3da05efc3f5c90b653092726","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-11-01 03:00:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8827d61864e3562e668f2b6ca45b660c68299cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 06:23:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08521cb14f8d4e4ba6728b564d4d8918 : SUCCESS in 5m 20s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b7e83725c2024737b2e80ca60a3d06e0 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8a34f5a21ec4d31ace710b0c138f7dc : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e439d500d8764418a6918273a431834d : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48fe2b9de19645f78f155c0524acc392 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5c2e9f151ac4aee89293b4dfd9b23ca : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46cd6a263e3b4058a56d580fee1100f1 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81ce91d1e0574dc0a282287f83b8a649 : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f5d0ccfbe5994c009ef7d19aea69fb33 : SUCCESS in 7m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/228b131fc9824ba2ba2447e2b95d0c55 : SUCCESS in 8m 11s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e151f6e09e454400bbd316fc5660fbbe : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/126e780c327f4febab5a0b7f1bfb938a : SUCCESS in 9m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/02d0f8e5b51e4d119eff41bc1e48a334 : SUCCESS in 4m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d34907a0dd56427e94864334f33a3c01 : SUCCESS in 4m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e214756d7fbc468fa235451024410cc6 : SUCCESS in 9m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/09afa5479d9748b0a3192f280abc3010 : SUCCESS in 8m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc8042afba854332ace8356a0ea66b28 : SUCCESS in 8m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/38c0dadd3c2648fd8b533ed779ca2d88 : SUCCESS in 25m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a399d2396e22457c8a0e16cc67e8c27e : SUCCESS in 25m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8a2c7f310dfc402e9bb41963e48c88c4 : SUCCESS in 19m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/628f1760acdf4ccb9d924ce2af46bc3b : SUCCESS in 18m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c81a1fef19e7464cb888e10942e40b96 : SUCCESS in 14m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5dc711e2dd714611a066c1863ac3bafe : SUCCESS in 13m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f6fcbdef53564346ae16b80ade5bfb59 : SUCCESS in 13m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/13daedc6597044f59adb7c53c5ff97fb : SUCCESS in 17m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4fca95cb7a314e6c88d6734e337519b0 : SUCCESS in 4m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/10484ee750b748c2b873e199004db88e : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec82ee29063740ed93e4a7518b59441b : SUCCESS in 6m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c72680593e424f089534777753f8d41c : SUCCESS in 9m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aa1521ddc678469a895a71f172d87138 : SUCCESS in 10m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5b86af09dea042598755a4fb0fcc3d09 : SUCCESS in 51m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/15811c92183f410ca44eb40c806e7b72 : SUCCESS in 1h 05m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5c0982e6518042beaff7debc0d4c3e43 : SUCCESS in 1h 25m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ea5cdb53085d467bb81578e3e506a398 : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":1},{"id":"41c84bccb789d77a3ee84d60b7b660da38f7c016","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 06:23:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5593c0f9ceac2138cc87d0f3c13ed3112c003236","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 08:23:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e7110c3ee647cb9c93ac2995a65b1e : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79ce119b19884396b74807c370c332ad : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7508b9b8744145a8b7eefad1a59fb13b : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3960fedf1b8742f2b53630b8a000c826 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/653fc41ecf4441ae8877d1df1e80b8e7 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a092ea956ce14e58a097c87283e00f2b : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ca42155f85ca430ba0679769b5eb0f12 : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d9f2690eb5f447eaba4d60a5aa2cec91 : SUCCESS in 6m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3dea61f4f0864915af15c4c3234ca29b : SUCCESS in 6m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b3d149961ba642f2be85e92dad7a912e : SUCCESS in 11m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a83b2683728848cb956f23da9c1cce80 : SUCCESS in 9m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/abefd9a8e81b43f19841b0f17a0d3a5b : SUCCESS in 5m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0cada34c1e5b4b0e9ee08d68446e2b7c : SUCCESS in 4m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3b1b1fa9cd5f4ec3b0cc67cbd1c800a1 : SUCCESS in 8m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b8077021aff141a18da4a4c726c32bee : SUCCESS in 8m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b685d2dabcb4a4ab17a144be05d0a60 : SUCCESS in 7m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c85a282e3ad04a8087ca2e2c9b3fafd9 : SUCCESS in 12m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/80f13d3e2ca146499531d912f7591096 : SUCCESS in 20m 49s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/90a663d315c34fb98be1acf723b3b0ee : SUCCESS in 19m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0afe3e97ddc34564ab9f0e3ea3cd48aa : SUCCESS in 26m 47s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c6b33954cfee40a9bb8851fd79932cc6 : SUCCESS in 12m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f09861f62b05455885fda1a0d2c58ce6 : SUCCESS in 17m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3f19f965d26644f0ba3233c8ffcadfe1 : SUCCESS in 13m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/af8670390e154393be2dc135cd274b56 : SUCCESS in 19m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d73f8bb723c7432d9bd5cb48a7eb3205 : SUCCESS in 4m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5fb58ba2706e4cc18541eff10a9da1ab : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/940a38e50d5a4148889216cbdfae63a4 : SUCCESS in 5m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dc9e86001a254a9f915f82b37da8fcc8 : SUCCESS in 12m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/eb38fb51c89c497a8ee90d70c8469ea1 : SUCCESS in 10m 05s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d2e1ba0eedd644788c9bb6f1899cfa4c : SUCCESS in 54m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d4738c100054c038c318a8d95afba9a : SUCCESS in 1h 12m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/80aee0732ac341ff944cf78ff43274aa : SUCCESS in 1h 48m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5593e6555c4341efba8b922dd7d25b1b : SUCCESS in 1h 11m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c0300e353f24a69548a0ac6c11e8dc9d0537867a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 08:23:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"df07bdf9de24c39d4c9f8aad61575762b1ad9b20","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 08:25:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e31d9a13b1e4705b2dde2f5edc56bd5 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7aad2cb1556d9779b1a45d7685d747e3e1261854","revisions":{"7aad2cb1556d9779b1a45d7685d747e3e1261854":{"kind":"REWORK","_number":1,"created":"2019-10-30 14:30:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/692153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/692153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/692153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/692153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/692153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/692153/1"}}},"commit":{"parents":[{"commit":"4e7d3ce46bed5a7c9f267b49c05663f82400bb95","subject":"Merge \"block versions of salt that cap pyyaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e7d3ce46bed5a7c9f267b49c05663f82400bb95"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-30 14:29:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-30 14:29:58.000000000","tz":0},"subject":"update constraint for aodhclient to new release 1.4.0","message":"update constraint for aodhclient to new release 1.4.0\n\nChange-Id: I67e50a959007000eaa79f24b7b0535efe2279289\nmeta:version: 1.4.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Simon Merrick \u003csimonmerrick@catalyst.net.nz\u003e\nmeta:release:Commit: Simon Merrick \u003csimonmerrick@catalyst.net.nz\u003e\nmeta:release:Change-Id: Ib14db0c2344db6f1c664fd22235303eb9c7c6f29\nmeta:release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7aad2cb1556d9779b1a45d7685d747e3e1261854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7aad2cb1556d9779b1a45d7685d747e3e1261854"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
